)]}'
{"id":"openstack%2Fpbr~949188","triplet_id":"openstack%2Fpbr~master~I358e2f5a016ca81dc2f3424de8cb5f090106b0c7","project":"openstack/pbr","branch":"master","topic":"pkg_resources-removal","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-04 23:54:46.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2025-07-17 12:26:44.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I358e2f5a016ca81dc2f3424de8cb5f090106b0c7","subject":"tox: Stop calling \u0027setup.py\u0027","status":"MERGED","created":"2025-05-08 13:54:01.000000000","updated":"2025-08-04 23:54:46.000000000","submitted":"2025-08-04 23:54:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949188-pkg_resources-removal","meta_rev_id":"75865d42c27fa081f1075b60008d65800944099c","_number":949188,"virtual_id_number":949188,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-04 23:54:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-08-01 16:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2025-07-17 12:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-17 10:13:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2025-07-17 12:26:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 15:37:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-03 09:46:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2025-07-16 16:15:46.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-08-01 16:46:30.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-08-04 20:05:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"01602fc2607f3bc38c138d83862b6b0275544da0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-08 13:54:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84d735e92d2b0d04b718b8c326ad945f9cd9b239","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-08 13:54:43.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e374e68aedca70cc7db12cecb0585a8a93f0152a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 15:37:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2ae2fe71df6420085713b43b5024061\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/463e5c7b1e1c4eb29f00a943cc4faa06 : SUCCESS in 1h 33m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42caf48ee18a42cfadd675d088120bb3 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d11e0222a51446cf803f02c1c254ce68 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e4815ef4aa943ce8f98048cc3268d93 : FAILURE in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cfd07665d34a4fdfa5a14cd135fddc62 : FAILURE in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d51d73dbfee9473ab0d897640abfe8b9 : FAILURE in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e27dcf0bba548c881eff8cf789424d9 : FAILURE in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/151214b80b614ce183c198a86cf2cbfb : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b11b6167eb5b4c089b93389086b432d6 : FAILURE in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6fdb0ca24c34748ae712f994e44fd09 : FAILURE in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95d620e6bd9242b5b5c34bf23768e28b : FAILURE in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0ed7cc098e840d1ba34361f983cb1a4 : FAILURE in 5m 51s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e8cc5376209443219cd7325e1628fa7c : FAILURE in 25m 12s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/523f514230ce4a56a3ae717adafdbdfb : FAILURE in 24m 32s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/64c0fb8e582a4d7baf8b862e33d2d06c : FAILURE in 16m 12s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/eb4740d0f74e4bc9888f34b714c16dc4 : FAILURE in 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d75e7d01dd6ba267206372070250e238d82c4426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 09:38:27.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"86421b48f7f3ae5345e962ba8b632b689e7260a6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-07-03 09:46:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"334b5fb66693975ed4e76ffb41bda442e70f4fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 10:30:25.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ce64afb7debb151c6fccb393fc648f53172082a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 12:28:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0276f505d2e4e8d9340afb401df3deb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f290dec5abb4476e9258d31eee8a4f48 : SUCCESS in 1h 44m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b456fbb6f5d4c09b3420e9aa89f68c9 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69695914d422452eb1c9d36d9821b985 : SUCCESS in 1m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b199df9b921c4daaaefe1a3235a4eb43 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/66e75e1ea91940b5bcdd8ee70cceeb2a : FAILURE in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/273773e56dca456397df762b719e86dd : FAILURE in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bddad89738db42f389569e5fff6275a6 : FAILURE in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edf9497cc1114007adf5c6d29ce234d2 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55a4205ac33e44cfa4f457c9e6fdd373 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab8bd54186b84feaa89bd6123e745cbe : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f52aaba16944866aa8eaaba9fa5bc8c : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9a7f56bc1dc423a86e4858a8f4ea3de : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f282b084eba7487d9dc6c6d7817c30b5 : SUCCESS in 11m 17s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e9415b30ebc3483c89a5e7efebce3509 : SUCCESS in 24m 55s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/fa3d605178fa4f4c8ef1796264182fbf : SUCCESS in 23m 46s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/85516179b16d4635a510579fd45f3f1d : SUCCESS in 31m 08s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ace6cd3ab3da4b3b845a989d2f06ac48 : SUCCESS in 17m 55s","accounts_in_message":[],"_revision_number":3},{"id":"37c5eebd1c8438f8e2c75a5a56e06c1659fddbfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:32:31.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"f7cf68a9b43c06971ffec260ca4b6dfc589a0190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:40:04.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"97f43ced689d88162c425b8e32c87d61f3ec5650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 16:15:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d597aec384b64c96a1b47fa4921e6881\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d139655d83e34f88882c650bf362beaa : SUCCESS in 1h 28m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d45c66075a74734a63b8bd1d3caa8c5 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da2e62f0ab24a0482d6a8ae08496ba6 : FAILURE in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ba70e75cdae4eb3be2f0249cbb7a988 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f0ed6221b7a49ff80168d68aeea19cd : FAILURE in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/690a8b1c091b4b0599b6dbe1a8bbc2a0 : FAILURE in 2m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5a99de291a6426d8c3861f3a8c9a444 : FAILURE in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c08cad181874971917415f30c4d6007 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa75f554e16c421e95573301a9b62c8d : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bec5652b8aa4cd48cd147667cf94c10 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8b99ce4e9ce47feaca8060b810c1191 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f96fa8ef47f418e84aaea18c7a93e11 : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a80e11092e24a58904eb14316eb5bea : SUCCESS in 9m 18s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/120ca96a4a4d470ea95cc20d784c7646 : FAILURE in 8m 20s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/f329d5e61ee64ee0b9252e09e5c8e2f8 : FAILURE in 10m 50s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/388284a2f39f4876926f45fe124b8240 : FAILURE in 8m 09s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/86c9d4767f334723b518944a974e59c7 : FAILURE in 7m 36s","accounts_in_message":[],"_revision_number":5},{"id":"e00e9014e57d80c33bd9a4b23c5c6ebcea641209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 09:13:15.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8b406cf55176a68be8b9847946c2185ae83852e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 11:00:37.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7548b487451c2229f2ca183e664fd28f777cb598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 12:34:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bf682144593413e82ddda73eeb8770b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0cc39aafecf406fa189743573728acc : SUCCESS in 1h 26m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f95be153bb4cc0bc55f4ef210c9406 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90547f20b1034f609359b3ba5f4d3112 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/973f64500d1c4dbd9de932730d0c6feb : FAILURE in 10m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8a25030cfb004d0cba703b19545ee2ba : FAILURE in 2m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fed2e78dc36245489f255dd605594bcb : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/430cd61f58cd4f4581a5e245ef209c0f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cd5cbc8f0a44fb6baa07f12d57de63c : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0fc38d5711742389a26a46eec3e4a66 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf4d526906424398ae63cf01840bc509 : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/776cf14bedc948f2b96279c0688e1e9f : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a98277357a674d1f9bc408439d453eb5 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0caffb6e0a9c45718dc1b1d342ef8266 : SUCCESS in 11m 55s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/057bf442bc2f4558b3f295d0a92f9f68 : FAILURE in 28m 45s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/115a52b59b904727b87aea750bdaa0f4 : FAILURE in 32m 36s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/655aa8e52cc241d4b2533699673f313a : FAILURE in 32m 13s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/450d5f5a74174c70b2d0d24e2cca6c50 : FAILURE in 45m 04s","accounts_in_message":[],"_revision_number":7},{"id":"8d01143bc77b5390dc27adc02e794ad58508a390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 17:59:08.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"84be64ec47bdff470f25dea511ca02da8d040fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 20:05:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a72e0306fa140eba2ef8a067317495b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5e0f161c6cf4c55a7b01ae4e9e24053 : SUCCESS in 1h 50m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1675f1b5715247e5a1b87341af8641df : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48611e0adb304eee83d5fd578b700db6 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74e5814b03bc480686d55bb11ac0bc33 : FAILURE in 8m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df43fae083a545bcad2db1450598afb9 : FAILURE in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58656f4a09d34230a6fd8066d8c717e2 : FAILURE in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3653e9dcc1d3444ab75760a3d8dabb61 : FAILURE in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21b0527bc86b4cd4880e58eeba53874a : FAILURE in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26b9f5107a76431c99a36daee7c3b5b2 : FAILURE in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e577e89e675423f8afe0a1142384d12 : FAILURE in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c374a8606af44cf6880fc79ddb15deda : FAILURE in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8736d43ba6054e1f9b62b8ae90e4f69d : FAILURE in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a387ffb5c864f71bf353fd175f20c63 : FAILURE in 7m 20s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/6ebd5bf7f75f456f98d58ad0765faab9 : FAILURE in 34m 13s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/2945ff8aab1f444aa396555f12fc2913 : FAILURE in 23m 16s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/225d41dddeeb4a6cbb0b92ff7d1c59dd : FAILURE in 33m 21s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/7cde17cb22f24fd3859bf6e280a691fa : FAILURE in 28m 22s","accounts_in_message":[],"_revision_number":8},{"id":"5f87b1b6f44f82c2c49ec5d155d91c55e2f1d35a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 20:35:45.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"5053f1bbf24ab16ea16f19b0d3381357a15ac9ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 21:37:58.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f79f7360c8047b8a1d8039cb49e1c26\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b65dd87a292942aea3d3f67ec44184c1 : SUCCESS in 59m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d122c5cea240401889d487db4133a539 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f664473c43b47608eba3c249b97032b : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/969324ecb630403bb989df4e494d4472 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f9500bf1f3e482193bd775107fbae16 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31d6ed640e6c443ba7244ce2a0833bcb : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89e988e8c03b4a0b897442f380cd792c : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/651c34cd4aac4de1bbf5352b320a4f0d : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/611d4468a61c43dcaf2ca58448e651ff : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5bc7c8d4d8c4cb4bd98c9ab4e4842ea : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdda49d965d84b268f4e628ee4b9744f : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cf1c14ff225464a9e02e71aa1834871 : SUCCESS in 10m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a9f8e0106754888bced2b47399cf4a8 : SUCCESS in 4m 13s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/a19c5ac0cca94927a081df376eb29741 : FAILURE in 13m 43s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/db4f006c1c2743ed8c74d63369d13be1 : FAILURE in 18m 59s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6e916b633f694ab48ee5ffd8bb29c7f3 : FAILURE in 16m 43s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/5dea7709156843d2ab92a80afcf8c9f7 : FAILURE in 18m 48s","accounts_in_message":[],"_revision_number":9},{"id":"9e56f8c622f6f48219cf7d707f5185a091e19ff7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-05 17:42:22.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"8f9723461dedd3ffc23a6b6959b99c36131e08e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 19:24:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10349ec774304c1daaf9369b778dd21c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2b6f5a7379d44c787bf88e555bf695b : SUCCESS in 1h 32m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d0590d556049b7a872e8f0a3e29049 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/374547e98b224a2290db332b5c4e7c3b : SUCCESS in 3m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/355162390ae642158ed0287cf7a17186 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/565154a4f3ef45c397aad4feef2a67e8 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce890baaa92d4a24bc1573ff2dfcb8e3 : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b47f1b1021b49f98e7907114b3812dc : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70f8a23d620747fba70d8ae86558ed69 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ce19a0ec79249c1b7170cc290e190f3 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf12114cf9424ed2b43656e6d1a9d191 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7221b6deaa16408185556263909c535f : SUCCESS in 6m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/531c0c9ffa244ef99524e8883524cf06 : SUCCESS in 9m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d032e8bf6c164403b05b19ad792738cb : SUCCESS in 9m 26s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/a084e7fcd88a43adb7bba62ef633b911 : SUCCESS in 28m 21s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/fe120310ed42467f9afb1909a0c736cc : SUCCESS in 23m 35s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e18df01bb24a4608bc2fc3dc0405954a : SUCCESS in 38m 54s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/6f38927b24a64adeb29cd6e9acf92571 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":10},{"id":"387f8bbb5a8ce613e30ae04e6e4b227f2059e8e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-07 08:30:36.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"5099d503f5e22f2225485016218f52288b2665f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 09:31:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82ab034233bc4e178004655c80116d36\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b64b58e27d04b27a200ac0d331efa9f : SUCCESS in 57m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fac95ef4fdc437b9dbc50d11680287c : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f937fb8416204f50becef29a15d8d660 : SUCCESS in 2m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2d7da461a0f4552b0e0f0493663b643 : SUCCESS in 9m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6cf4ab7fa5054889a2fcc03dbf0a90a7 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d1e24ed96844081b087a6b00891c929 : SUCCESS in 8m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd73e4b3c3d84ca197a2fd9472ec20ae : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dfc64281e20483eb1748ffdca5931db : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c160f599de9f4995967769571897b40e : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88d3360f60cc452a93f4175f1dafe64e : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8303f720ad6e4ea684df4e650e4ce2c1 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/565eed88c9db4eb19a048da7d220cfca : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a194f7f5397448b58030d5f5647476e6 : SUCCESS in 10m 19s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/0f39c7666c7c46b3a02c7f150d9c558b : SUCCESS in 31m 16s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/c4402870f73249498665e2350361fb66 : SUCCESS in 23m 11s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/92e8816b221241ca8163d6a0ab29bb4a : SUCCESS in 24m 14s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/d1e93a42c1434b37aa09c1dfc0f7210f : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":11},{"id":"e76d7886ec9fdbebbbb1e1ab3839a3ba09daed34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 15:04:16.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"c4a00b8eb96f784add343d98c0386cddd4d87440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:42:54.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0747bd5d1dcf495a856eeb6e207564dc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac6e80c50fe0420a84929ea1ffc25f01 : SUCCESS in 1h 27m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72fd34bf44fe4729ac3ddcb5a65749bc : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504a6e3e77154aada231d20b19e06530 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a34e9064fdba45cc82fdb576b58cfbae : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a26618e5026740be9afcd3670e525193 : FAILURE in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dc0fc4400a8431eb1f8d00d6f1f2c18 : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/448a90de569f4bd49e4949cdef3e4a49 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03fe296995d140dd9333307c5f6d93a5 : SUCCESS in 8m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42939f3ad9284653b21bfbc9477216ec : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9e3d5943ee64a1db9cea39bdd478c1d : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e03641dcbb64644af3b9e12157f63dc : SUCCESS in 6m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/abe104cc010941e88756eb47e7009b5b : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc4f7369f73f4e6fb57ec46a812b270a : SUCCESS in 8m 47s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/113f9f932ada464da901617860f55499 : SUCCESS in 23m 54s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/810bd33671fb4174b1c4bb75b5b16b53 : SUCCESS in 24m 54s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/5f61ebf476134bf5a4a7ecad9d4405df : SUCCESS in 30m 01s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/8d61cf3fa6b94f8687b61888af965043 : SUCCESS in 39m 19s","accounts_in_message":[],"_revision_number":12},{"id":"af27b64bb79b35b5ea1ecb5b8b1da4551bf70888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 17:03:50.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"4b47a05b82e1d1e348dcee8e22f652119d63b119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 18:50:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/257ba7396a704db3920c9600067c86dd\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/447bb0f214da454e8eac3d1a776cbe86 : SUCCESS in 1h 39m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d8847cc059b4c3d9e4cddf5eac075d6 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/badb74ddb6fd477ba5418778f7a8cd6e : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cc76216b4ed460d838cde4074a7efa1 : SUCCESS in 9m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2eda17a0af7645919efe1832a91f8b2f : FAILURE in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c60d2dd920424a0481fa4ea81a71b946 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72a3212fd8a44b0e9bfb9e6e2690f785 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/952200340c0b48b582e0b5a17e6e7a62 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58782c6a939244d6a51af2c3557303cd : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ddef1235498411091c62fd0801604d5 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0221b7ffc4a4b37b2757b5ffbad4d2d : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3d8bfee1928434b8e4915a73253f80e : SUCCESS in 10m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91ab320c2d0e47e9bf4efd2ffea79e4f : SUCCESS in 7m 26s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/5949383c2f5e4d13aeee5f0e1fc0665b : SUCCESS in 22m 10s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/f810f1f7030b4afbbc41b8164d8c3f94 : SUCCESS in 29m 59s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9f98dd25f29440b088e25072fa2b0c75 : SUCCESS in 43m 28s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/281d4f04e7ca49d8afa8e3922ba14ce6 : SUCCESS in 36m 55s","accounts_in_message":[],"_revision_number":13},{"id":"7e41bfa701e0dd965b70c570387cab9c996ed64a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 10:37:26.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"85edc1e60cf810e76f00d2c560e53453d686b17b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 12:32:23.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db9b6a656a604ed584b67d8b710f4d4f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb639a48d5704128a192bfdbd4c4cb9e : SUCCESS in 1h 35m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45482e170f9d4ce9b1fc38f263f5538c : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92696d598dd8418892b39acb77637feb : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbb0edd9773c41cdb7007e7ec7bf0560 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6315a5c4f5a44921b73c63a401f9ba47 : FAILURE in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2c14779b898457e9d05937d63e58e03 : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13036c16459f4ce4982d3c0a2c2cd07d : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c923331a80bc4999acea9f8b7680fd28 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13fba4addfff4fa88b6e60dd03cf40c4 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5f5b1ef0b59404392f817d563f5d1b7 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89348ba442eb48a8bc0c1c568ea33262 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2cb1aebef4640eebcbe17f7c4092545 : SUCCESS in 9m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c4ab9bdcd674434b46b94b4b2eed32d : SUCCESS in 7m 26s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/1d5449db1c444688a02a08ed45ff1cc1 : SUCCESS in 22m 46s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/1c378fdfaec247828571202ea03a4053 : SUCCESS in 24m 22s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/3af9f4795ca84940b454f0a3e0f0e870 : SUCCESS in 33m 12s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/26821e400313488c8effcd359446b79b : SUCCESS in 34m 43s","accounts_in_message":[],"_revision_number":14},{"id":"68e713449b1bbc8b9b2df45ac9da163fe44fab79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 12:46:27.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"04cde89ff9d6e316541d71a6f7aa6e23ee93c197","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:31:47.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30c4c3ff3e8a4101808bab0956511c5a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a26319dcd46748368d470c1d07983d67 : SUCCESS in 1h 35m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/074787db717b452bba6073338f16a9ed : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de9c8fc40e5f43598f1f01eb25ec539b : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10e124265c314286819b5f615e54058c : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69e78fc01fe54931b5a93559984b5ab6 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b11288396744ee78eeb147864dd8f32 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55e4abbbfe494495b1170b389905453a : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/718ffb78185f49acae3775df336fbb81 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/effdc5ca86ef4a25a61ad8186a5cdee1 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d0a9d5c2f9844a48d5ebc00f82019ce : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d21deb54b15b4b85b51cb87b8fc5697a : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6c025809b6e47b58523ac2dc49c67c4 : SUCCESS in 5m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e7787b94b7d486d8f8381255940b8c9 : SUCCESS in 7m 38s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/fc40f116550044ac96a9ab7f841f412e : SUCCESS in 30m 09s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/1a2b1dd5d5394307942411af3f87755a : SUCCESS in 29m 52s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9f7377869a6f40aab1b97e619765de42 : SUCCESS in 33m 45s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/3fb95fbbaa704153bea28da3054aac12 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":15},{"id":"1d5257ea474c62eb47f2bf53552db55398bd991b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-16 16:15:46.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"ee995d0ce9c530722a7300016fdc3c087ab5e965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-17 09:32:27.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"650faeea68eb133530026480fe08b0826fe89aad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-17 10:13:39.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":17},{"id":"7fe30295e97803fb05d1fce803644210c14ca630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 11:59:20.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b831074d3d9a4dbc829e54bd8dcfe8f9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/018ffca884164e2b81d38db51ff10241 : SUCCESS in 1h 40m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d95228971024b99acc2c6658e443527 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3367f85015d48e799c1af6e1452f5a2 : SUCCESS in 2m 45s\n- build-python-release https://zuul.opendev.org/t/openstack/build/d5b5eb36784c494fbed379ca29d251d4 : SUCCESS in 1m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b65edb0f317475baf1eb543e48898f6 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/db14f12b16f4430ca47b77e23aabfec1 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83b96e21a2fa40fdbcdebd94e68c949c : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1049cab8484f4320ad7da72899274143 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd84bc0e5e1141f48305908c40d6f730 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef1e64e820a94365a0d51f23cb7e6224 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5822a910509c4aaca0bb52ee3992fd10 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c65b27f060c4830a39f6e3e65a420c0 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65ed7cd058fc4d94b979d05717b3c447 : SUCCESS in 9m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ef85a0c425b4479a71523eb4d47bd93 : SUCCESS in 7m 17s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/cc485df84e6a445990ddcf9f7cb6bd14 : SUCCESS in 14m 46s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/8230475bb5154061b93e384c38c92e0d : SUCCESS in 30m 24s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/8485910cb3174310bb4a21f1c8b7046d : SUCCESS in 28m 54s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/353667317b0b49f2b371658fcd70c9ef : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":17},{"id":"d88506df4072a695ebc92e51a4244e57efaf7746","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-07-17 12:26:44.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"f187e908652f74d0b8f72e715cec00a3b90571ea","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-01 16:46:30.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"292ab5fc51efc7496b6d392c9b977e12b992fddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 18:05:34.000000000","message":"Patch Set 17: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/954050 is needed.","accounts_in_message":[],"_revision_number":17},{"id":"d1c44e25cf52713e8b74914cae28976b4baa3798","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-04 20:05:18.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a115012c133edf71e349f285c15b2b2d2796fb5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 21:53:10.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fe04f2a17f64bb0afb2c5830d4d06f4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0f2e4bad9614bd8b3d668fffca5d330 : SUCCESS in 1h 41m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deae2c9dd664416692fe1386695e8d07 : SUCCESS in 1m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17a4a9c71d5b45d9ab79626c59132b7a : SUCCESS in 4m 21s\n- build-python-release https://zuul.opendev.org/t/openstack/build/0233b7ac89014b4e918edf2f7403e33d : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9901dfa528e846269332214e5ddcde43 : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/27916378549d463a81d0c8f99820f020 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f24ff32e4f3b46859732f3e9c3100ff3 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41d3c5f9e06347ce925e6bcdbe4de62f : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0603d2001d4c4e6b9d6a87b290b1b05b : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b78ea731a6b4897958d3eea61d39a07 : SUCCESS in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29bff3eb45b34ebd9f17c9cd91c46422 : SUCCESS in 8m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ea7b6365b7c49b58e0f575759dfac0d : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/495775b01b424524b0855671e041924a : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f2fc2563ea54b3e872590d78ecf996e : SUCCESS in 9m 22s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/1ac0d64ca862426d8a104412127bb924 : SUCCESS in 25m 58s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/cf4b1eab935b4e50a760af72c8acc9e0 : SUCCESS in 24m 27s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/c0dca54f2faf49a185eb788cb438f43d : SUCCESS in 40m 14s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/793f253aa7fa4161b7e760ed5513cc11 : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":17},{"id":"2da6cb6843e0d372295fa5f5e549712a2b87dde2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 22:07:04.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"c53d1221f6e390d10ff91fdd4ac899e439900f95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:54:45.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607140a7e49a40e2be8d25d221d71b5d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce13ba74abbe4435a48af85577bf4695 : SUCCESS in 1h 43m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c87f43e5fc425392746ffbc5f3d813 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b2372a8b26e454ca70ceb68bdadb95a : SUCCESS in 2m 38s\n- build-python-release https://zuul.opendev.org/t/openstack/build/17c8605741504055a99e4340a4f4773c : SUCCESS in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/057ffaaec3244105afc0dd79a117e925 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3ff52c3e454466aab11b053ede53b43 : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f186f6399d34926a1b905ae29cc2da8 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3841feb328fc4674bb77404f2fac7c4a : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64a586dc51cb41f689ac6d66c57ea2be : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d348f793dc040e09d4c7f77c7680216 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba65e5e3837f4c93903d7527206d9ca1 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4004385569e94a07a61d188aac0664da : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78c7afd4f089441da3e3913813b63730 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01786b3e3e264f32bb1b84f2409765b8 : SUCCESS in 7m 41s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/949b4eded79d451496a713ba5536ad2c : SUCCESS in 25m 14s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/0ded1545fea64f768c952284f63b82fe : SUCCESS in 31m 08s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/dd07f966fb844d47ac29e1148e8d2650 : SUCCESS in 34m 36s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/667e85ee87c64f27b4cc17c440769903 : SUCCESS in 31m 06s","accounts_in_message":[],"_revision_number":17},{"id":"75865d42c27fa081f1075b60008d65800944099c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:54:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"1ed80ed617cf99b5a4850fc1545be24c76a31f93","revisions":{"ecd5130e64b5200ba63b6b482eab03fd1234f421":{"kind":"REWORK","_number":1,"created":"2025-05-08 13:54:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/1"}}},"commit":{"parents":[{"commit":"7aff914dbee0c594283fc4233bef93aa3fe34e89","subject":"docs: Remove use of sphinxcontrib-apidoc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7aff914dbee0c594283fc4233bef93aa3fe34e89"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:49:25.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ecd5130e64b5200ba63b6b482eab03fd1234f421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ecd5130e64b5200ba63b6b482eab03fd1234f421"}]},"branch":"refs/heads/master"},"3514b1c650605b7789765c969924794bacaa4310":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-03 09:38:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/2"}}},"commit":{"parents":[{"commit":"717bf51bb6c04ba7abdc79e903392b56a648610f","subject":"docs: Remove use of sphinxcontrib-apidoc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/717bf51bb6c04ba7abdc79e903392b56a648610f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 09:38:23.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3514b1c650605b7789765c969924794bacaa4310"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3514b1c650605b7789765c969924794bacaa4310"}]},"branch":"refs/heads/master"},"db65d0b5fa09f7f66490def0c55d3c28bc4232a7":{"kind":"NO_CHANGE","_number":3,"created":"2025-07-03 10:30:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/3"}}},"commit":{"parents":[{"commit":"0c1fb38317974b824bc0c22dcae569552b874c81","subject":"docs: Remove use of sphinxcontrib-apidoc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0c1fb38317974b824bc0c22dcae569552b874c81"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 10:30:09.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/db65d0b5fa09f7f66490def0c55d3c28bc4232a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/db65d0b5fa09f7f66490def0c55d3c28bc4232a7"}]},"branch":"refs/heads/master"},"abb57b6436403f0d0febe2622daac4003dfba44c":{"kind":"REWORK","_number":4,"created":"2025-07-03 14:32:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/4"}}},"commit":{"parents":[{"commit":"0c1fb38317974b824bc0c22dcae569552b874c81","subject":"docs: Remove use of sphinxcontrib-apidoc","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0c1fb38317974b824bc0c22dcae569552b874c81"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:32:31.000000000","tz":0},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/abb57b6436403f0d0febe2622daac4003dfba44c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/abb57b6436403f0d0febe2622daac4003dfba44c"}]},"branch":"refs/heads/master"},"ccc0e3602d2630be72a7275c7a67993c00ae4904":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-07-03 14:40:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/5"}}},"commit":{"parents":[{"commit":"ab88857d534c9df5124f6ba8fc398471a3ed350a","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ab88857d534c9df5124f6ba8fc398471a3ed350a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:39:55.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ccc0e3602d2630be72a7275c7a67993c00ae4904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ccc0e3602d2630be72a7275c7a67993c00ae4904"}]},"branch":"refs/heads/master"},"21b04b1e5763e140e61e0dce4a68f0b95fb7dfc8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-07-04 09:13:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/6"}}},"commit":{"parents":[{"commit":"558de97cdbad88e35505db4315bcb15c83470d5a","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/558de97cdbad88e35505db4315bcb15c83470d5a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 09:12:21.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/21b04b1e5763e140e61e0dce4a68f0b95fb7dfc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/21b04b1e5763e140e61e0dce4a68f0b95fb7dfc8"}]},"branch":"refs/heads/master"},"d8411a791eea359f7a724252ebbc907f89abf80c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-07-04 11:00:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/7"}}},"commit":{"parents":[{"commit":"c9b9c71fad6a8d3044f4a02a74d255ddb89fdf3b","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c9b9c71fad6a8d3044f4a02a74d255ddb89fdf3b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 11:00:30.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d8411a791eea359f7a724252ebbc907f89abf80c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d8411a791eea359f7a724252ebbc907f89abf80c"}]},"branch":"refs/heads/master"},"17254a05273a3a77210be4061026366bcfe940ae":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-07-04 17:59:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/8"}}},"commit":{"parents":[{"commit":"a2666d7ca8027dfb450f4ed7716d451a1015db06","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a2666d7ca8027dfb450f4ed7716d451a1015db06"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 17:59:02.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/17254a05273a3a77210be4061026366bcfe940ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/17254a05273a3a77210be4061026366bcfe940ae"}]},"branch":"refs/heads/master"},"388eb70229ace2dfd8c4675d4ba4f940cab8aa4e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-07-04 20:35:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/9"}}},"commit":{"parents":[{"commit":"e49830c38c76250913d64c888044831f79d548f1","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e49830c38c76250913d64c888044831f79d548f1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 20:28:47.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/388eb70229ace2dfd8c4675d4ba4f940cab8aa4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/388eb70229ace2dfd8c4675d4ba4f940cab8aa4e"}]},"branch":"refs/heads/master"},"1f8f779ca70a4e2baf24d501141a6cbe1cc3b472":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-07-05 17:42:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/10"}}},"commit":{"parents":[{"commit":"53e4695b0a9ef18d552ce817f9364fc6e0d82540","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/53e4695b0a9ef18d552ce817f9364fc6e0d82540"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-05 17:42:04.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1f8f779ca70a4e2baf24d501141a6cbe1cc3b472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1f8f779ca70a4e2baf24d501141a6cbe1cc3b472"}]},"branch":"refs/heads/master"},"7827aa36c5b1d6c801ec638ee72eb44f34c21900":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-07-07 08:30:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/11"}}},"commit":{"parents":[{"commit":"600e18c68d9cf1caabdf5088fccc30cb51b2092a","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/600e18c68d9cf1caabdf5088fccc30cb51b2092a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-07 08:30:29.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7827aa36c5b1d6c801ec638ee72eb44f34c21900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7827aa36c5b1d6c801ec638ee72eb44f34c21900"}]},"branch":"refs/heads/master"},"8eefce99f5ae676c95db908d37f552782c3e0c19":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-07-09 15:04:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/12"}}},"commit":{"parents":[{"commit":"6eee0b21cc6df0f1cd216c184049771069e47f56","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6eee0b21cc6df0f1cd216c184049771069e47f56"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 14:27:31.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8eefce99f5ae676c95db908d37f552782c3e0c19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8eefce99f5ae676c95db908d37f552782c3e0c19"}]},"branch":"refs/heads/master"},"673e1930872101a0dd9b038a5466b45227d52241":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-07-09 17:03:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/13"}}},"commit":{"parents":[{"commit":"f0b47db068f07affd36050f1f5ad40bcdbfa6e9a","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f0b47db068f07affd36050f1f5ad40bcdbfa6e9a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 17:03:38.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/673e1930872101a0dd9b038a5466b45227d52241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/673e1930872101a0dd9b038a5466b45227d52241"}]},"branch":"refs/heads/master"},"204f913a2ba3b18c32b55899be50c359616f83bc":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-07-10 10:37:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/14"}}},"commit":{"parents":[{"commit":"96597ad35b04cec102f1a94a4b0d0c1acd1f8a8f","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/96597ad35b04cec102f1a94a4b0d0c1acd1f8a8f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 10:35:29.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/204f913a2ba3b18c32b55899be50c359616f83bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/204f913a2ba3b18c32b55899be50c359616f83bc"}]},"branch":"refs/heads/master"},"4f23b1e5bee14e92eb78a885c3eca17b4b5e65b1":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-07-10 12:46:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/15"}}},"commit":{"parents":[{"commit":"fdde4e9b6d3d8eccbe05bc1b72b1d89289e0f06b","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fdde4e9b6d3d8eccbe05bc1b72b1d89289e0f06b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 12:41:23.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4f23b1e5bee14e92eb78a885c3eca17b4b5e65b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4f23b1e5bee14e92eb78a885c3eca17b4b5e65b1"}]},"branch":"refs/heads/master"},"4d167910f80b6ba1a9d4bcd9fd7e57602711d43a":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-07-17 09:32:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/16"}}},"commit":{"parents":[{"commit":"2e9c911e12ca3fd0b0ce14d51a5abcc2d212425c","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2e9c911e12ca3fd0b0ce14d51a5abcc2d212425c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-17 09:32:05.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4d167910f80b6ba1a9d4bcd9fd7e57602711d43a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4d167910f80b6ba1a9d4bcd9fd7e57602711d43a"}]},"branch":"refs/heads/master"},"1ed80ed617cf99b5a4850fc1545be24c76a31f93":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2025-07-17 10:13:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/949188/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/88/949188/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/88/949188/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/88/949188/17"}}},"commit":{"parents":[{"commit":"e177eaeb2314455273129d38a96fa29be8fc5562","subject":"Deprecate \u0027tests_require\u0027 option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e177eaeb2314455273129d38a96fa29be8fc5562"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-05-08 13:38:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-17 10:12:02.000000000","tz":60},"subject":"tox: Stop calling \u0027setup.py\u0027","message":"tox: Stop calling \u0027setup.py\u0027\n\nsetuptools doesn\u0027t like this and is removing it in October 2025 (per the\nwarning issued). We need to start using PEP-517, which means we lose all\nof our custom distutils commands like the reno one. Alas, such is life.\n\nChange-Id: I358e2f5a016ca81dc2f3424de8cb5f090106b0c7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1ed80ed617cf99b5a4850fc1545be24c76a31f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1ed80ed617cf99b5a4850fc1545be24c76a31f93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
