)]}'
{"id":"openstack%2Fpbr~951379","triplet_id":"openstack%2Fpbr~master~Id26c14c5db9b50ea8111a7fd75ea8b7a6808341c","project":"openstack/pbr","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-03 02:31:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id26c14c5db9b50ea8111a7fd75ea8b7a6808341c","subject":"Add pyproject.toml to sample package","status":"ABANDONED","created":"2025-05-30 08:26:53.000000000","updated":"2025-07-03 02:31:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e82b965aed57e18d64817b24b3d2735c1e71d1a3","_number":951379,"virtual_id_number":951379,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-30 09:57:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-2,"date":"2025-05-30 08:28:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 09:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4cb869e31b0bc52991969f9242c1cb600526c75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-30 08:26:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb97d072ba4a58cdd6d6ed3b7aa3b55566398e45","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-30 08:28:36.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"3b079362d420983e495b273c957581767da4089a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 09:57:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/739a7020f4694bbaa77d27d8dd70e3f7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c8a23e6aca54d72bbad1562ffa07620 : SUCCESS in 1h 27m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ed10243b004369a7bca7ee5ee3b438 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21dbb629ec4945a8903249bde54dbf32 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10e1b846e1b1413bb23a5479da485c49 : FAILURE in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b7536dacbe164683aa3f77845f4b110d : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a38e21e6b20467eacf236cfee982945 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61c40445462f48c3ad51b0763e6a56f6 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53c46fe7124942b580f5880e9d5d2a4d : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b58062c1deb4e32b21163f16a042f82 : FAILURE in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ad0ba4e6f004a6aaa867254d1a6b9d9 : FAILURE in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bca746e7a713417bbaf88848bb20797d : FAILURE in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c8eb4b17ecd74b5ca0b137390e3b6af8 : FAILURE in 10m 28s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e710bdfb75c841adb4b7e5ab1a32902e : FAILURE in 25m 30s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b38bdc35a8c44493bbc6a634e8a82014 : FAILURE in 31m 21s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/58d04ac0b2a64962b5428a4edcaa547e : FAILURE in 18m 49s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/34f557fd1e5a4781a5740e33a7a13fb4 : FAILURE in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e82b965aed57e18d64817b24b3d2735c1e71d1a3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-03 02:31:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46e9599e5432342431ac509188c0fabc71646081","revisions":{"46e9599e5432342431ac509188c0fabc71646081":{"kind":"REWORK","_number":1,"created":"2025-05-30 08:26:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/79/951379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/79/951379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/79/951379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/79/951379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/79/951379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/79/951379/1"}}},"commit":{"parents":[{"commit":"5d4a1815afa920cf20e889be20617105446f7ce2","subject":"docs: Drop SetupTools from pyproject.toml example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5d4a1815afa920cf20e889be20617105446f7ce2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-30 08:25:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-30 08:26:51.000000000","tz":540},"subject":"Add pyproject.toml to sample package","message":"Add pyproject.toml to sample package\n\nWIP\n\nChange-Id: Id26c14c5db9b50ea8111a7fd75ea8b7a6808341c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46e9599e5432342431ac509188c0fabc71646081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/46e9599e5432342431ac509188c0fabc71646081"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
