)]}'
{"id":"openstack%2Fpbr~953893","triplet_id":"openstack%2Fpbr~master~I8f9935721d56ed9b778d58590d198f9352086da9","project":"openstack/pbr","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-02 20:27:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f9935721d56ed9b778d58590d198f9352086da9","subject":"pre-commit: Bump versions","status":"MERGED","created":"2025-07-01 15:58:19.000000000","updated":"2025-07-03 09:35:53.000000000","submitted":"2025-07-02 20:27:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"953893-py313","meta_rev_id":"b037a1b9106847c97843b83469a446751c81bc1b","_number":953893,"virtual_id_number":953893,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-02 20:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-07-02 18:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-02 15:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-02 18:37:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-01 16:31:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-02 15:05:21.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-07-02 17:30:09.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2025-07-02 18:18:53.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-07-03 09:35:53.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"}],"messages":[{"id":"c0eca0caf53f61c0ab64033d32ba4ed0af5f5dae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 15:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa84b4524f86dec3b6bce009b10b9c9051d97f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 16:31:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cac16e8cf5c44a1b0dcd2bb3740daee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad14943091b4e509326b05b4373b7de : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b328553bf459450cada5a0ab117b772c : SUCCESS in 4m 43s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/6095da9868c2435989be92136b4104b3 : SUCCESS in 14m 05s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/bb68ed1517cd4a13b0b32db1e176ae1b : SUCCESS in 28m 36s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/759f834fe68f44eda615d891de6c703e : FAILURE in 6m 47s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/77b4ecff192342a98fa654fe96d8c200 : FAILURE in 21m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8ab2fbdcd5c5dbcf5645cadb715b84023115cfab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 17:24:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"916df09cc05dd7660a92a2ca799b6143cc1f72db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 18:07:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab455691a55f4f7a870296a46b785bdf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/337dae84ada24c95bb84e85ffa7628d6 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f21c040df3c0484a861f5e8745311113 : SUCCESS in 4m 46s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/2d52fc7b576b4a91884b9bfee00d1bee : SUCCESS in 25m 02s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/48e60ea5da804ed8b64a7fcbafed491c : SUCCESS in 29m 58s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/65aaed4342b04f9e8247b21a0f240161 : FAILURE in 16m 57s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/637ff21bc6d243dfaa04d7a4fac998fa : FAILURE in 19m 48s","accounts_in_message":[],"_revision_number":2},{"id":"6255007f1f6bf138e002571afeba2257200578c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-01 21:58:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3433b4f8cbf0337f64ac8f7d685b33f93b434c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-01 22:39:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0619de54675c4a2088c17a586e9f6e4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50ad91232df4f6e8805572a0802eadc : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba93879d30c74495ac4d8aff6370a855 : SUCCESS in 3m 00s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/ae0d341b8e9f481d8940c04d051a3172 : SUCCESS in 28m 49s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/7e1b81ffcab347bcba1f3fed86c34454 : SUCCESS in 20m 04s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0cd5d05b86014a43adc1c0af7b3f6ea3 : SUCCESS in 22m 49s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/265486de5b5341738856747f9861ef40 : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":3},{"id":"6118ac4fd9c00df5c27484bea1179b8be58a2ad3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-02 15:05:21.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aee03f15e7ab25949e18a199794aa81c4342accf","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-02 17:30:09.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d5ffbfa96deaeb74a06223d64de1bd061f117554","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-02 18:18:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb7627545162c436121486daf9079f016020849e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-02 18:37:01.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2f9dae238a6b7e52c809f48f30b72f7b06b1c724","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:38:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c90535aad1317cc82d7f23042de925522a09a8d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:27:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d42b9928b3dd4feba316577a75ff34d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d664d0a7bc7460f8a9cf702b1e945f1 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/209e26b76d354de4a022c35efa42457a : SUCCESS in 3m 06s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/b9afe4256f7d4edda866e69342e3cf65 : SUCCESS in 23m 05s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/6b22c1a640c8431ca19fe41e03c1415a : SUCCESS in 25m 03s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/8b2c9a049e7b43ae98467d8e6fa7d7b8 : SUCCESS in 41m 14s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/9e2ee7fb5aff4044944663b96b057ddd : SUCCESS in 32m 37s","accounts_in_message":[],"_revision_number":3},{"id":"82e45de3f5c936ad4345c307634f9551d3c7b354","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:27:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"96850cdc00f5ff7e7c22d60d9488d6123938fdcd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:29:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90555dde73ad4d2391abd8fb92daafd1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaf3b97e761744be8de00e5a0b4e3d6c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3},{"id":"b037a1b9106847c97843b83469a446751c81bc1b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 09:35:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9956df0d19dc3dab8a0f855e7251f2fc1fb24b29","revisions":{"d9efb9b8dc314337662d444083c91badd68c3813":{"kind":"REWORK","_number":1,"created":"2025-07-01 15:58:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/953893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/93/953893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/93/953893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/93/953893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/93/953893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/93/953893/1"}}},"commit":{"parents":[{"commit":"fd5d5866a3e9eac5e62e684b77871298ff1dadfa","subject":"Add openstack-tox-py313 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fd5d5866a3e9eac5e62e684b77871298ff1dadfa"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 15:58:12.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 15:58:12.000000000","tz":60},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nSwitch to git tags and the native hacking plugin.\n\nChange-Id: I8f9935721d56ed9b778d58590d198f9352086da9\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d9efb9b8dc314337662d444083c91badd68c3813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d9efb9b8dc314337662d444083c91badd68c3813"}]},"branch":"refs/heads/master"},"e2dafcbf8ded7e5d3617c9da270294aa050ae2c5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-01 17:24:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/953893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/93/953893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/93/953893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/93/953893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/93/953893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/93/953893/2"}}},"commit":{"parents":[{"commit":"496bacc7f654b1e4102805df6ff3e98649d277df","subject":"Add openstack-tox-py313 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/496bacc7f654b1e4102805df6ff3e98649d277df"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 15:58:12.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 17:23:55.000000000","tz":60},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nSwitch to git tags and the native hacking plugin.\n\nChange-Id: I8f9935721d56ed9b778d58590d198f9352086da9\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e2dafcbf8ded7e5d3617c9da270294aa050ae2c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e2dafcbf8ded7e5d3617c9da270294aa050ae2c5"}]},"branch":"refs/heads/master"},"9956df0d19dc3dab8a0f855e7251f2fc1fb24b29":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-07-01 21:58:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/953893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/93/953893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/93/953893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/93/953893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/93/953893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/93/953893/3"}}},"commit":{"parents":[{"commit":"141d0eac9e10aea033f1f6fe4d255ddac28ca3c4","subject":"Add openstack-tox-py313 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/141d0eac9e10aea033f1f6fe4d255ddac28ca3c4"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 15:58:12.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-01 21:58:47.000000000","tz":60},"subject":"pre-commit: Bump versions","message":"pre-commit: Bump versions\n\nSwitch to git tags and the native hacking plugin.\n\nChange-Id: I8f9935721d56ed9b778d58590d198f9352086da9\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9956df0d19dc3dab8a0f855e7251f2fc1fb24b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9956df0d19dc3dab8a0f855e7251f2fc1fb24b29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
