)]}'
{"id":"openstack%2Fpbr~953982","triplet_id":"openstack%2Fpbr~master~Ic6ee03da59f5e99d1c52548c1414e7559de7b300","project":"openstack/pbr","branch":"master","topic":"py313","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-02 21:12:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6ee03da59f5e99d1c52548c1414e7559de7b300","subject":"Don\u0027t pin setuptools for all tests","status":"MERGED","created":"2025-07-02 18:21:03.000000000","updated":"2025-07-02 21:13:37.000000000","submitted":"2025-07-02 21:12:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953982-py313","meta_rev_id":"44b7d356b54bf5a2bf35d2bb464801fb7e4c77ac","_number":953982,"virtual_id_number":953982,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-02 21:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-02 18:36:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-02 18:36:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-02 18:36:11.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-07-02 20:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35c49f88a173c97484853031ef44f22fb71db94f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-02 18:21:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2ed102fd8592825da0d042c8a2f1a39a11f96a5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-02 18:36:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d70206ee48f061d181c8bea1699d1b5fd1c28501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:09:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f401db4ec1d54752a095741991797a6f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8961819d53a1494c8b4642e74b04f007 : SUCCESS in 1h 45m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aea2819e3ed4bf2b7e67340ca6b9108 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194abe8f334941d0835201ad1f07115d : SUCCESS in 1m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/330884315d5648fd83a3cc3b5e0861ac : SUCCESS in 8m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36900baf9bf5460492718857f19aed00 : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0df9ba9a55aa48f49c30b4c7f249669e : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/940528bcf3df43ba93a040344136fbb5 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c19a29c1c28c4cbba195274b89deadee : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bdb28247f9a467f9a10022ad3b6f208 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ee15c8e01784555a8c2a656368c390f : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f1c25cf14cf4fe7b2b00f8e7c410a97 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/209551e7a1854c9bb91d8c33e5a54a1d : SUCCESS in 2m 38s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e73efa4ebb6045c68ff730bca59cf622 : SUCCESS in 31m 03s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/2ce9416f41be4a9dbbd101fa54f8b2d0 : SUCCESS in 13m 38s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/f746b44b2a3c44af81d159d9d37fa4e3 : SUCCESS in 33m 41s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c39994005ab14eda801517fcabc9265c : SUCCESS in 17m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1f0fc4382a87644f442e291c51a8142b8deaef35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:11:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"454e4c079d1ab445ee94dbaf802e4e82344627e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 21:12:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c5756eacadd439da5c90959e74ff26d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec55b329a03e41e5a0e8ad9906a3c787 : SUCCESS in 1h 01m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac08b6fc0f149b4b482de34866c403e : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c7a8511a984ed0a04a55c76d66f4e7 : SUCCESS in 4m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3368e019d4dd4b689687eccb2c6af19a : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d5ce5c9001847bf8fe9f6c30908845d : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68cd122aa7f942db848b34af8e3b09bc : SUCCESS in 2m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9bcba8523c94dab9c780d636850a76f : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11290174f9d542d688a4ae6bd2d3b982 : SUCCESS in 2m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0366f2ceda34b659906a78fe757ba78 : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61d4abbdb9714236804345a114ba7586 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a263e8c079041f589070ea8d8b37e9f : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c33df0a643fe4071935ab4bbeb417e9b : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/142cbd347c224591a92cd8dad2c9a0f2 : SUCCESS in 8m 42s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/dc1445c49c1746da99a4eacf21d41e71 : SUCCESS in 25m 36s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/83ed805082f6497da3865d71ff86d968 : SUCCESS in 33m 54s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e5d25f496e8e41c786b5557e2a4b8e72 : SUCCESS in 33m 20s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/6af8cc0cc2364cb3a096f51bb102ffaa : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":1},{"id":"fd61e21f52fa41c89b489496701e000022695a8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 21:12:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"44b7d356b54bf5a2bf35d2bb464801fb7e4c77ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 21:13:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86e1a48c7b1843418b42b7f8cf2802aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7d74e0c0eeb436f996372f3274d3b36 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c74b961539228a0825f55a3e648eff95dd04dcac","revisions":{"c74b961539228a0825f55a3e648eff95dd04dcac":{"kind":"REWORK","_number":1,"created":"2025-07-02 18:21:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/953982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/82/953982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/82/953982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/82/953982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/82/953982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/82/953982/1"}}},"commit":{"parents":[{"commit":"3f9c3d44883acc5be86d68a8195c0be2139b58ce","subject":"tests: Skip tests that break on newer setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3f9c3d44883acc5be86d68a8195c0be2139b58ce"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-02 18:20:49.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-02 18:20:52.000000000","tz":60},"subject":"Don\u0027t pin setuptools for all tests","message":"Don\u0027t pin setuptools for all tests\n\nAddress nits from I599cb05706753c89f0ba74ee17446e62abd79b5d\n\nChange-Id: Ic6ee03da59f5e99d1c52548c1414e7559de7b300\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c74b961539228a0825f55a3e648eff95dd04dcac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c74b961539228a0825f55a3e648eff95dd04dcac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
