)]}'
{"id":"openstack%2Fpbr~954042","triplet_id":"openstack%2Fpbr~master~I7006197c9799520bf79cbffd5e2bbd6f2152f6de","project":"openstack/pbr","branch":"master","topic":"pkg_resources-removal","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-07-10 17:55:58.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-11 13:36:34.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-11 13:36:34.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-10 18:54:51.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7006197c9799520bf79cbffd5e2bbd6f2152f6de","subject":"Use importlib.metadata where possible","status":"MERGED","created":"2025-07-03 13:49:32.000000000","updated":"2025-07-11 13:38:08.000000000","submitted":"2025-07-11 13:36:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954042-pkg_resources-removal","meta_rev_id":"80d6839fe41fe6ec88eceee2dcb47b1a981a86d0","_number":954042,"virtual_id_number":954042,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-11 13:36:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-07-10 17:55:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-10 18:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-10 18:54:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-03 15:47:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-03 19:44:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-07-08 16:51:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-07-08 21:43:51.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"e418828655a95cbbf8006224bf43d9248e032c53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 13:49:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3b77971df3c0beb2c648eebcda2339ea4a84463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-03 14:40:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9275af93a616afc8ca857bed12a95601d060c063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 15:47:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab92db19297f46c39727f13619d491f8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f6ea5d1515d41d59b9940fc7b2fc1ec : SUCCESS in 55m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/564a12dee2df4e18bcc45246053e8fde : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/792af76e97ae41bc9a18e3e869c9d521 : SUCCESS in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb7e8492ad484c598928a8f1607ba650 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa5a4fa7b7d4401982932cbb8b8cd1c4 : FAILURE in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84241ba23e6349c4901d75c4e9e28afc : FAILURE in 2m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b3eb3637a444c2a8700cb4f060b79e3 : FAILURE in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d8e65704aee4a7791af19c0e487b7b9 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8494f89b4dcc43c99a018a745bc5fc1b : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6efbe158a2bd40fdab9865164233518f : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/768175a2560b4a10940bbe9df8d77e72 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e959a90290d943bd968cf7f14261c23d : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d10d1f7b9584cb3acaa769e7ebe3f43 : SUCCESS in 6m 55s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/97d1e185117a45f49e06717efbf70c22 : SUCCESS in 34m 58s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/8e608cfa99104d4fb43575f282bf0cd5 : SUCCESS in 27m 48s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0cf02b521e7940ae981a3ec7be8a2c65 : SUCCESS in 40m 39s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/48bd1b8244594af6a6dd64bd2acfa11e : SUCCESS in 17m 12s","accounts_in_message":[],"_revision_number":2},{"id":"0c87b44ca640df716bab69c292a86471f945ba61","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-03 19:44:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"677f6277ac3da5a793b1f9eaab1ff892c304ed43","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 09:10:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eaa79ef0f8bbf70ed56745ec4849d8f591811458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-04 09:13:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b7a227e194513f16b6da12fc445560bc8f4edd0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 10:20:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fcf195d2f1b49b68b60e163573350ba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ef2100e4c12499d82872238850cf025 : SUCCESS in 1h 04m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9a8f1d75f64e1d9d2acd15738537ec : SUCCESS in 1m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/022185ba277d48dd90b4cdd24f67d98f : SUCCESS in 1m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ad6208269eb4a4c934f4e1f3b3dd3c6 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc7299d6d9df4e89896c3624eb14c07d : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5031a78596542b5a3be4cb12a5eb1bb : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d02e9b3b43c5408d9866f8d6eff8c3d3 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ed3560b3ff9478e9d8b51d57eb5dcca : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e0c1440eca5473190e068ffa1791520 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5aa3fa9e89204808b6f814b780124900 : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97cf1e3bd6bc4a7f8f6084a6519121b1 : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4abe952e31684cd48c93352616b2b76d : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7249aa49991749d5a23b7d7d0f00885e : SUCCESS in 4m 10s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/62e62ffc75524570ac8f395c0d6ff442 : SUCCESS in 18m 27s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/1bf52c4bdacb4fa6908f2a7759d3fa11 : SUCCESS in 18m 54s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e3227d47b7994a7d9ff330e946e86549 : SUCCESS in 20m 52s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/f8322002b6e44b63b19f810fedc1ff7c : SUCCESS in 22m 46s","accounts_in_message":[],"_revision_number":3},{"id":"04277c144ea660bbe512926611308331eafe0f21","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-08 16:51:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"274966e54e8afdadee4556406c49924c10398152","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-08 16:55:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2a586c8c7733fdfd98b151746d14fc29542a6849","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-08 21:43:51.000000000","message":"Patch Set 3: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cad57a7b15a96440fb41fa47212c08ebd2b39bc7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 13:46:53.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"12f7f76094e141f8dffa76b0b744b1012ad71c1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 15:04:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5e22793a3025726c8747e1fd4cfb37d3b79771cb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-09 15:08:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"939172850a3e152adcc3a1089f305f46b56685b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:51:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0e362cf6b844ec288becce28675e547\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fb8d7dfe87e491188534a279d5fc4b0 : SUCCESS in 1h 40m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5afafec7befa43289a410036dfc5a60c : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f7210b2c1e43e4a7204f52f8c6bf9c : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3799ef3cd34e4d52a09ba7e00fb7421b : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a571030aaa094bfcb7596670e3df3336 : FAILURE in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/486c00b3f9264525a09db7b79b3a2b7f : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc628f7a819e4932a046d93d1e7084db : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a92c5b372e9342c2bd29ff5368a84f61 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f61e067f453e46c087e6a9a3ffb0bf31 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1a3ceb3677c46fbaf4bf24f4618b7a9 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fcc43f48c7143d9b8245eeef712b65a : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5382819c75a34f7eaae11b8532c7bec3 : SUCCESS in 8m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2d2394308a543e7b059829bdec4bfbf : SUCCESS in 10m 02s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/5dbfd3b1c20b49889aa18d254c3c8726 : SUCCESS in 25m 38s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/09bc97d766b24d75ab02df061d3d9cbb : SUCCESS in 28m 30s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/963d51267f77409396fcfca936c194f2 : SUCCESS in 32m 07s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a8d98c620a454a33a0020c32fa7e3c93 : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":4},{"id":"69323c602c911894b81825a99a88afd307bc596c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-09 17:03:50.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"87b08d89951cf14eb0d1c95b774477c0d9d148a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 19:10:00.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/625298be9ab640488f71091ef8c31b77\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55895ab21ff74ca2a761fddaec05d467 : SUCCESS in 1h 49m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab5fa2d76374660bd80b19acb36d444 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96850e10ef6d43788d0c2dfa37f250ee : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/634dc0931c9645d891e8c4e26adcea19 : SUCCESS in 8m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3711cc534d349b3b476eb7b3cb1d344 : FAILURE in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/740c3d8bbaeb43fb9d9be5befcdd6721 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44da2b90992b4c4592b7954bf0258b0f : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b4590f36f994c47a89bf7ebef723f17 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/883e3cafda734aa6b0cf5e11181b460b : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01ddf2c3fc724520b1aeee856bcbb70f : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98cfa0b3631e430eb9e4f937ded04e76 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35085481b8014a4095f7d3ed9e0f0a0b : SUCCESS in 5m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17fe639660e245e8b1f1406dee9941ce : SUCCESS in 7m 49s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/550233ff5495414599d1c6ca5f797662 : SUCCESS in 27m 50s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/0db9620f35ca4a2e9c08491049ef2927 : SUCCESS in 36m 55s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/5d0dadd480e44bb29b74b9e7fe5bd1ad : SUCCESS in 43m 32s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ef9db3d167eb4a1ea88788b4a87c61d5 : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":5},{"id":"9eceb79c2b0eb7c6c7b8c482bfa2e49786ea4960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 10:37:26.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"14226abbcb480c6098bee1fa40d4c9f73b2eb03a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-10 12:46:27.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"1f80b389c686b83f80cd241353771d9b40d96fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 14:21:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c90c910906a4d998539bf4883c25bfe\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/531e50a2ab6a4686b5deba22492cdc2e : SUCCESS in 1h 29m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a633fb428e484482dc84bbcb76b161 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4514a9f0df7c4f67ae0708ff05ef3e1e : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3278d81adc654f3b9feaf7ba15396a25 : SUCCESS in 8m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f095b9c1f62436a85d9a49ecfbd854c : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f87e10936574ef8831e23b4ff4c3bd0 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47d65e88643b415183d07853ecf08029 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00e333e66c8243569271d9e5f3dfa31f : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9ce17b056c4425899565d3734442a84 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a17d8d922efd4cf4a935b85babe0b388 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61f4755b2edb4b0a8b2bdb06e86e4cf0 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b13e03705a54b1cb827c88632e38507 : SUCCESS in 5m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27a43932591e4484b9319c767ed3b059 : SUCCESS in 13m 07s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/3762907e46fd4e8f82f107a6c592d09b : SUCCESS in 30m 39s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/8909a07e662840db8d1fd703df7333c3 : SUCCESS in 26m 36s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/c96dfe36aabc40edbe1e7bdf714bbbb5 : SUCCESS in 44m 07s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/2c89456c63cd4cccac55d8c6c42f239e : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":7},{"id":"7411145f889ebab75b65d0987b2d0f5c61e7f5f8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-10 17:55:58.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"da2c087da6659d954fa42f066286e6aa1e50b627","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-10 18:54:51.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3d3d0a5382eb9809e029bf03ec62ee0d1737dec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:47:38.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b6a55564e65845ff65edf64784b171af52f88ea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 13:36:34.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a71693ae8dde4c14bb73ea8bff264c0c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5595987faf24fa9976a904bcbf6d419 : SUCCESS in 1h 35m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6513a7f8ac9e4469aebcf345357aaecf : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8f826cc2bd4e0a815ab2e3e60500fc : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/022c7fc046b44945b8e5e97768daa13f : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/100a3d0809c948979b487c788cce9af3 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be12d8d94aa8406a8038d1e188a06def : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8da4b5c38c8944e6836dfd4e9f710740 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1dcd7eb23e42fba56f3898e6fa1816 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3652a9f15e0b426ba5f15ab0c1c9ba45 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f6501be1258481797bd20c641c0a0c8 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28e0cb83c95d4ab28a9771ddc6dd7edb : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6896e58aa6494a5e91cef0a3d869bd0d : SUCCESS in 5m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e972147715b4333a8663fb022145245 : SUCCESS in 7m 41s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/7d6a62be764b4c84b86501e9c4e40ec0 : SUCCESS in 23m 15s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/89f1ff6b69e244738345cb3013663c55 : SUCCESS in 28m 20s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/62b8850c37b947a184bb249c4fbea795 : SUCCESS in 30m 38s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/9f4b5246bad04e68b1a0d911ca102322 : SUCCESS in 33m 17s","accounts_in_message":[],"_revision_number":7},{"id":"abebf454bc7d4c2aaa406e0063bb0fef3f1a5f78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 13:36:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"80d6839fe41fe6ec88eceee2dcb47b1a981a86d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 13:38:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/163de1b01f05484fa739d9382e9fcd6e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6598bb2c060b40e7bba5770e1d5e408a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0256e756a3810f5057e4e48d5f55d3b018b47481","revisions":{"9c6aa10094b66f05edfa13df039449de2c51bf8e":{"kind":"REWORK","_number":1,"created":"2025-07-03 13:49:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/954042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/42/954042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/42/954042/1"}}},"commit":{"parents":[{"commit":"378a7b4eb0c1a930a9fa7fd95ce679a5dab68cad","subject":"Tweak loading distribution version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/378a7b4eb0c1a930a9fa7fd95ce679a5dab68cad"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-02 13:36:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 13:45:58.000000000","tz":60},"subject":"Use importlib.metadata where possible","message":"Use importlib.metadata where possible\n\nThis patch adds the initial infrastructure for handling differences\nbetween importlib.metadata and pkg_resources and migrates a single\nutility over.\n\nChange-Id: I7006197c9799520bf79cbffd5e2bbd6f2152f6de\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9c6aa10094b66f05edfa13df039449de2c51bf8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9c6aa10094b66f05edfa13df039449de2c51bf8e"}]},"branch":"refs/heads/master"},"e172d2e1f87f3bf15788e62c67e27af2300a162e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-07-03 14:40:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/954042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/42/954042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/42/954042/2"}}},"commit":{"parents":[{"commit":"8dc49aa4c6315fccd5ff0528dc37a3457a5f88ea","subject":"Tweak loading distribution version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8dc49aa4c6315fccd5ff0528dc37a3457a5f88ea"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-02 13:36:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-03 14:38:45.000000000","tz":60},"subject":"Use importlib.metadata where possible","message":"Use importlib.metadata where possible\n\nThis patch adds the initial infrastructure for handling differences\nbetween importlib.metadata and pkg_resources and migrates a single\nutility over.\n\nChange-Id: I7006197c9799520bf79cbffd5e2bbd6f2152f6de\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e172d2e1f87f3bf15788e62c67e27af2300a162e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e172d2e1f87f3bf15788e62c67e27af2300a162e"}]},"branch":"refs/heads/master"},"2677b29364bd6f1f8760c5a85efabffcabb80229":{"kind":"REWORK","_number":3,"created":"2025-07-04 09:13:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/954042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/42/954042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/42/954042/3"}}},"commit":{"parents":[{"commit":"8dc49aa4c6315fccd5ff0528dc37a3457a5f88ea","subject":"Tweak loading distribution version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8dc49aa4c6315fccd5ff0528dc37a3457a5f88ea"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-02 13:36:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 09:04:35.000000000","tz":60},"subject":"Use importlib.metadata where possible","message":"Use importlib.metadata where possible\n\nThis patch adds the initial infrastructure for handling differences\nbetween importlib.metadata and pkg_resources and migrates a single\nutility over.\n\nChange-Id: I7006197c9799520bf79cbffd5e2bbd6f2152f6de\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2677b29364bd6f1f8760c5a85efabffcabb80229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2677b29364bd6f1f8760c5a85efabffcabb80229"}]},"branch":"refs/heads/master"},"5ca1f2f1ea0bd3a47463533c42559b2e4530b7bf":{"kind":"REWORK","_number":4,"created":"2025-07-09 15:04:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/954042/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/42/954042/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/42/954042/4"}}},"commit":{"parents":[{"commit":"a159ee74dfe9f3ff448ddc278418721ad94b9025","subject":"Tweak loading distribution version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a159ee74dfe9f3ff448ddc278418721ad94b9025"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-02 13:36:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 14:23:13.000000000","tz":60},"subject":"Use importlib.metadata where possible","message":"Use importlib.metadata where possible\n\nThis patch adds the initial infrastructure for handling differences\nbetween importlib.metadata and pkg_resources and migrates a single\nutility over.\n\nChange-Id: I7006197c9799520bf79cbffd5e2bbd6f2152f6de\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5ca1f2f1ea0bd3a47463533c42559b2e4530b7bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5ca1f2f1ea0bd3a47463533c42559b2e4530b7bf"}]},"branch":"refs/heads/master"},"38865d5495167f15dcbc4eabd6d0ae7d9b962690":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-07-09 17:03:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/954042/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/42/954042/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/42/954042/5"}}},"commit":{"parents":[{"commit":"81346568075d91cc4aa139083a405c1887931ccf","subject":"Tweak loading distribution version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/81346568075d91cc4aa139083a405c1887931ccf"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-02 13:36:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-09 17:03:38.000000000","tz":60},"subject":"Use importlib.metadata where possible","message":"Use importlib.metadata where possible\n\nThis patch adds the initial infrastructure for handling differences\nbetween importlib.metadata and pkg_resources and migrates a single\nutility over.\n\nChange-Id: I7006197c9799520bf79cbffd5e2bbd6f2152f6de\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/38865d5495167f15dcbc4eabd6d0ae7d9b962690"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/38865d5495167f15dcbc4eabd6d0ae7d9b962690"}]},"branch":"refs/heads/master"},"f62eed50a2d02bc4c35525ca9438d9df314907ea":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-07-10 10:37:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/954042/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/42/954042/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/42/954042/6"}}},"commit":{"parents":[{"commit":"b92acd3698c4210ba112b4d11ff34a983f86d1a2","subject":"Tweak loading distribution version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b92acd3698c4210ba112b4d11ff34a983f86d1a2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-02 13:36:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 10:33:43.000000000","tz":60},"subject":"Use importlib.metadata where possible","message":"Use importlib.metadata where possible\n\nThis patch adds the initial infrastructure for handling differences\nbetween importlib.metadata and pkg_resources and migrates a single\nutility over.\n\nChange-Id: I7006197c9799520bf79cbffd5e2bbd6f2152f6de\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f62eed50a2d02bc4c35525ca9438d9df314907ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f62eed50a2d02bc4c35525ca9438d9df314907ea"}]},"branch":"refs/heads/master"},"0256e756a3810f5057e4e48d5f55d3b018b47481":{"kind":"REWORK","_number":7,"created":"2025-07-10 12:46:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/42/954042/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/42/954042/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/42/954042/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/42/954042/7"}}},"commit":{"parents":[{"commit":"b92acd3698c4210ba112b4d11ff34a983f86d1a2","subject":"Tweak loading distribution version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b92acd3698c4210ba112b4d11ff34a983f86d1a2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-02 13:36:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-10 12:38:28.000000000","tz":60},"subject":"Use importlib.metadata where possible","message":"Use importlib.metadata where possible\n\nThis patch adds the initial infrastructure for handling differences\nbetween importlib.metadata and pkg_resources and migrates a single\nutility over.\n\nChange-Id: I7006197c9799520bf79cbffd5e2bbd6f2152f6de\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0256e756a3810f5057e4e48d5f55d3b018b47481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0256e756a3810f5057e4e48d5f55d3b018b47481"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
