)]}'
{"id":"openstack%2Fpbr~954708","triplet_id":"openstack%2Fpbr~master~I43bbcddd87e6038148d22ea76a916d8a05f20574","project":"openstack/pbr","branch":"master","topic":"trivial","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-07-22 12:42:32.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-07-22 09:14:12.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-07-22 12:42:32.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-22 12:42:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I43bbcddd87e6038148d22ea76a916d8a05f20574","subject":"Silence Python warnings","status":"MERGED","created":"2025-07-11 10:25:55.000000000","updated":"2025-07-22 12:43:26.000000000","submitted":"2025-07-22 12:42:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954708-trivial","meta_rev_id":"82e2b232272f9f2ececeb426e3837cc4b83c068b","_number":954708,"virtual_id_number":954708,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-22 12:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-07-22 09:14:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-07-22 09:14:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-11 11:57:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-11 15:17:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-07-11 15:17:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-07-21 08:37:12.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"74fa756c27cbfa58f60f1fe21f5209c6b56ab7d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-11 10:25:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3776bb84cbd5c49a14fb5f72a17bd55657ffd431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-11 11:02:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6b4d6dbc4bccf8275cd3afb3f933d71e43eb92fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 11:57:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc5174b244e7409faf11f0b9877d3f82\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f1f0da482b043cf82f903dbc76e1d06 : FAILURE in 52m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94347252c14f4a63b01a5d768e7fec8c : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e952f45ab3492982ac283ffe1e079f : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f2ef85218af40fc9fbb7149d7f78e40 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71e03d07328a41bab035f6164b5af335 : FAILURE in 1m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc2bc2cf49064dc7ae17f9c7d5981d6a : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4247b7d2fd77468596f3c6fd8ab02db8 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c22d89dd05da469eb635ac45078cefed : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf11abe9665b4c9c85f43611bcfa85df : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fbeb7832e4c40b1a261e042df35cd8b : SUCCESS in 2m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea2e850501f643d4a8b03753cc403918 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2076c292272c455d8fea34cf117c0dd9 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/abe7e19923db40bd90bd6676597376fa : SUCCESS in 5m 15s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/7f64c851278c4795be5e800d55af31bd : SUCCESS in 17m 48s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b0bfae337b0645d18cfc940870c8a071 : SUCCESS in 18m 48s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/4e95aad7616846d0abf8db1e2d84b225 : SUCCESS in 21m 30s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/adee1a9e557c4db799cb3b9dc958ff8f : SUCCESS in 29m 47s","accounts_in_message":[],"_revision_number":2},{"id":"e950e20d294b44028a803756bcc1b2741e8bb179","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-11 12:43:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9e1b9784583c256dc792e7c54cbba08b38cad447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 14:36:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8c9f1cbf90d4ec5bea788f038bad6ec\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb4676fec9e74701934c0a1dc528b5c7 : SUCCESS in 1h 47m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995533867c844e64bdc8758ca11fbc3f : SUCCESS in 1m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/239b3ffd9bf743e7814373bb4b7269de : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4943dd3c3291424c95704cc0b0fd10c5 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b5b366d5c8248fa9ad42bf09fe3a384 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d495ff2b7ef34da3b77401bdcf918fb2 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98def66f8f924947a92bbae72a439e46 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f150e41ca43e4a6496c5bf54efe63dd7 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75a54dee94dd4cc4b07e0ab043045755 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8d41543699a474799134d24b625992b : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8544b09bdce64584ab36e18f45ea2753 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af028ff82e80478ebf849d9a707d390d : SUCCESS in 6m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a8e5986ed824fa0bfa53f1bb09791f3 : SUCCESS in 7m 35s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/88a68bdf287c48119c874e32fed3ba7c : SUCCESS in 22m 35s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/0b660997924c4cd487f4f3ad4a48d250 : SUCCESS in 13m 59s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0234b87137054a8190d135d4d04ff13e : SUCCESS in 18m 48s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/1f6b6dc5155b420cb9f069faaa62075e : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":3},{"id":"8d02bce0955f442b7591cfcd4cde4d4f63ec742f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-16 16:16:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"df3772269d7786b6935e800abbcd05017481bf44","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-07-21 08:37:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"940a36d2fad218f00208e5ed5a8cc4e79292a4bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-21 08:37:17.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/pbr for 954708,3","accounts_in_message":[],"_revision_number":3},{"id":"8c95fd90d35a7c5f3e28d16468692be3f464fe5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-22 09:13:59.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d462829de00f5f2a380e1b652f7e5a1dbd830a07","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-22 09:14:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"720eaa7093463ca5dd4931437ec9339ed0557ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 11:06:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca245fe2bd70462ea768cead238ab90a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b657c9e16749404bbacfeea157aa07d6 : SUCCESS in 1h 49m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238f1f4e2f6a42b292de2771b06e7914 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae2bdad10534097b3cf6416830660fa : SUCCESS in 4m 00s\n- build-python-release https://zuul.opendev.org/t/openstack/build/a844d606a704464193769570f505278c : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d21c720e9d040b0b5f90623d9119358 : SUCCESS in 8m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ff6fff824c94327b319efcfb27c6b75 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aef2cd9f69c4301b49dc97c298dc2a0 : SUCCESS in 7m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab5806c5433249529f15c6723d1dddce : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c58f363449947b7991fb788cb59ea9b : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d75babde96d4987ad68481482e49bd4 : SUCCESS in 8m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7595523914645b0818d2b0274f44fee : SUCCESS in 8m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c16591b7a04644e0942a213efd96dae2 : SUCCESS in 8m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ed54fe504e540e8b91bf514659db5d0 : SUCCESS in 8m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/76f5e9457f914378930835f3ef470a10 : SUCCESS in 11m 11s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/ec7bc7264bfe4038bc35dc250a337275 : SUCCESS in 31m 10s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/d895a35e7bbc457f92f9fcca2d34437a : SUCCESS in 34m 08s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/ecea722d740d4d89a0c19597cb4797fd : SUCCESS in 33m 15s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c31b4a0d7bb0455982f4c8f9b8d67a66 : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":4},{"id":"e896e93964c6cf4b1bd82d8526d38b56d0193070","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 11:07:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e9a266c72cb16661aff54d386b4d05b16d5f0bfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 12:42:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba91bfbbdf38498591a563b34794224b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc9068ac00a042c8a17ee8dd5932aaa6 : SUCCESS in 1h 33m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b180b9f6739a4b9d9cb075be7af2f431 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd52665cbeb1486796f5c1a9a6799676 : SUCCESS in 4m 05s\n- build-python-release https://zuul.opendev.org/t/openstack/build/e34d3bea546b43309ce76d4ebe3b968f : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2595b93797a4d799d16ee7a3850455f : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4049be5e95684a88a1e87de7aed5ccc1 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52270f31768c4a6e92c65d3fdad0c824 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7da688de47ba4cbfbb6af1f93a1e2336 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3469627cc71d4af7a671baea8fb208a9 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cfc5f894cd749d59d6fd2999e607cc0 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b60ca76264c45f693e02cfa019fd1ab : SUCCESS in 8m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e127364c582e4a5b9291ba506774fa63 : SUCCESS in 7m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71df1ab3a9204b25bb4ee383bdf18217 : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6c8de6ce8904e3a9e8385500cc8b8f1 : SUCCESS in 8m 12s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/1f226e71e9e54d5082aa16ab0c08c895 : SUCCESS in 31m 48s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b0f7416457d94c9a8633c5a50e12ef71 : SUCCESS in 17m 02s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/17c06d8d76c24c7db53775c1c9c00e8c : SUCCESS in 19m 00s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/3f0807edd7a7469a9c6015982070c38d : SUCCESS in 32m 05s","accounts_in_message":[],"_revision_number":4},{"id":"c4ecdfffa0c7045149fb6b2b61a0070ee32b09bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 12:42:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"82e2b232272f9f2ececeb426e3837cc4b83c068b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 12:43:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96157eeee71f4a02922b6ef901a2888a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e44fdfcce004b34a7b1ea209d9ee55a : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3261691831db4393493916ac10894abbd8c722f6","revisions":{"b5d09bdf0200291baf4f153e0382474c206c9854":{"kind":"REWORK","_number":1,"created":"2025-07-11 10:25:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/954708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/08/954708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/08/954708/1"}}},"commit":{"parents":[{"commit":"248450dd805cb7dda37e94f21e2fb4840aa3b435","subject":"Add __future__ imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/248450dd805cb7dda37e94f21e2fb4840aa3b435"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-11 10:16:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-11 10:23:40.000000000","tz":60},"subject":"Silence Python warnings","message":"Silence Python warnings\n\nMake out build logs a little more readable since setuptools is currently\nvery shouty (we\u0027re working on it). We do this \"universally\" in all\nplaces where subprocess is invoked, rather than at each call site, since\nit\u0027s simpler and this envvar is a no-op if executing non-Python code.\n\nChange-Id: I43bbcddd87e6038148d22ea76a916d8a05f20574\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b5d09bdf0200291baf4f153e0382474c206c9854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b5d09bdf0200291baf4f153e0382474c206c9854"}]},"branch":"refs/heads/master"},"c434ce47f48fbf49ed7a1d9d0937b1288826fd5a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-07-11 11:02:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/954708/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/08/954708/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/08/954708/2"}}},"commit":{"parents":[{"commit":"248450dd805cb7dda37e94f21e2fb4840aa3b435","subject":"Add __future__ imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/248450dd805cb7dda37e94f21e2fb4840aa3b435"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-11 10:16:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-11 10:58:05.000000000","tz":60},"subject":"Silence Python warnings","message":"Silence Python warnings\n\nMake our build logs a little more readable since setuptools is currently\nvery shouty (we\u0027re working on it). We do this \"universally\" in all\nplaces where subprocess is invoked, rather than at each call site, since\nit\u0027s simpler and this envvar is a no-op if executing non-Python code.\n\nChange-Id: I43bbcddd87e6038148d22ea76a916d8a05f20574\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c434ce47f48fbf49ed7a1d9d0937b1288826fd5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c434ce47f48fbf49ed7a1d9d0937b1288826fd5a"}]},"branch":"refs/heads/master"},"8d760dab2132b5bfed77ef371b312f7813349ef5":{"kind":"REWORK","_number":3,"created":"2025-07-11 12:43:28.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/954708/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/08/954708/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/08/954708/3"}}},"commit":{"parents":[{"commit":"248450dd805cb7dda37e94f21e2fb4840aa3b435","subject":"Add __future__ imports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/248450dd805cb7dda37e94f21e2fb4840aa3b435"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-11 10:16:09.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-11 12:43:22.000000000","tz":60},"subject":"Silence Python warnings","message":"Silence Python warnings\n\nMake our build logs a little more readable since setuptools is currently\nvery shouty (we\u0027re working on it). We do this \"universally\" in all\nplaces where subprocess is invoked, rather than at each call site, since\nit\u0027s simpler and this envvar is a no-op if executing non-Python code.\n\nChange-Id: I43bbcddd87e6038148d22ea76a916d8a05f20574\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d760dab2132b5bfed77ef371b312f7813349ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d760dab2132b5bfed77ef371b312f7813349ef5"}]},"branch":"refs/heads/master"},"3261691831db4393493916ac10894abbd8c722f6":{"kind":"REWORK","_number":4,"created":"2025-07-22 09:13:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/954708/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/08/954708/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/08/954708/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/08/954708/4"}}},"commit":{"parents":[{"commit":"15d65487017e9d977958431b8a83c8d5e57b7c01","subject":"Merge \"Remove support for custom compilers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/15d65487017e9d977958431b8a83c8d5e57b7c01"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-22 09:13:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-22 09:13:44.000000000","tz":60},"subject":"Silence Python warnings","message":"Silence Python warnings\n\nMake our build logs a little more readable since setuptools is currently\nvery shouty (we\u0027re working on it). We do this \"universally\" in all\nplaces where subprocess is invoked, rather than at each call site, since\nit\u0027s simpler and this envvar is a no-op if executing non-Python code.\n\nChange-Id: I43bbcddd87e6038148d22ea76a916d8a05f20574\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3261691831db4393493916ac10894abbd8c722f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3261691831db4393493916ac10894abbd8c722f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
