)]}'
{"id":"openstack%2Fpbr~955274","triplet_id":"openstack%2Fpbr~master~I6c1a16cef5b0ab25b0c9affd3fdab5d36ac065ae","project":"openstack/pbr","branch":"master","topic":"pkg_resources-removal","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-04 23:54:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6c1a16cef5b0ab25b0c9affd3fdab5d36ac065ae","subject":"docs: Remove deprecated options","status":"MERGED","created":"2025-07-17 10:13:39.000000000","updated":"2025-08-04 23:54:48.000000000","submitted":"2025-08-04 23:54:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"955274-pkg_resources-removal","meta_rev_id":"f3ebbee572fd9465b18f962fd248b7e7b3a39b1b","_number":955274,"virtual_id_number":955274,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-04 23:54:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-08-01 16:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-07-21 08:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":2,"date":"2025-07-17 17:15:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-21 08:35:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-17 11:44:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-17 17:15:41.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-07-21 08:35:25.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-08-01 16:48:14.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-08-04 20:05:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"da17225af64bbaf42875182a53b5e692aca3fc5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-17 10:13:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff3d05a7c28d3618f78a4b527226b2bc2cf3bebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 11:44:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b425e064a8c4ea38f50dbaf5b85a3b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c57417eb894d55977ad7b9cd4315cf : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16e7211a5b68452da38b837179c1e5f0 : SUCCESS in 2m 23s\n- build-python-release https://zuul.opendev.org/t/openstack/build/57ca872a9250473092694fd0af25906c : SUCCESS in 2m 19s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/9aa0d6218d274149a508b9a6e6c9a929 : SUCCESS in 22m 58s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/e28b5254783f48f2a2b0b2349380b557 : SUCCESS in 25m 06s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/00c940fcc8b14171a875b663266ee5d9 : SUCCESS in 33m 29s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/919f1b400d6146dcb04996b836b98362 : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1bf242b1ff77919daa25e6967d365ce625fae958","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-17 17:15:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf4e772f4bcbaef7a6d9a4ef7b1903006f9f0205","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-07-21 08:35:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43b9550da15e72570523798423d1af7d1d2f701a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-01 16:48:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5e2be528172997ebc87213dfc5ae248e892c2a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 18:05:34.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/949188 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"0061077412defbab5735e38c71d7ac69fcbb1c7d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-04 20:05:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"443b28702fe9c7d457dffcccae443a67c87f6e28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 20:51:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/365c86af85774dcaaafe4d12453fdfca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b2a5a6a6f5d4fb3a19aa0e76348629f : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a731524f253449b83a52524ce7e29a5 : SUCCESS in 4m 21s\n- build-python-release https://zuul.opendev.org/t/openstack/build/204efd1f3fb64e5f8944d08d978196c6 : SUCCESS in 2m 12s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/d176babcde7049eb81de8646ce422137 : SUCCESS in 28m 53s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/0a554315722a468a8aa65a5057098c46 : SUCCESS in 14m 49s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/4147b14814f84dfda8797afd373548ec : SUCCESS in 38m 56s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/35423fb525fe4f7eb3dfdb1e34fe8feb : SUCCESS in 16m 44s","accounts_in_message":[],"_revision_number":1},{"id":"24663776c63f3dd4da07380689f683098ad1a03c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 22:07:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cfe38cb399666caee88d66fd0026f734b3349db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:54:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ea87c68bfb94fb5a900a2cd258a80c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a203bdfa628e4559b9efb09ffe0c23c0 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f85dc0b19d4e52bdeb539b7e970480 : SUCCESS in 4m 15s\n- build-python-release https://zuul.opendev.org/t/openstack/build/8641304de0f34a7aa914db8ad9333cae : SUCCESS in 1m 50s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/74fdbe9c23a446758e0f6ad1e6ff6c3c : SUCCESS in 14m 57s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/87d0de898ed4462b83a6c5a39c02ec6b : SUCCESS in 29m 47s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/19e081fc75014e40bea8da2d1d44d302 : SUCCESS in 39m 51s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/2c2cccda8984479196d523ed3aa17063 : SUCCESS in 35m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f3ebbee572fd9465b18f962fd248b7e7b3a39b1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:54:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3bbb792b7946e398aef6a31ffdb08d95172f22d6","revisions":{"3bbb792b7946e398aef6a31ffdb08d95172f22d6":{"kind":"REWORK","_number":1,"created":"2025-07-17 10:13:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/74/955274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/74/955274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/74/955274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/74/955274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/74/955274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/74/955274/1"}}},"commit":{"parents":[{"commit":"1ed80ed617cf99b5a4850fc1545be24c76a31f93","subject":"tox: Stop calling \u0027setup.py\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1ed80ed617cf99b5a4850fc1545be24c76a31f93"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-17 10:12:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-17 10:12:57.000000000","tz":60},"subject":"docs: Remove deprecated options","message":"docs: Remove deprecated options\n\nWe do not need to support Python 2 for building docs.\n\nChange-Id: I6c1a16cef5b0ab25b0c9affd3fdab5d36ac065ae\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3bbb792b7946e398aef6a31ffdb08d95172f22d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3bbb792b7946e398aef6a31ffdb08d95172f22d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
