)]}'
{"id":"openstack%2Fpbr~956389","triplet_id":"openstack%2Fpbr~master~I46f57996dce8fbc1a511e939cff00ddaa87df84e","project":"openstack/pbr","branch":"master","topic":"pkg_resources-removal","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-08-04 23:54:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I46f57996dce8fbc1a511e939cff00ddaa87df84e","subject":"Remove duplicate line in test_generates_c_extensions","status":"MERGED","created":"2025-08-01 16:58:47.000000000","updated":"2025-08-04 23:56:00.000000000","submitted":"2025-08-04 23:54:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956389-pkg_resources-removal","meta_rev_id":"132647fd965c0fa2aba5f16c2f5d45f45d3c5b8d","_number":956389,"virtual_id_number":956389,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-04 23:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-08-01 17:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-01 17:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-01 17:15:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-01 17:15:36.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-08-01 17:20:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-08-01 18:39:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42a8043bd2c38ff36820353effbc9806c095b7c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-01 16:58:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30df9c69e9058deb720fbb6fe36dcfbdc8568332","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-08-01 17:15:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"722c032489c84d326695029d5a0b3b153ab02f3d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-01 17:20:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63c24536c27fe3959be1ab6391b3110174f30bda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-01 18:39:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0152f9dfb6448418a22aea755b98ed8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/286a59dbce324752bc8c20976f6691a3 : SUCCESS in 1h 37m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47c1418203d44a2e993c2567db46af0d : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e53a720023046ca903effcd685d7b20 : SUCCESS in 3m 08s\n- build-python-release https://zuul.opendev.org/t/openstack/build/4f4cdd3cf9ce43079a460b41d5b59f4a : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efbba22101d945db97bd44c483c3ebe4 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ffdfb62c9764cc79472b2673cc56f1e : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/846e090f591f4b46a26584f9297275e7 : SUCCESS in 2m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/772f0cb4d6804c04900bf0f10e008d69 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4893eff6814d4bbbbcf911206811ea1c : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/057dd238be7f4f3bac30099067f1fbbe : SUCCESS in 8m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/169b2c82931146c6b044df42e6ddcebe : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7ac16a1c7f54877a304b87cdfc4bd68 : SUCCESS in 6m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e304bf4e70ac4c7caa8a0cef7ae0d75f : SUCCESS in 5m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f0be67128be4e6b99f9a41ae406763f : SUCCESS in 8m 08s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/d469bbf16bb44e6086ac2966f22e1184 : SUCCESS in 18m 59s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/1fee27bb60024030a52937b654f04bb8 : SUCCESS in 28m 41s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6975a564da204ac89f98e0f8fbf7f386 : SUCCESS in 18m 43s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/392aa7bf927147f89989100f3b9ca265 : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8af4e75078e87a5270f86f8b841327730ece6760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 22:07:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0db107b90570a4be1208cbf56edcc4bee7d5d1c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:54:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c296313e509746b2802694791049010e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fbb8fc5cbadc4f99a460ad05816efc21 : SUCCESS in 1h 35m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d92df769e09748a1a23cfbfd35edbd6b : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d4d8c48751a4073bf6721795b680f68 : SUCCESS in 3m 11s\n- build-python-release https://zuul.opendev.org/t/openstack/build/c8268c21daf34683a413ce80b5269040 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f0dca61d5d84b08bf6fe3f3d677ea42 : SUCCESS in 9m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd0cf859663649c6a342f0f952db0e4e : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2ae8a4b06e545b4a98552dd16749921 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1ee087578cf4a8b835987626909188a : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/762f414777214875a7a5e6b3af7b4659 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c21b9024e00b4467a871f1df99130b68 : SUCCESS in 8m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d74be5276976469db8f6ab2e6f3524a9 : SUCCESS in 8m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7a0cd057e4741758effb907fc6e5d53 : SUCCESS in 8m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4491adf82f684a108ba903f1f6b894d8 : SUCCESS in 8m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8dbfbdfca61b4bb793aaab96c6e18273 : SUCCESS in 11m 18s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e01e2e5b866d4db389d825dc1758df17 : SUCCESS in 25m 24s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/5dc6046fb97b4510a696a5797897346e : SUCCESS in 23m 30s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9e7c829b746f44848720e0d3509a55fb : SUCCESS in 32m 26s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ae02f1d96d0c4469aa24cc9a531ba3ff : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1dba2dcd734284b598263f5edffb709bbbf2785e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:54:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"132647fd965c0fa2aba5f16c2f5d45f45d3c5b8d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 23:56:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8971c6c7356749c5b797cca45fe96cee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3965ff07825c4c9890cea48ad19c45de : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"63c8efac0a8d4e800270270c7b21c6d4ba8720d8","revisions":{"63c8efac0a8d4e800270270c7b21c6d4ba8720d8":{"kind":"REWORK","_number":1,"created":"2025-08-01 16:58:47.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/89/956389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/956389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/956389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/956389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/956389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/956389/1"}}},"commit":{"parents":[{"commit":"2e8db96e34f25287767b8c9121b25a38f3764741","subject":"Update setup.cfg usage docs to explain pbr vs setuptools better","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2e8db96e34f25287767b8c9121b25a38f3764741"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-08-01 16:57:55.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-08-01 16:57:55.000000000","tz":-420},"subject":"Remove duplicate line in test_generates_c_extensions","message":"Remove duplicate line in test_generates_c_extensions\n\nA prior refactor of the test suite accidentally added a redundant line\nto this test case. Let\u0027s remove it to avoid future confusion over\nwhether or not it is necessary.\n\nChange-Id: I46f57996dce8fbc1a511e939cff00ddaa87df84e\nSigned-off-by: Clark Boylan \u003cclark.boylan@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/63c8efac0a8d4e800270270c7b21c6d4ba8720d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/63c8efac0a8d4e800270270c7b21c6d4ba8720d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
