)]}'
{"id":"openstack%2Fpbr~957264","triplet_id":"openstack%2Fpbr~master~I29a44fcf38c9c19a3f1ee4884b3d2f09efe45765","project":"openstack/pbr","branch":"master","topic":"fix-description","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-08-18 15:15:17.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-18 13:42:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I29a44fcf38c9c19a3f1ee4884b3d2f09efe45765","subject":"util: Skip normalization of description if README opts present","status":"MERGED","created":"2025-08-13 16:05:19.000000000","updated":"2025-08-18 15:16:17.000000000","submitted":"2025-08-18 15:15:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957264-fix-description","meta_rev_id":"d60eb1d94832355cbdc681720c781a86abaae499","_number":957264,"virtual_id_number":957264,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 15:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2025-08-17 12:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-08-18 13:42:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-13 17:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-14 15:32:36.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-08-17 12:26:06.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"67426a440974211e79e4f43c0f50649f543597a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-13 16:05:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a73b2d797722803b0e84caec9a274620fb56973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 17:53:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc1d737fd964b82bbcce948eddd975a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/500e84fc97a645b4a729bd0a006c6810 : SUCCESS in 1h 42m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/987f1466ebc744bbb63e88965c26b6f0 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc941d32e09944bd91bf012c16b151bf : SUCCESS in 2m 36s\n- build-python-release https://zuul.opendev.org/t/openstack/build/65347a227a3747d79db3cfc01d53a528 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/307c93aa3de14b928e617fdb697630de : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f8cd1ce9fd14b64a3ecb9b5d54f0b4c : SUCCESS in 9m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5bf993031314331b22a9759cd0fdcf9 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7661d042765442e58c522eef8896cf15 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fc4c5b9a1ca438d9a92229ae773348f : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43a06804befa4aaf9d1b8294b6393fec : SUCCESS in 8m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f32c4833dee42beb0c9f51aa2295ec2 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20544a38c6bf41e29b4006783d902096 : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dd942a077764b55a2c9011e43e40475 : SUCCESS in 5m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/865575669d20492fb162ef54c87417bf : SUCCESS in 8m 01s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/756dca386ab34d53ac638fe8f3ea5cba : SUCCESS in 24m 17s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/684254bca8a344c6b1721617789e9b0b : SUCCESS in 22m 58s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/62dd691c072945f489257096fb0c8332 : SUCCESS in 44m 36s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/14faeffcea4146d7bafabe445d6273ad : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":1},{"id":"11bac53bacd27ae216610b5da210cf62877d9b07","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-14 15:32:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1cd90b156374fa1b20fdad6962a8377b3bdc3a66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-14 16:07:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c7f9d78aa4d8352ffde219942b2b025a61d3085d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 04:08:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddb63f0b6c8d47ec99da6844485e8561\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9a16f7e243240099036ceba84a8290a : SUCCESS in 59m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/298defbd1fcc4d9c8ef6ecac6cfba39b : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b12796be1bc455c84f2629cf479ad5b : SUCCESS in 2m 12s\n- build-python-release https://zuul.opendev.org/t/openstack/build/d5cac3bdafb045ea9e75d71f0cbe8ad3 : SUCCESS in 1m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9382e17c0afc4059972cde99e98d78d1 : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/39bdaadb0580459d80f3d14544150fb3 : SUCCESS in 2m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6b49692373c4ed09dbe223927199d92 : SUCCESS in 2m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61709392858f48d186947453b0d52872 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c729ea57566b46c498a494d0b8522bf0 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44c5a88a3f4246eaa394f96b54637282 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6632d2ad75474219bee1ba1cd0c55e70 : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c40e7a6e87ee477e9dc05c14fccb1565 : SUCCESS in 7m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58aedc0c17e64ed9a18d37a47674b473 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f667ca3907254b90ab8bb9d0075c662e : SUCCESS in 4m 30s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/32de384b408643dd8b62523a064e8fac : SUCCESS in 16m 09s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/7c68a675a043449ba1ea4e53e6a76a3c : SUCCESS in 16m 55s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0ed304ccf88b492ebb9cb2d95a964550 : SUCCESS in 17m 42s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b3b6f10242ce466089dd21945507d857 : SUCCESS in 17m 13s","accounts_in_message":[],"_revision_number":2},{"id":"b9a2e9e1808e2f2079caea11d941300b713ba617","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-17 12:26:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"93c4f082dd14a13dc8601424bf00ca806a3963cd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-18 13:42:34.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b3d43f11c5e9e0ff6199a08a5d2464bf7330d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 13:43:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ee94a510fe1e5e443f7e0df8c57a48f5c32c6a88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 15:15:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a5b78fa1f4646d4bd43d795dbe7823e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d84951cf934240e2a3021673c004b389 : SUCCESS in 1h 05m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d739382e07148c68bb7006a664b9de9 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25bb6a29403d4beea16ca301718f2473 : SUCCESS in 3m 07s\n- build-python-release https://zuul.opendev.org/t/openstack/build/dc9fbb1f1f884452a53751f06fde37c4 : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9cd49f2133d4a99bada41f7efbfb849 : SUCCESS in 8m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9adbd33d4a20441ba79852eb3dbb39a2 : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97a98f1a7b4247148c79e1a26207b40e : SUCCESS in 2m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64d19267c96748c788ca6efe30b8d21e : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adb32809634847858d7c765f1b4e0e15 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a1dcfd341c5454899249c5b7f6c9b30 : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f56f2ff05024bce851c1daa5116841c : SUCCESS in 8m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5436a1abd7374a8b8942b604b2c3518b : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45021ae80459488b85f82f1e13fedb64 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9bf0ec3b67e945b3b22093a3b46798a4 : SUCCESS in 4m 52s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/0e02976fa5744bc2839012e76d4d5aef : SUCCESS in 28m 17s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/4f28311da7074aed895b8001d6fd8726 : SUCCESS in 14m 15s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/ac1b04def5d740f79c7a2fa732f73807 : SUCCESS in 42m 38s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/78b3a97af59642209350588fefe5fe02 : SUCCESS in 22m 42s","accounts_in_message":[],"_revision_number":2},{"id":"af7a687fe048dc124f4be6aa8b77ad669012ac43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 15:15:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d60eb1d94832355cbdc681720c781a86abaae499","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 15:16:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb29fc71d97449ea9dd4e64532938507\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de98500053b24738a30c86945c6b9107 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eb42fb3dd8e3aa95e63dfafc1fc96f4b5fb8e849","revisions":{"275e9cf9d33ed13d5c899b43b8406acc1e3eeeab":{"kind":"REWORK","_number":1,"created":"2025-08-13 16:05:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/957264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/64/957264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/64/957264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/64/957264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/64/957264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/64/957264/1"}}},"commit":{"parents":[{"commit":"d358c87b8576f75951f78e1ffe635b857d96fb05","subject":"util: Make handling of description_file clearer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d358c87b8576f75951f78e1ffe635b857d96fb05"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-13 16:02:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-13 16:02:31.000000000","tz":60},"subject":"util: Skip normalization of description if README opts present","message":"util: Skip normalization of description if README opts present\n\nChange-Id: I29a44fcf38c9c19a3f1ee4884b3d2f09efe45765\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-bug: #2120575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/275e9cf9d33ed13d5c899b43b8406acc1e3eeeab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/275e9cf9d33ed13d5c899b43b8406acc1e3eeeab"}]},"branch":"refs/heads/master"},"eb42fb3dd8e3aa95e63dfafc1fc96f4b5fb8e849":{"kind":"REWORK","_number":2,"created":"2025-08-14 16:07:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/957264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/64/957264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/64/957264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/64/957264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/64/957264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/64/957264/2"}}},"commit":{"parents":[{"commit":"f198c2e15fe368abb9c9a756241108100f90a60a","subject":"util: Make handling of description_file clearer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f198c2e15fe368abb9c9a756241108100f90a60a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-14 16:07:00.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-14 16:07:00.000000000","tz":60},"subject":"util: Skip normalization of description if README opts present","message":"util: Skip normalization of description if README opts present\n\nChange-Id: I29a44fcf38c9c19a3f1ee4884b3d2f09efe45765\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-bug: #2120575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/eb42fb3dd8e3aa95e63dfafc1fc96f4b5fb8e849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/eb42fb3dd8e3aa95e63dfafc1fc96f4b5fb8e849"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
