)]}'
{"id":"openstack%2Fpbr~957265","triplet_id":"openstack%2Fpbr~master~Ie8889bb66773cf656076e627b8032cd99978fb15","project":"openstack/pbr","branch":"master","topic":"fix-description","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-18 13:42:38.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ie8889bb66773cf656076e627b8032cd99978fb15","subject":"util: Deprecate description_file","status":"MERGED","created":"2025-08-13 16:05:19.000000000","updated":"2025-08-18 15:17:03.000000000","submitted":"2025-08-18 15:15:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957265-fix-description","meta_rev_id":"312879a5b87a4165d786c66bfd7efbd0cea11c41","_number":957265,"virtual_id_number":957265,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 15:15:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-08-14 17:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2025-08-17 12:38:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-08-18 13:42:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-13 17:45:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-14 15:33:16.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2025-08-14 17:09:16.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-08-17 12:32:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-08-17 12:38:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"97d611144bca1a03be5135d0fffc955a7912dad1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-13 16:05:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"209cad5c05683a8c8ac384a133816a91728a1089","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 17:45:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55bde354e7904c51a9b132f08a704f92\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66d9a67412944608bdb23ced5e5f78d2 : SUCCESS in 1h 32m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75116077b65c47f1aaa8a52c19e516c7 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef50cf871be4ecba1442f3ed8b72950 : SUCCESS in 3m 14s\n- build-python-release https://zuul.opendev.org/t/openstack/build/5bc3a2591f1240bcb4ccd6aaa915a0c0 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acb556879e584e1494bd3fa22f83a320 : SUCCESS in 7m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f68861b909e54fe3b9353e4e7cf71390 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31b7e9f7d16a48cf89b4e36deca9d513 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16856177088d450faae51bf152f52cbb : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79a2ebf800c94393ba43a91c4cbaedfb : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8386d3e86775497693c467673326dc21 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc395a3b564642e4838ce0dc327a0222 : SUCCESS in 8m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6cc50796bece4140acf0e40b4869777d : SUCCESS in 6m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce5bd938024f4bcc9fb76182b43c1f2b : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b801ffba78b456796bfe301ddf66c01 : SUCCESS in 9m 31s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/955f54ad2b9545fcaea21f734fd5fb6a : SUCCESS in 30m 15s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/52e7bd3d593442c7ace691cf8a9ca426 : SUCCESS in 25m 43s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/b77dec201fe1479abe7551964d1d3da0 : SUCCESS in 31m 48s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/7eef6c276ef042ad893d74f4b7489976 : SUCCESS in 30m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6ab0d1fa63982141bdc325b0de9de314568536a0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-14 15:33:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"590558077f9ff05007b75999281272c024b91079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-14 16:07:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4cb16abc6ceb48fc2ed9e74c257f327d5b7d22fc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-14 17:09:16.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9d03089458ffa13acf32a632e1c48eb27ba52bbe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-14 17:24:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"586cee66b1ac42d278a0ed5e9227b3e9a2960962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 01:19:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4cd3eae68524178b9ad5ab210455103\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4827cb10b3324ac1b198c2eabc3f57f4 : SUCCESS in 56m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a572e008c84fa49d5dc88b7e0459b5 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97493a3dbef346b28f0155ec194f8b25 : SUCCESS in 1m 42s\n- build-python-release https://zuul.opendev.org/t/openstack/build/18c41152326b4480b70d9c016debc80f : SUCCESS in 1m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4f30dbfbdf04b1691acf5d8ebb904cb : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eed4e423112a473697817d0c5bc14355 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87bf1d39b6a74441a1fa5495f69a3d1b : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/307601448e2f466582c1ccaab78faeb1 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c403d62624894903b35959b5b6bc1908 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f26dd819ffc4270b29ffd17ba3d5773 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ddeb02f791a423487fbd88288ad90ce : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb6546651f6046a784a0a4dca9560c58 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e27cfd9b5764fe79e9c2388a124d804 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f26c68f039e6446c948a2914c32868d5 : SUCCESS in 4m 26s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/8e071739a4074fb78ade9e6d6732fec0 : SUCCESS in 17m 39s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/618dc95fc86047a68a0bd960de53a671 : SUCCESS in 27m 03s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/70792df94d994cd1a7d3e4bcc50753d9 : FAILURE in 20m 45s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a2adb352465e4248b1cc04d678c4e93e : FAILURE in 22m 42s","accounts_in_message":[],"_revision_number":2},{"id":"db326c48311c127d43a63ad0f6540c73fce39b11","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-17 12:32:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3006445f44cc92faea990bd91597682064cceada","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-17 12:38:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"842d37adc036cc611c4a26a8f8acaa95956048c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-17 14:25:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1537c2d2731b41e6908bfb0a7d07afc0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe7b2c2c19c74b72b3b427270fb6474a : SUCCESS in 1h 49m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12787aeaf3204cd1ac0f894b89b9384d : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c9fbe483b24d3b83c3c8a843b68695 : SUCCESS in 4m 12s\n- build-python-release https://zuul.opendev.org/t/openstack/build/d68b7d93623e480da19e2bad92de3ce4 : SUCCESS in 1m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c4b8c44c7ab465787dd48c2f172a84f : SUCCESS in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cdea2bd71f4e45f79e7e7e672b67b497 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c27643efb32247699267006a8234da81 : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf504dc52532455eb96bfedc9239a14d : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13bc14b0927e4cc1814b0fb54288ef6b : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2ac257c21b6492586faa0452454c161 : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/540e89252ab2415a9f56efec4a21ddd8 : SUCCESS in 8m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02480c955443417185f1e7df7fb779ee : SUCCESS in 9m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bcbe89adced941d283a1dc4ebeae555b : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee8405074e7e4e5c830811174a9adba7 : SUCCESS in 10m 59s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/71082871d3e54fc48197fc758db85343 : SUCCESS in 33m 24s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/12cc06f118c2413dbd9f1cda4903a9cb : SUCCESS in 29m 40s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/ddda31ba4d1a45d0aa9d876c584ee8ef : SUCCESS in 40m 16s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/1f8165d5fa574a38bc096063cd95f45c : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":2},{"id":"31c7038c2c7a310a9516fc10d5b1af360c03e42f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-08-18 13:42:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91eaa838eebba508ae79baac6da0b9953c2a51e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 13:43:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f84f5829b096f9da8cf9ff1dbced695a605539d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 15:15:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfe1aa60b0a4466e86f5149b5ee257d1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b98facb9da24bbe973a5adc27217cdf : SUCCESS in 52m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd82a06ee964d5dabb220ccb091db90 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5a54888069b4ec4bb3d4b9e5ec5b070 : SUCCESS in 3m 13s\n- build-python-release https://zuul.opendev.org/t/openstack/build/3a6779d5b471458eb7547e669e03fab5 : SUCCESS in 1m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d92f10959eff4b1583ad307ab3c68c70 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1dcf555a026476f830dd27023fe098d : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40fa33f23b5445468ef8d3a1cee2515f : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6abd127c244143b390e9f1630220e032 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c95bf65c3004a4c8175ab99c06f957b : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c6d0d59bf0e42f1bd12ebf6ee698e47 : SUCCESS in 8m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9cb55be2c204525a60ef7c461ccb284 : SUCCESS in 6m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3804dd3e9a14e04b012b2ddf3b8b687 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/804162f0f66b489b84f493ea6a6153a5 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b727c464dc241ebb8ef4d2d70aa81e6 : SUCCESS in 5m 06s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/9a90f9963d5c4916908bb1be5c482612 : SUCCESS in 13m 56s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/cc564cbfba85405e9a431adbdb184303 : SUCCESS in 32m 29s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/7eddff375389411db0519e63453c7dbe : SUCCESS in 21m 47s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b387b2466e334f9f83187c15c6ac1108 : SUCCESS in 22m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e082e7dd6932af6635bf34a6401c1ea753c7b506","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 15:15:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"312879a5b87a4165d786c66bfd7efbd0cea11c41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 15:17:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b79e6b7215c4dd88fc8b41939b3552f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f3a7eebad14d6da94bef1872a544f3 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a0d431c5dec60b674ba726323edf442c845de2c2","revisions":{"7f7f8d871b5101401bfcbdc670b75485d870527e":{"kind":"REWORK","_number":1,"created":"2025-08-13 16:05:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/957265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/65/957265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/957265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/957265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/957265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/65/957265/1"}}},"commit":{"parents":[{"commit":"275e9cf9d33ed13d5c899b43b8406acc1e3eeeab","subject":"util: Skip normalization of description if README opts present","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/275e9cf9d33ed13d5c899b43b8406acc1e3eeeab"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-13 15:26:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-13 16:03:39.000000000","tz":60},"subject":"util: Deprecate description_file","message":"util: Deprecate description_file\n\nThis was missed since it wasn\u0027t in \u0027CFG_TO_PY_SETUP_ARGS\u0027, but it\u0027s\nanother pbr option that has a setuptools-native option that people could\nuse.\n\nChange-Id: Ie8889bb66773cf656076e627b8032cd99978fb15\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7f7f8d871b5101401bfcbdc670b75485d870527e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7f7f8d871b5101401bfcbdc670b75485d870527e"}]},"branch":"refs/heads/master"},"a0d431c5dec60b674ba726323edf442c845de2c2":{"kind":"REWORK","_number":2,"created":"2025-08-14 16:07:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/957265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/65/957265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/957265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/957265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/65/957265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/65/957265/2"}}},"commit":{"parents":[{"commit":"eb42fb3dd8e3aa95e63dfafc1fc96f4b5fb8e849","subject":"util: Skip normalization of description if README opts present","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/eb42fb3dd8e3aa95e63dfafc1fc96f4b5fb8e849"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-14 16:07:35.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-14 16:07:35.000000000","tz":60},"subject":"util: Deprecate description_file","message":"util: Deprecate description_file\n\nThis was missed since it wasn\u0027t in \u0027CFG_TO_PY_SETUP_ARGS\u0027, but it\u0027s\nanother pbr option that has a setuptools-native option that people could\nuse.\n\nChange-Id: Ie8889bb66773cf656076e627b8032cd99978fb15\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a0d431c5dec60b674ba726323edf442c845de2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a0d431c5dec60b674ba726323edf442c845de2c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
