)]}'
{"id":"openstack%2Fpbr~960028","triplet_id":"openstack%2Fpbr~master~I909916dc861d086c449ceb4df4ad919a99a724d4","project":"openstack/pbr","branch":"master","topic":"pep517","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-10-24 13:41:39.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-28 22:12:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I909916dc861d086c449ceb4df4ad919a99a724d4","subject":"Inline LocalInstallScripts._make_wsgi_scripts_only","status":"MERGED","created":"2025-09-08 11:55:24.000000000","updated":"2025-10-28 22:12:51.000000000","submitted":"2025-10-28 22:12:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960028-pep517","meta_rev_id":"92b2cf18c048ce682d93d8f713ee18c8e63bd888","_number":960028,"virtual_id_number":960028,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 22:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-10-24 13:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-27 20:26:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-27 20:26:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-08 14:19:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-23 16:54:46.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-10-27 20:26:59.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"48b7052b866632c9e7fe72c445b9420935ae9ff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-08 11:55:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e272c6de32c06fab54e06f3d1677e8969ac18a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 14:19:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c55846a8ef774fd3801bc3c17687ce44\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3967161b3e6045e5a2ca62832f1dcce9 : SUCCESS in 2h 22m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caca2a20b3834900a2d6de73b1f22ce9 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1719ba735a874c33bc4e4d79c27c05a3 : SUCCESS in 2m 16s\n- build-python-release https://zuul.opendev.org/t/openstack/build/de9b6c281e5e49c6a36a8fbe59a165e7 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd88a41923494e0798c20f4781b6c9a1 : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e85cf2d667641fb99bae6ae3fb3c02e : RETRY_LIMIT in 9s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b697aa2ea0a4ca08902e8550e0c97f3 : RETRY_LIMIT in 8s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e7d57287b1748afae0a9ea83dc58032 : RETRY_LIMIT in 9s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35d3d6ff89754ec9b667e03c108ccc05 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f7dbb64823a4aa7a21a13cbdd32ee73 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a2d8b0bd30a4061aac19ee06c3c7c7a : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d8a5fb49d7d4aaa9dc0343779ea9032 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a393ab4b1ac460cbda6b656b97942fa : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0926868d5ec24906b37e74ae68090f3c : SUCCESS in 4m 48s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/783cf59c4e2844f2a171499c58b91d1f : SUCCESS in 21m 22s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/04afff9a6e1c4dfc8a3f2034911d8754 : SUCCESS in 19m 21s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/7013a840b4c14256a635a07c98d1e5ec : SUCCESS in 17m 00s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/0645aeae570946cfbd67febd12b87934 : SUCCESS in 29m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5afecf189fd8d844d8e478e091552c85776b9190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-08 15:20:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"94cab896c7d84383374562b723c301ec17e65a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 16:29:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7241769290bc4a3f87a2578c7c82d8c5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38e621a4467a42bf86e2d6e8409aeb17 : SUCCESS in 57m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944f36ab5ddb425e8a10a16b09b077d3 : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405201a7b0784dccbc0836dd1719e156 : SUCCESS in 2m 11s\n- build-python-release https://zuul.opendev.org/t/openstack/build/46c3b267ddcf4316b3c65467d4496922 : SUCCESS in 1m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaaee5c493b54ffe8e7a4978ace5dbdb : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a520673529444138905369f9167eb8c4 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dcda19591524fd9b659f6f34cdd4103 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3daf4656f33745ce83f1b997daa2ab6f : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd6dab7023a1416caaa143d0144759b6 : SUCCESS in 2m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13f31a66704d46a79622f6e23d47d920 : SUCCESS in 6m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02b0afcb8a114444800a281653b679c5 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/633db627fbcf4d2aaa424549bde463e4 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f512ff0eeae498e9ba8fdd6b8a4fc43 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b344ea955756403ab6842535e74e3275 : SUCCESS in 8m 38s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/ce5b41ed488a495bb4ec389655f1d233 : SUCCESS in 14m 13s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/cb40f1edd69949a3bad685c669bdeb98 : SUCCESS in 34m 49s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/89f6ff6fe61345a9a5664df99274d52f : SUCCESS in 58m 08s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/1eb83c454ba4464990c7336d4fb80ac1 : SUCCESS in 35m 31s","accounts_in_message":[],"_revision_number":2},{"id":"7f5decdb76299ebbe956c55758024f26df40f03a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-23 11:50:10.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"bdef1438fa81ea1eb2b1513be62c8b3d6293b7df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 13:47:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97c15fd798554455ad0befc1a67334b8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1864448f7c84eb2b86bd73e4de4f874 : SUCCESS in 1h 49m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50b9fb61073248d5a608dbf436be3026 : SUCCESS in 2m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e0ffca9a2742b0835f1137eaa11078 : SUCCESS in 1m 43s\n- build-python-release https://zuul.opendev.org/t/openstack/build/7e400fb9c4b44b0981b79f88653f0b8c : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1370cb05a96849fab472d801e3f8ee69 : SUCCESS in 2m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bef51010eff4ce2b66dcc98c22dd738 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c399ea850954b0eba1e28f4c201580c : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/260331fc22b04c7a97d98de36134d1d9 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/652198b80a1f4893b93e513c0f387356 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e2e11624a7b4177831b662e35e23f3a : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97e46c74367e40fead3461e4f82f54d0 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4820533516a241dfb61cde0544efea0c : SUCCESS in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a6bd61827994172a96426d46d21a76b : SUCCESS in 4m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ecbd6944f59245248ce12173e908891a : SUCCESS in 11m 17s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/edadd4a3c452434095a95f9ea6e945cf : FAILURE in 32m 01s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/3440e18bb2fc4232b0ad9e8ed583e41e : SUCCESS in 33m 01s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/cf35dfce3a204a85a9c9927740d5bc17 : SUCCESS in 31m 00s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/32749764989c4a5796e311e3912b2843 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":3},{"id":"88a762d5fada2ab57134fe82aab8ee4267afd7eb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-23 16:54:46.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a0df571c39a53fec907347b45ee1391bb7c422a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 18:49:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2360686fc0884df18f018309939a5074\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e16a4c18e0da4fb995190ade426168b3 : SUCCESS in 1h 48m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af4acf82d7b4491b4095ff5f51da999 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95fcfa2d0ced47678c2ff93d4b8e7eb7 : SUCCESS in 2m 39s\n- build-python-release https://zuul.opendev.org/t/openstack/build/32c687191518402a884675d8f8776e9e : SUCCESS in 1m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1decf9d1861147c3a70fa96aef92dc6b : SUCCESS in 8m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa7672b252664c94bc651092459e1375 : SUCCESS in 2m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7880940282b4da8a8d4d6d1e8e15789 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b02c6563e42f4d658e6dd3f64abb3fb8 : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0b3fd88950d48b2b7d1ba9531314349 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c11714a5241438b9db6037a954fbdba : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3834269354b4493791dec287f946bcaa : SUCCESS in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b16e83a594e4ca8ad502d058ad0adbf : SUCCESS in 9m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eab05f5951544984becd2df337b37a1b : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c11dab8496974abe8489b9642d0abe5a : SUCCESS in 8m 23s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/6926ef45dcf94bad908a3edc7323a610 : FAILURE in 30m 30s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/e472689011c34fe88cd2c1125bd8cbc3 : SUCCESS in 30m 57s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/3e9be7b7392b477aa66862545debef0a : SUCCESS in 37m 35s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/07e48451013445938f36a1c40956be48 : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":3},{"id":"cef54363f46e4b5c2b68a664eee7aaa912544452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:25:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cf779fa8d9e3a0dbed6806b4115572034ec38544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:00:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7b67f9f231742649a2da6f935a137b9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a7299839db774ac3b271b7f2044b101e : SUCCESS in 2h 27m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa6b989325a04a0fb18ca2364c6d91ad : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0af2a26b4a14031b4c5b32ce1b5e4b8 : SUCCESS in 3m 00s\n- build-python-release https://zuul.opendev.org/t/openstack/build/f364566037c549b7a60779e371972f7d : SUCCESS in 1m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54fe135778db4ef3840d29061c45d665 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73f32fd4c01a456086958cedd350fc93 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/764d0cba669f4918889571f92673b1e8 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/950f7232a29445aea4e31703c9e5c88a : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abfb4d3e44624cd88480a9858a3a4dfd : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21ffdc6cab1f41b690b498c46730c0d8 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd8aac07f90d4c36965eabb4cd44338d : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17e611e7133c44cb8a1b65ff2de29bae : SUCCESS in 8m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/82b61c8bf8b4488bb4b22e1903134342 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d96be4e31c2475eb1463493b6cb5646 : SUCCESS in 8m 33s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/010b3504f2724795bde219b85b99b8fa : SUCCESS in 31m 52s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/6df6224991c24c7b8197991adbd05a0b : SUCCESS in 41m 25s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/234df41c2068431d977a32aff83df579 : SUCCESS in 39m 22s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/1edc309d400a40edbafb2e6ea4ee935a : SUCCESS in 16m 45s","accounts_in_message":[],"_revision_number":4},{"id":"0373d715cc305a786fff214b7072c591efe3ebc7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-24 13:41:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7249b80a885ce2d7771dbda7999e0bedce98c7b9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-10-27 20:26:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c98933f6f49d9d4ab77f8061b2e87d64f466fc47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 20:28:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bc51ee56e7d04894c80f922c71bd4d3856bba068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 22:19:04.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/pbr/+/964713 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"826e20ce231145dd126412078eae06b3affd9b76","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-28 18:12:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eac2a68f8b09a54235a7ceffca8b422b0cb5e1c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:04:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d06777aca9674a3b9cb98275c8327640\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/282a7a26299f459ea8f2b217b5046c2d : SUCCESS in 1h 46m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41cd709266ff44b6ba9c11c96d540594 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56b0a736b1e34029872ce0bd6f7747d5 : SUCCESS in 2m 55s\n- build-python-release https://zuul.opendev.org/t/openstack/build/e58bb2ad8f7f4a458e2f75dc9332d3f6 : SUCCESS in 1m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/392e358814fe4a38bcdc4f92077ffaff : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85e44c8f8dd24230b436047aa97f802a : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0c09649a65c4e5eb852e771aa0fe48c : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5b44f8176564096a50e1d0bd6224978 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b08a854c6fc45179a975a1ec8d3e5d6 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5f9b7f1952b47a1b1d66870e50b6f8d : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/616abc7e8cff4f0d8bb5b486efffba9c : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0203d40446904ba7821302408c289d4c : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9330fb1bb56f401ab7f656a137422654 : SUCCESS in 8m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/156dfeb904e542a8b22d52a6fd7d7c26 : SUCCESS in 8m 29s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/97b22273b568454f953b19ec89c43f85 : SUCCESS in 25m 34s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/3c7ee57addb24d7991ceda149150ecdc : SUCCESS in 21m 51s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/7f24140331df48bcb1155652f6be9694 : SUCCESS in 40m 02s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/22aced4fb7054f6fb9fec21db06e881b : SUCCESS in 17m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c3c0d95c6afe155ea44ef67fc34e8a4dcde37160","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:18:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6a058a779d9a27dba9c8df3ec3e62f2180531d46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:12:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aaa2520c90248e895423ecb4472f974\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e83135951c754c9eb2a8cad891fba025 : SUCCESS in 52m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef012a443405413c8d3af57049a5c42e : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b22a5a525d624fc7b4a616f32924e20e : SUCCESS in 2m 39s\n- build-python-release https://zuul.opendev.org/t/openstack/build/3cc8c7ffa8bd40a593ee2d66de3dee2f : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b65c0c49350f49fda8740c5ddd42c0ea : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/887a1cc813124b5d8532253a99abbe6f : SUCCESS in 2m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ee966429cac4d11a2febe51cd0903f8 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af198c8e7bef46029b103f9fb4f8992a : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a433452a16af499793523f4534c34565 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6469292614f4e84bfc6925c286fe0ba : SUCCESS in 9m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d13ac4729404027bd2b3bbc22c1839b : SUCCESS in 8m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c91a518e1f0445d6a3fce0a2ad0a8ab0 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/856caa8d5f0d47658e81c91f36dd9b38 : SUCCESS in 5m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8734b92ee5534aecb01561aa091494c6 : SUCCESS in 8m 15s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/9d291ceb6f3b42dd8e018f67f0a2e790 : SUCCESS in 14m 29s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/d8ab8330fc094209ab7832724af96640 : SUCCESS in 25m 03s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e7f3cef2805c47c8abceb5ec8ceb9537 : SUCCESS in 39m 46s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/19b44e90a177411b8bc9fa6c24f3346c : SUCCESS in 32m 33s","accounts_in_message":[],"_revision_number":4},{"id":"92b2cf18c048ce682d93d8f713ee18c8e63bd888","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:12:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"562f4ebb5b3f1280beab1f7d1e3e56906e2169f7","revisions":{"1d7ed93c52d5212ddb91ae16ee0cbe90215b9759":{"kind":"REWORK","_number":1,"created":"2025-09-08 11:55:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/960028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/28/960028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/28/960028/1"}}},"commit":{"parents":[{"commit":"a0d431c5dec60b674ba726323edf442c845de2c2","subject":"util: Deprecate description_file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a0d431c5dec60b674ba726323edf442c845de2c2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-18 17:48:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-18 17:48:57.000000000","tz":60},"subject":"trivial: Inline LocalInstallScripts._make_wsgi_scripts_only","message":"trivial: Inline LocalInstallScripts._make_wsgi_scripts_only\n\nWe will be moving a lot of this logic out of the class shortly.\nMake that easier to do.\n\nChange-Id: I909916dc861d086c449ceb4df4ad919a99a724d4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1d7ed93c52d5212ddb91ae16ee0cbe90215b9759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1d7ed93c52d5212ddb91ae16ee0cbe90215b9759"}]},"branch":"refs/heads/master"},"9780d9b02ed233c5f91af76a100031e088386b1a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-08 15:20:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/960028/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/28/960028/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/28/960028/2"}}},"commit":{"parents":[{"commit":"a6e049df468f67af0980c7dadb71a1da3138e5ca","subject":"zuul: Pin ansible version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a6e049df468f67af0980c7dadb71a1da3138e5ca"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-18 17:48:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-09-08 15:20:59.000000000","tz":0},"subject":"trivial: Inline LocalInstallScripts._make_wsgi_scripts_only","message":"trivial: Inline LocalInstallScripts._make_wsgi_scripts_only\n\nWe will be moving a lot of this logic out of the class shortly.\nMake that easier to do.\n\nChange-Id: I909916dc861d086c449ceb4df4ad919a99a724d4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9780d9b02ed233c5f91af76a100031e088386b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9780d9b02ed233c5f91af76a100031e088386b1a"}]},"branch":"refs/heads/master","description":"Rebase"},"3203791940f6b8ae6ecdb0cbb4a6dc93fee09d31":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-23 11:50:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/960028/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/28/960028/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/28/960028/3"}}},"commit":{"parents":[{"commit":"7181a4aa517c783256c6cdef19d31da02df4a18b","subject":"Pin zuul ansible, pyperclip versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7181a4aa517c783256c6cdef19d31da02df4a18b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-18 17:48:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 11:50:05.000000000","tz":60},"subject":"trivial: Inline LocalInstallScripts._make_wsgi_scripts_only","message":"trivial: Inline LocalInstallScripts._make_wsgi_scripts_only\n\nWe will be moving a lot of this logic out of the class shortly.\nMake that easier to do.\n\nChange-Id: I909916dc861d086c449ceb4df4ad919a99a724d4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3203791940f6b8ae6ecdb0cbb4a6dc93fee09d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3203791940f6b8ae6ecdb0cbb4a6dc93fee09d31"}]},"branch":"refs/heads/master"},"562f4ebb5b3f1280beab1f7d1e3e56906e2169f7":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2025-10-24 10:25:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/960028/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/28/960028/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/28/960028/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/28/960028/4"}}},"commit":{"parents":[{"commit":"0c98b07609c0354296795b10906aef21e222e7c2","subject":"Remove unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0c98b07609c0354296795b10906aef21e222e7c2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-08-18 17:48:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 10:18:15.000000000","tz":60},"subject":"Inline LocalInstallScripts._make_wsgi_scripts_only","message":"Inline LocalInstallScripts._make_wsgi_scripts_only\n\nWe will be moving a lot of this logic out of the class shortly.\nMake that easier to do.\n\nChange-Id: I909916dc861d086c449ceb4df4ad919a99a724d4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/562f4ebb5b3f1280beab1f7d1e3e56906e2169f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/562f4ebb5b3f1280beab1f7d1e3e56906e2169f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
