)]}'
{"id":"openstack%2Fpbr~964712","triplet_id":"openstack%2Fpbr~master~Ie666200fdad0ee2e436daa9c3e76fd50effd9d7a","project":"openstack/pbr","branch":"master","topic":"pep517","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-28 22:11:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie666200fdad0ee2e436daa9c3e76fd50effd9d7a","subject":"Remove dead code","status":"MERGED","created":"2025-10-23 17:18:19.000000000","updated":"2025-10-28 22:11:55.000000000","submitted":"2025-10-28 22:11:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964712-pep517","meta_rev_id":"8a087d674a7408db9b09ce392a2a850f1abcd76b","_number":964712,"virtual_id_number":964712,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 22:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-10-28 18:11:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-27 20:25:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-27 20:25:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-23 17:55:54.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-10-23 19:14:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-27 20:25:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"dd00d031d67b249d2a37dbbd6e052a664f539b05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-23 17:18:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8b459497499eccb08ef856e80937488dc788155","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-23 17:55:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5169a53d7ec658917254c012b707b1c386e243e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 19:14:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5eda0f5774b0483095f29412c7c5eee6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a08a60d6437c453597ff975b453b2b03 : SUCCESS in 1h 42m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e277f1160d9e4b399878f9c9c38de2ee : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4675c12885974931afdaa4f368a07e5f : SUCCESS in 2m 23s\n- build-python-release https://zuul.opendev.org/t/openstack/build/6da916ba592747c88d0db798fdee51e8 : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffa962d5f70c464aa96f6fa3bae45a2f : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0059bceecf254b6a85c9f18671e3a598 : FAILURE in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8b07d442a994b73a6b05f9884b04d4b : FAILURE in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/004f87a564da4a64bd7b4d623909ea71 : FAILURE in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40b78e0191e1474498fa5af31d7d91b3 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc636223d71b43c084f9c33d09786af2 : SUCCESS in 9m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3af62beb49f747e5b54ef3c0a179fdea : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47a968e9fa23487ba1fca4deef6539e2 : SUCCESS in 8m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/814c6765f056468c8042b42e8f2eed2a : SUCCESS in 5m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96c723ce087b4c779223dc94219b8598 : SUCCESS in 8m 17s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/0b275cc39c0649de9faffe977922485b : SUCCESS in 20m 44s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b447c146c5784aefa10eadcfc313d03e : SUCCESS in 20m 43s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9eef7ac10bf84988a4654aa5a7a397f0 : SUCCESS in 29m 11s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/44e747dfc17b470cb8236bbbbd3d4223 : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d0f113c9ae39e1e897d00a40db4d68d97aca45dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:25:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a4044652c243e6434625764fc494201d34ce7cc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 12:19:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e7ba60cc3d14cb8ab74aa992b6bd881\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08b2a056eb024dc381bd2e96a291c500 : SUCCESS in 1h 49m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d259f87af7e4a6bbbd14fa985b85bae : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7cca02f9a0f4059bf1844a65b477ac7 : SUCCESS in 2m 21s\n- build-python-release https://zuul.opendev.org/t/openstack/build/8b526e33405e4466bc1795176ac67852 : SUCCESS in 1m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b9d42253b9f47b6a99e0fa9c71692a4 : SUCCESS in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49d95be9cfaf44e3a407356e79d4816c : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8ca90167a90446b86496fe6c9d4ce97 : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a7a7c958a8d417ab4f74afced131b9e : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed46b62051b34de0a8ea80d685668f62 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71cdc13473f245d696b874a8c6a1eee4 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d766450fdd74fffbf9bad3db0eb6cd5 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a032ece8826342ab8b5df2c184797ec5 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cfd0d73f4e742088d2a53a4a361f0f5 : SUCCESS in 8m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fbe450505cd450ea6f627f43666bc01 : SUCCESS in 6m 23s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/ed7ff9492bd94952af7d936405528639 : SUCCESS in 31m 11s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/4482cceacb8946098cae78c0fa5d4dd4 : SUCCESS in 38m 23s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/7b1dcaaa3ce44addb5df4e3652c61fd0 : SUCCESS in 31m 32s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/6bf20e1bace0403c84527f938e5dccad : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4091cd9a2359d633d4e6cb0a0033758148a89ed8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-10-27 20:25:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c8853e1a98b5e9b72e28cea4a13176e8620861ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 20:27:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2cb00aa1dd594c875944800015713d817513214d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 22:18:59.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed7c6cdff643408fa4274ed519e3b3a2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0945db916fc4aec9735659272f2385e : SUCCESS in 1h 47m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df740ab2976f4fa2a00b3f69e51e44ec : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a503ed0cc074bbabf357e7b326bf143 : SUCCESS in 2m 30s\n- build-python-release https://zuul.opendev.org/t/openstack/build/3c3436bfcbba4c35bc778a8e8bb7c716 : SUCCESS in 1m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b76b7fccbdfd445ea6b1bd7db3a2243d : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/910e97d969a74fe6a30c60083fdbeffa : SUCCESS in 2m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4b5cd5412c940328b5fc419946e568c : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0249e480544448ba90871380f7b017ae : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59b618b340cf43849ca0e22dfaa28b44 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c0158e0f3af457682b3d6746e92bf22 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5a081fa56854a67a5a6fa2b9ceb16c7 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4bd1cf7a4bd4db0a8ba606e783b7c4e : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15e35c5b663043d2bdbb0ec88388f6db : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d73d296f5714203a44aac1a4fc140a5 : NODE_FAILURE Node(set) request b10708ed285648ed9a38be54ce13288a failed in 0s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/3331f228fd65478ea1a4ef3a61ebc8a2 : FAILURE in 30m 16s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/7b3d397b3eaa4067821e7be00dbd6141 : FAILURE in 30m 42s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/c28ec433389b48e2a1862778bdc64c82 : FAILURE in 31m 58s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/015b54ea7da440bfbc881deb8534ecd3 : FAILURE in 37m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4017131eb12703154a5fac4a940120dba58f4759","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-28 18:11:31.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ded3e8e9dceb00f5bec5ebbe82be64052fb0456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:17:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21dc54a3a14d4849b5e9a2a40683ad07\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2233218914c643c3ba88545337241b7a : SUCCESS in 1h 59m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2fcb7e249e49889e12701a90b2190c : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec3679e42fa437d8d929fad8add2f78 : SUCCESS in 2m 55s\n- build-python-release https://zuul.opendev.org/t/openstack/build/85145d07874b423f8483a1a10b1d9588 : SUCCESS in 1m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee5be3792b774213b452d2c20c57c336 : SUCCESS in 7m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d0eb6a957ba493e9ca596edc3c01c4b : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/518d40ab83134b7dbdbb434941de2487 : SUCCESS in 2m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a819ced6adca4375b82d6c49e5a9da63 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/857c592fa16c4f9480e5a6035ab49345 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4baa32f3fa944b028bd938225a391e62 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9ece1256bed4c7d9fd64011ba46183a : SUCCESS in 7m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddcae98648c74747a01c175973109c5a : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6feabe9ec66b4a71b8da5b6f20a65626 : SUCCESS in 7m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bc699484b9644d2866c871432780774 : SUCCESS in 6m 18s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/b12e15c0a31a49b08abb598fd2b35146 : SUCCESS in 35m 15s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/490b6599875e48e090bf2d40c5ddf7d3 : SUCCESS in 40m 28s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/487df732fdcd4368bb6a7e792de0a67e : SUCCESS in 25m 55s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/8c6344d46f814548ac45ff49a4dd6ba8 : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":2},{"id":"7592a54032460b341e4472fc069b5a45c79d03c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:18:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cfb7a99bf4b95d6c5c4db7acc7739bcbf813fc56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:11:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a88a01c1f17485e9b92918cccbd9bd6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92c4019d65954165b0062318c37ead17 : SUCCESS in 1h 47m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/307b7838c73b4de1aaecc5468266b243 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77024c5f305c41e5ba9cdf2ff1ae2db7 : SUCCESS in 4m 09s\n- build-python-release https://zuul.opendev.org/t/openstack/build/7ded78a94fb34ef483546c9e72c02cbf : SUCCESS in 1m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21ea539b8382419d86a574b9c063922d : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eed0eb4c4caf456ab7bb674fc625d5f5 : SUCCESS in 2m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b07645d36194898b8e7e29b9d1d070e : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a14f63d5b8046fe83aed1a01c9d7be4 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/702a8784a7e446dd8ba56ca8d2148eba : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f37c7ef75e24c638002d065f7bdbbbb : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/808568fc8c2c4e7aa86a9f86c118ce40 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/822b4f92b4ea4b4ba9ccf3f22b4f0261 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48948595042a4e7a9c3fab33d9e43c94 : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/006b5fa99f604b4bb91099eaabceda7c : SUCCESS in 7m 44s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/934970ed48ce450b9f9b0aabf40341b7 : SUCCESS in 29m 03s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/4144fe7a9e6a4fbd97eac97addb3dfb0 : SUCCESS in 24m 22s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/56474f768671480b835bd42e740a81d2 : SUCCESS in 37m 47s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/154b83bcf7384f00aa5a07397774b994 : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":2},{"id":"b41811d8abe8b7b1417a37a25283787627d4fba1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:11:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8a087d674a7408db9b09ce392a2a850f1abcd76b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:11:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13b4359e1be244868703d56ab9a124fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db36a370d0204f3e9a6223fcc2828fd5 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f4c91de312f09eb0e251bedec2e1664b72a327a5","revisions":{"e0fd2be6684fcdd920a43b84b311f5bb1704872f":{"kind":"REWORK","_number":1,"created":"2025-10-23 17:18:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/964712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/12/964712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/12/964712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/12/964712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/12/964712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/12/964712/1"}}},"commit":{"parents":[{"commit":"d95512191b27fa3d4a2446c21069223eabd0d7e9","subject":"Remove remaining user of pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d95512191b27fa3d4a2446c21069223eabd0d7e9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 15:55:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:20:03.000000000","tz":60},"subject":"Remove dead code","message":"Remove dead code\n\nThis class has not been used since commit 32c90ba598 some 8\nyears ago.\n\nChange-Id: Ie666200fdad0ee2e436daa9c3e76fd50effd9d7a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e0fd2be6684fcdd920a43b84b311f5bb1704872f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e0fd2be6684fcdd920a43b84b311f5bb1704872f"}]},"branch":"refs/heads/master"},"f4c91de312f09eb0e251bedec2e1664b72a327a5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-24 10:25:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/12/964712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/12/964712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/12/964712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/12/964712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/12/964712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/12/964712/2"}}},"commit":{"parents":[{"commit":"7181a4aa517c783256c6cdef19d31da02df4a18b","subject":"Pin zuul ansible, pyperclip versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7181a4aa517c783256c6cdef19d31da02df4a18b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 15:55:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 10:02:02.000000000","tz":60},"subject":"Remove dead code","message":"Remove dead code\n\nThis class has not been used since commit 32c90ba598 some 8\nyears ago.\n\nChange-Id: Ie666200fdad0ee2e436daa9c3e76fd50effd9d7a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f4c91de312f09eb0e251bedec2e1664b72a327a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f4c91de312f09eb0e251bedec2e1664b72a327a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
