)]}'
{"id":"openstack%2Fpbr~964714","triplet_id":"openstack%2Fpbr~master~Ia7bec21525ea2789b635fbb03776f722f965a45e","project":"openstack/pbr","branch":"master","topic":"pep517","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-10-24 13:59:45.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-28 22:44:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7bec21525ea2789b635fbb03776f722f965a45e","subject":"Rename setup.cfg parsing utilities","status":"MERGED","created":"2025-10-23 17:18:19.000000000","updated":"2025-10-28 22:45:03.000000000","submitted":"2025-10-28 22:44:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964714-pep517","meta_rev_id":"0c90e5eaf51c46c008cac2c4808429f2f7f6ef96","_number":964714,"virtual_id_number":964714,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 22:44:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-10-28 18:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-27 21:12:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-27 21:12:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-23 17:58:30.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-10-23 18:50:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-27 21:12:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"963e2ff11d9cbc43c82c6ed3be3b9cfa2efa53b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-23 17:18:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"108e453408af03d8e821f0a20bd61f5167f4acc2","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-23 17:58:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8499fa8ae9ea6513218b3d0a465f07dc2f10db90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 18:50:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d21219f4a80d4c8498406c5473013eb9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d4c6242384b941318834281ff88a3a50 : SUCCESS in 2m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6e0e4a7d31f472d8bf56196760f632f : SUCCESS in 1h 27m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d465158cd547cf84543f5e1af776f7 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c61a3d27ce084a89af6b39e66be14e0a : SUCCESS in 2m 23s\n- build-python-release https://zuul.opendev.org/t/openstack/build/060f47f3c5af4bc6802d4dd34db9567b : SUCCESS in 1m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a6840bedc1b43afbb1628d63c761105 : SUCCESS in 7m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1560b6b7e5894044bfacf2017bb2a888 : FAILURE in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49199ce6fd2645da8c2c3afb190bb8bc : FAILURE in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33a8b01ac8b24896b10ae4774594c9cc : FAILURE in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37918624fe814c8998934b2843b4fc72 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bb34d6c698d4310a5ea942b2bf7f67d : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fdd48044921f4bc38d37468231dfac35 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b47315dcf794e5d9f27561e530eb5ea : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f730c8eb165e46b1844ff31b24576ae2 : SUCCESS in 7m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8f265c44f8e4838b837e42e763247a7 : SUCCESS in 4m 10s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/147ea782ff9c4299bd769fc381055443 : SUCCESS in 31m 09s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/0aa2cf49790b4491ae9570331a7f98f0 : SUCCESS in 38m 58s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/4120479ad01d4952a48860d619619a4a : SUCCESS in 32m 39s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/9d190b71740e4f03bd4eeaa390bf767f : SUCCESS in 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"33dbd02ea4bf905c3aa1dbe323c5234266c535a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:25:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"10bd18cc2b4edd7544d40406df77b39b0465731c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:30:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86c38997cb7216d64d56693381fa0ba59c4bd514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:33:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"cb0912b913b3a64008d6d77daa050138fbb1e098","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 11:10:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f67ff93b97d9e584ebdefe80fb5dfe695d75814a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 11:58:57.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"71e32bc212b44e5c3966a8b9e347d82415ca7619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:32:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6d5544e5ff74c8a8b229c90757687c4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa14fdd5e3ed49869d214f1806537d3c : SUCCESS in 1m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a30a7d23d7dd490386ad9b72f3bd4c4e : SUCCESS in 1h 29m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faaa291e08ae4e4486d1625d22300a9a : SUCCESS in 1m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ad9ac797f454390b7090fb6ad072c59 : SUCCESS in 3m 01s\n- build-python-release https://zuul.opendev.org/t/openstack/build/e21a9f2bf52c4477a2234c94fa9a6b45 : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46f3996f32a94094a5260969881f33c9 : SUCCESS in 8m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f663bf6592524af2bd18e0dd1f327e0b : SUCCESS in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1f348a332fd49bcbcb6babb154e8275 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5026d17422be4a3f837d3e0e18948b30 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3599863f0d564f16bba6f3a1e1d4edb2 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a68bf2b112984cac99059367989035cc : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9b2f8437b62475cac68289de95b868f : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb12805f43da4e3ca6c6c46a745efc8a : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98701270e2b7490db91885c154d0f430 : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d328de103a0f48bf869d5e8040955cf4 : SUCCESS in 9m 18s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/cdecb016ede14060a23971fc87cc6368 : SUCCESS in 17m 23s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/fb5589ac59124025b857c275d8f011d3 : SUCCESS in 38m 27s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/2f6370e37f104f06878ff435591bc9f4 : SUCCESS in 36m 02s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/36df758ffbbe4a67812cf597960f01f3 : SUCCESS in 16m 12s","accounts_in_message":[],"_revision_number":5},{"id":"a7af7fed4ceb42541c9b4f92afde94803ef14dd1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-24 13:59:45.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"eeb736f2328ae9ce435aa8a4dc275176e75d680f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 14:48:48.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fddfd5a82c547d90d6d2ea81d9d130e4e73c030d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 16:34:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/808e9fe3c74f4c83b9d09acf566dc69f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8f54ac98b45e4d0ca47a3cf16d49ac8e : SUCCESS in 2m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38ee70e00b744396a097e425cf39afd0 : SUCCESS in 1h 41m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bad9992104b4ecea5350c55c2b2b204 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74f6508777c2466fb577ad836a952a8b : SUCCESS in 2m 53s\n- build-python-release https://zuul.opendev.org/t/openstack/build/6c160c8a3e2b4edfa983c7d8c22cccbb : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24abbd059efc4139913f5fff87c74547 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/198c5d58b93446fdbd87009c9451b7ae : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01d6db59d5eb439ab93034a8c78d2737 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7e84fa4946041f582a09cde7baaceed : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bb2c676b4184bae9b9a851002c19b0e : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a91f05e86b1d456e804d7a41c4a0bca1 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97e58931d1db407ca1e127253d855be5 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6728683426f645898df91750ad99d1b7 : SUCCESS in 6m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3038e7a0bf3e44308ebad4980e17b57f : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/916fb13bd3914b8e8f400908cbfe6888 : SUCCESS in 7m 58s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/8efe064cdb904021a6a348038427754f : SUCCESS in 16m 13s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/bd22c577ccf64d8787506581fd271883 : SUCCESS in 37m 28s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e25e43ebc49446189fe817ed99f4ef80 : SUCCESS in 37m 06s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/1b6c79bc510b4ff788dd0c3afd759c13 : SUCCESS in 29m 04s","accounts_in_message":[],"_revision_number":6},{"id":"10e82d4c8e162d024246cd1f83a8ab31a14797a4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-10-27 21:12:20.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"21a571aa20becf9684292009b3b9952edf1d7f6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 22:19:06.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/pbr/+/960031 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"155015fe687e57de2eff8a95056de960b1265f90","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-28 18:12:40.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8fc0a3c1a0c220a31a739a6ea98173dbd9a248ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 19:43:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7db710e10794b46aa01c495895a9be8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ebe7c7d8e16641df9cd4e9821d34bbed : SUCCESS in 2m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6015b41ac2f64c278deaef31c555196c : SUCCESS in 1h 27m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d5005bf52948118385a3aea4c27e15 : SUCCESS in 1m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8844dc072b7a4d2a9f270cff633438ac : SUCCESS in 3m 14s\n- build-python-release https://zuul.opendev.org/t/openstack/build/f1cd7921b7a640d6a6cd95ee7af363f6 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3747c48e4cf946da857630739b9dd796 : SUCCESS in 8m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a4c3079a60445d2b1fc6581735388fe : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bded198b956443a2b4a3f203a9dbdfdb : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e03198751fe4434a83c24abd55087f9 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44e9db5a76274f53a8b14ae88e793071 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f867d1166c74d0a84aa4d47dea3b7b5 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da42bcb6c09c4a5289b69ca2ce30d679 : SUCCESS in 8m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c010c9395f04da9903f0cc591c646c0 : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f93b03b75e214e149c4fcb578e80b686 : SUCCESS in 8m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cdc552dd5aa4375a42cfbedf9acaefc : SUCCESS in 7m 37s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/16b07eb219aa451e99d733a29f3f20c9 : SUCCESS in 25m 54s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b19a55e68cca4142ab8a6e961ccd0569 : SUCCESS in 35m 56s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6b1652800b2b4577b874f341695a71d4 : SUCCESS in 24m 06s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/cf61b4f877744863b27aa1b3e0a29a0e : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":6},{"id":"4b89526c26825be13a2018c7c06b02ae5c68c76e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:26:15.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"03ea3c0b49f1178854148f5d51a41ff43faf226a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:44:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/254a76c1c2284ce8a919e15374c8d093\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5082c0825210415cac78c3397081aa14 : SUCCESS in 2m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/98b10cf4b11947b592760cf87b89357f : SUCCESS in 2h 13m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b4ea50653884a57af9c8263d0ad1406 : SUCCESS in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08df08315f934cc4b507e4b53a8e3a20 : SUCCESS in 1m 53s\n- build-python-release https://zuul.opendev.org/t/openstack/build/da4674a59890403888e66a01dd3b09df : SUCCESS in 1m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f425ff787904f67be9a0b9f53671ba1 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a22817994da649139b96a250aa08632e : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50d10ae5c10c4d66aa07433d0a0c5058 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b54e62d4a80b4212829b17f1795f02e7 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29852611cdd745b1a2941f56beafe44e : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3aedc5afa3684967b8f62c76fbc0094a : SUCCESS in 8m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed3d1b53df654f26abf4d4f4a3cd1b8c : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80b5535b8d6f45a18738e6d57b4e2e82 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/962d48cd57254565acf58cd16e14dc58 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb85f1424aad409287d7ee2eef7ec4d8 : SUCCESS in 8m 26s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/eae5f620d1914b80ad4bb7ca633e6c2c : SUCCESS in 24m 45s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/0c47852fde3d428a93cdb1a43f4aca68 : SUCCESS in 14m 29s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/02afdf990e7649dd98bd54418781c305 : SUCCESS in 38m 20s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/f820fd20009c4f52949cf5e92a8d221c : SUCCESS in 34m 29s","accounts_in_message":[],"_revision_number":6},{"id":"949b92176991e80a247ae060c5658e99faccbdf9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:44:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"0c90e5eaf51c46c008cac2c4808429f2f7f6ef96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:45:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f21eadb5884a58b4d394704762f86a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e342803ebb014286b739f44896b59d9f : SUCCESS in 35s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a3cf42d4941938e57c9581c0338cd79a5d795e96","revisions":{"91157285f676b43a5c6e349a95b53bd573aaf01b":{"kind":"REWORK","_number":1,"created":"2025-10-23 17:18:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/964714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/964714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/964714/1"}}},"commit":{"parents":[{"commit":"b77203d24ea8fb6e3d336aa6886151a9d5fa63c6","subject":"Remove unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b77203d24ea8fb6e3d336aa6886151a9d5fa63c6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:00:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:20:03.000000000","tz":60},"subject":"Rename setup.cfg parsing utilities","message":"Rename setup.cfg parsing utilities\n\nWe also tweak some quoting.\n\nChange-Id: Ia7bec21525ea2789b635fbb03776f722f965a45e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/91157285f676b43a5c6e349a95b53bd573aaf01b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/91157285f676b43a5c6e349a95b53bd573aaf01b"}]},"branch":"refs/heads/master"},"18647a4bc391f155bb1ba5dd4c7c2c029a0bde97":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-24 10:25:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/964714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/964714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/964714/2"}}},"commit":{"parents":[{"commit":"52b35f8ff866b187a28fa6debf70598827c083cd","subject":"Remove remaining user of pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/52b35f8ff866b187a28fa6debf70598827c083cd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:00:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 10:25:15.000000000","tz":60},"subject":"Rename setup.cfg parsing utilities","message":"Rename setup.cfg parsing utilities\n\nWe also tweak some quoting.\n\nChange-Id: Ia7bec21525ea2789b635fbb03776f722f965a45e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/18647a4bc391f155bb1ba5dd4c7c2c029a0bde97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/18647a4bc391f155bb1ba5dd4c7c2c029a0bde97"}]},"branch":"refs/heads/master"},"5b33bebaa8ab689300ae45a8ba204fa0590defbc":{"kind":"REWORK","_number":3,"created":"2025-10-24 10:33:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/964714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/964714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/964714/3"}}},"commit":{"parents":[{"commit":"52b35f8ff866b187a28fa6debf70598827c083cd","subject":"Remove remaining user of pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/52b35f8ff866b187a28fa6debf70598827c083cd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:00:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 10:31:48.000000000","tz":60},"subject":"Rename setup.cfg parsing utilities","message":"Rename setup.cfg parsing utilities\n\nWe also tweak some quoting.\n\nChange-Id: Ia7bec21525ea2789b635fbb03776f722f965a45e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5b33bebaa8ab689300ae45a8ba204fa0590defbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5b33bebaa8ab689300ae45a8ba204fa0590defbc"}]},"branch":"refs/heads/master"},"dcb32acf40b7ffa674a686c95c35c52adface93b":{"kind":"REWORK","_number":4,"created":"2025-10-24 11:10:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/964714/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/964714/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/964714/4"}}},"commit":{"parents":[{"commit":"52b35f8ff866b187a28fa6debf70598827c083cd","subject":"Remove remaining user of pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/52b35f8ff866b187a28fa6debf70598827c083cd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:00:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 11:10:35.000000000","tz":60},"subject":"Rename setup.cfg parsing utilities","message":"Rename setup.cfg parsing utilities\n\nWe also tweak some quoting.\n\nChange-Id: Ia7bec21525ea2789b635fbb03776f722f965a45e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dcb32acf40b7ffa674a686c95c35c52adface93b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dcb32acf40b7ffa674a686c95c35c52adface93b"}]},"branch":"refs/heads/master"},"609bf18285b661fd9b5109ddfbcd4e878ea5a04d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-10-24 11:58:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/964714/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/964714/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/964714/5"}}},"commit":{"parents":[{"commit":"5a77fc21c857ec77ecdfee282e3088b08a78b423","subject":"Remove remaining user of pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5a77fc21c857ec77ecdfee282e3088b08a78b423"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:00:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 11:58:51.000000000","tz":60},"subject":"Rename setup.cfg parsing utilities","message":"Rename setup.cfg parsing utilities\n\nWe also tweak some quoting.\n\nChange-Id: Ia7bec21525ea2789b635fbb03776f722f965a45e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/609bf18285b661fd9b5109ddfbcd4e878ea5a04d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/609bf18285b661fd9b5109ddfbcd4e878ea5a04d"}]},"branch":"refs/heads/master"},"a3cf42d4941938e57c9581c0338cd79a5d795e96":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-24 14:48:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/964714/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/964714/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/964714/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/964714/6"}}},"commit":{"parents":[{"commit":"1e1c6d5c7d3738d0b4eaf83bee051a1f3ba623ac","subject":"Remove remaining uses of pkg_resources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e1c6d5c7d3738d0b4eaf83bee051a1f3ba623ac"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:00:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 14:48:40.000000000","tz":60},"subject":"Rename setup.cfg parsing utilities","message":"Rename setup.cfg parsing utilities\n\nWe also tweak some quoting.\n\nChange-Id: Ia7bec21525ea2789b635fbb03776f722f965a45e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a3cf42d4941938e57c9581c0338cd79a5d795e96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a3cf42d4941938e57c9581c0338cd79a5d795e96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
