)]}'
{"id":"openstack%2Fpbr~964715","triplet_id":"openstack%2Fpbr~master~I6591bf05d01cc72cbcf64a43fc7bef5a9d708cb4","project":"openstack/pbr","branch":"master","topic":"pep517","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-28 22:44:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6591bf05d01cc72cbcf64a43fc7bef5a9d708cb4","subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","status":"MERGED","created":"2025-10-23 17:18:19.000000000","updated":"2025-10-28 22:44:22.000000000","submitted":"2025-10-28 22:44:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"964715-pep517","meta_rev_id":"4943c8a8d5317e66125bc51618b751ebbf3e75ec","_number":964715,"virtual_id_number":964715,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 22:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-10-24 19:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-27 21:51:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-27 21:51:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-23 19:19:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-24 19:45:27.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2025-10-24 19:48:34.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-10-27 21:51:07.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"32b155b5d18123034d6dd12693803b14cfd10a90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-23 17:18:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4996d449124733a64e2da9f574918fabb9fdc375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 19:19:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63e73a1f311741a08dcab2db586d0e66\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/19b0eb06e6544b69a59ad33cf6448a1b : SUCCESS in 1m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/710742f1d0a64f1e86249189310cf427 : SUCCESS in 1h 52m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a93bb98dba8046b2b98867be1a156609 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293db4762ee948bcb387d1bcae006c16 : SUCCESS in 1m 42s\n- build-python-release https://zuul.opendev.org/t/openstack/build/82cc9433525340508f63be8b5286e6dd : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e5eecee67954fc7b4b9bfab41600b35 : FAILURE in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c26e5fcceab64414b574f145747ccf2f : FAILURE in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94021a4fdf544f0784540f5c7c5959f9 : FAILURE in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9e75b872380408a81843ebb28155ada : FAILURE in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c940907f6e742aaabce8c2e75341925 : FAILURE in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ebf8aa766bd48efa8758679eb87e8cf : FAILURE in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b188fe1f3a0f4a5d81db998527485c8e : FAILURE in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1201cdd336b4fc5963987cc2a2fc678 : FAILURE in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e25f1106f85647f0addcfe28abd72100 : FAILURE in 5m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/870ec6b6933644498b611b1f87361d6a : FAILURE in 11m 20s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/92b27ed310704795b1d4b2aca5488183 : FAILURE in 25m 46s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/d37c938e75f34601a64b76d722dc6e64 : FAILURE in 30m 10s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/28450e8720fb4f589f52fad09c4607fa : FAILURE in 42m 50s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/831f383c0a5c4a6fbdfaef3759da20c7 : FAILURE in 35m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fba83fa0e6fe690368f195a5d22234ac7b0db497","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:25:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5058e52e8e1b284248036b39cb0e920fcaca96d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:33:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"494dd57626504aea3defb1d512f1838569a89289","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 11:10:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e66172c1038eabb0bd953d9325d24b78d242c2d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 11:42:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6e9521b5fe7d52f764b996b05c686dda13b715e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 11:58:57.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"22a1b9e6deb9d13af0a46c9497d8be55b4045de3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 13:35:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5090cf225e64ea7ae3e1494f119d454\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6d3fe49a379342048a240845579b30a5 : SUCCESS in 2m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f57d03ca1eaf4952bc513fd382fd1c19 : SUCCESS in 1h 32m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e4ee8324db4f278970576695a6ce9a : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/691be7dd8f0a43eb84d013232ef2d2d7 : SUCCESS in 2m 31s\n- build-python-release https://zuul.opendev.org/t/openstack/build/fd7417a2607049ab8169cdc11c7ea4b9 : POST_FAILURE in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc88ac5dda4144e8b9eeb7033a69bf3c : FAILURE in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/06597a9d5bc84dc7bea41f59b96a123b : FAILURE in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a9c187faeb84e31837e758b53158e2e : FAILURE in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/139dd2b5eddc46019ffccb280e3895eb : FAILURE in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21678e73542947d89b55feb228ddfc42 : FAILURE in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a70aefa337004fc7a25f3292e1a07849 : FAILURE in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc76d7c582bb442790739449c4800019 : FAILURE in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c535a4c8890649c8b1c44b657dc97b95 : FAILURE in 5m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b230c221be2b4dd289d2abc2ea90fbcc : FAILURE in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b44e1586eaa41e9ab7de84532aaecfc : FAILURE in 8m 38s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/5683291eea124818acf5160d8b675f5c : FAILURE in 27m 16s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/950dc6baab7546eb85d6c3dbc2aea5db : FAILURE in 29m 17s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/187a9aa32e6949a187fa6a1ddc5f3a36 : FAILURE in 37m 02s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/8406a36fb4304a3da76d4320a4209f77 : FAILURE in 31m 39s","accounts_in_message":[],"_revision_number":6},{"id":"41e3b499e14f3fd71925d6c4f067b485a473f087","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 14:27:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"916bc9241503a03e4085d91c66817e7dd222be2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 14:48:48.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"abee7d4d9936cb8de8e0b3f2e058a3ec24428a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 15:41:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/040816e71090408fbaf1e40a7ac19327\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5502ba7430674c2bad2171b9192fe87c : SUCCESS in 2m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2731cc8306de4b0a9b08f85b38aa1364 : FAILURE in 46m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca12cb1b343c4153bcffe75d4f59cb09 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387d174fe39d4f609154b298c78d3e01 : SUCCESS in 2m 47s\n- build-python-release https://zuul.opendev.org/t/openstack/build/8c9d0b110b8f49979150a451aee1039d : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03254c7daf9c4551a2154890744a27f9 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e918c52a74d24aaf9e1d9005bc066aad : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4ce8fcb7f9848ada7409611bc2e4c27 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14d791e4b413467abde61d61474a5b88 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8da5344db2624593b679544ef3c1ad1c : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c216b37d143240958652ed676723d4f6 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3df9274449b846628d207610dd4f2ae2 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e765c517e23743e3961a5ce267fbcf10 : SUCCESS in 8m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed233dc79ff34df9a10f45361a255ebd : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97fdee8bd2e44dbbb503e21f50689d71 : SUCCESS in 4m 15s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/b5e87f27f4bd45b6a7915e3186e4d2ad : SUCCESS in 29m 30s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/f68ed707667340adb6f16555d92df8df : SUCCESS in 33m 40s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/892b204bb39c4facb0e4d76895bd007f : SUCCESS in 34m 58s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/651a0775f47e433297af5dac8c1bc974 : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":7},{"id":"3db277172204ef95b1f676f70d111fad2a4a8454","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-24 19:45:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"99ac01a437eaf918fc69ca366a14eaac3696bf79","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-24 19:48:34.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c3f3e13289332795f67737cba4e20f9f07e903b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 21:47:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e82ec2f50f6f4516b904489042709b6a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c0f699c9f8c74fd4a21aaffce8bc8f12 : SUCCESS in 3m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1de4e01c14ff4b2ab651aa2f47d7ea1b : SUCCESS in 1h 56m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb9eca8c6b5f44a4b8c19e763f0360ff : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be89b71850c44c48ba981c30790f0d1a : SUCCESS in 2m 45s\n- build-python-release https://zuul.opendev.org/t/openstack/build/ae221678e03249999d5ee996f3ab55d9 : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/975974b4057d474aa59d969bda20acab : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e554de33c8c455bae2cb14398a6d438 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/640844386ddb460ea3b182d86f7a4003 : SUCCESS in 2m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e30d2e79266444d9c215092e19d6fd1 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f250ba4139724d57908537a1b5d6dd00 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36748cbc7f6c4674b62b73a2d6c9b714 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/089556385d184a138a75a840d1ff2545 : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b6f72e93d8f4ea1aa02fe235b6c02a4 : SUCCESS in 7m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/940a1228f8ad47f79d6097cc1b3e2e0a : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2255cbca36be4884ae3f393c6d9132ea : SUCCESS in 8m 06s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/cb8da5909e0945b9952da880296efd27 : SUCCESS in 16m 25s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/580fa306b3374cd8bd3a057e0b72a501 : SUCCESS in 19m 25s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/b7ecc69758ed4f889465881b607086b1 : SUCCESS in 16m 49s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/871ae7a97e4e4213b38a5b256626794b : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":7},{"id":"6e4aed161eda74627baf5a881c8294e7e586c442","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-10-27 21:51:07.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6857d960cacdd9709dc7828311413ff27437faed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 22:19:06.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/pbr/+/964714 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"d06253c147d5970c03d460550f8fbd17889a25eb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-28 18:12:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b51d6203c82901e828d738a9537011e5f0b3a1d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 19:55:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2eff0dca314b60bbd05a68b824ee5d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/726d5b3e0bed46379e6c44bda2192739 : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa10c7940b6a4fb5873b6b774f57714a : SUCCESS in 1h 36m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0c269cdb394b42b111f777d85c89c0 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e98488f7444d52b800016a3347be19 : SUCCESS in 3m 00s\n- build-python-release https://zuul.opendev.org/t/openstack/build/76d494b2e32c441f9c10ffb156b262f4 : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/845504cd60c14e1aa86867ccb7686680 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7362cecfc144158a46b63433e6fa41c : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c92de515f42849ce9b76d2ab3b627742 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eea16e5229ec49589491ebe667578bc3 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42753ff1bdf845d994a4e369f44ff1d0 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9ee87a640234bb2be6af284cd0fe606 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5932eb0965a24d25a4566f18cf7b9bcb : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9fc22dfaabaf4dc797a94ad7acc3407e : SUCCESS in 9m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cbf628a5cbf40ae97c902cbc4e00ab4 : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b89a360cb2964fa3ad370b6d469a58d1 : SUCCESS in 8m 48s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/9312239512b74fb59dc29369b8e2f17b : SUCCESS in 31m 23s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/dd8853fec7a04c049123ee25bd8abab2 : SUCCESS in 26m 47s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/729cfd132a164e768c0b8fd458bbef3b : SUCCESS in 44m 22s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ea9910eb5e014764a62da6f08af04699 : SUCCESS in 32m 45s","accounts_in_message":[],"_revision_number":7},{"id":"ab16178c0ed5996eac1b086e302df45f1aa09780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 20:26:17.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ae24b49bae76a656cb1c7c5458d24c7cc47d43cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:44:22.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df6069904e174bbca81a3ba3bea15791\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/427f3843dff04d34ac9329e2a4178b4e : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0c62c7c24e7b44759c0a2b4ad0411e95 : SUCCESS in 1h 21m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f192b37bcb1845b9801ab9b9b4c4f737 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8be07cc1ffb6448687c1122745da3b17 : SUCCESS in 2m 33s\n- build-python-release https://zuul.opendev.org/t/openstack/build/cfb38bccedcb4affaa7b1faa6c8d64b1 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b5519c0dd6c44f788f426a23c2c7ed0 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb943ae0c35248c089d341099b6e82a7 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67ca004fa3144e26ab8359aa457baa74 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9af8562639b54c0e93e3ff13752f4a5c : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2afe198b7a804bfb868f8443c8c1150d : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61bf2c4299f64ffa8494afd32afb16cc : SUCCESS in 7m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/471a4d6bb1494d66ab661ffcb99abcdf : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/843df5b41fda4a819e9c885ea14485ba : SUCCESS in 6m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d1494d6ef3f483397fc08010c6c2ab4 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c3e6e1bc2f3404f92deb0f063f71e15 : SUCCESS in 7m 50s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/17ff13294b80414aba806ffe8478ec95 : SUCCESS in 26m 31s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/14b268910dbe49a9a6909f4e7dbe96de : SUCCESS in 28m 19s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/3b19f08fbaba4df08726b4a58c32841b : SUCCESS in 33m 17s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ebb5da0418004a0b9cb384d5312a971d : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":7},{"id":"4943c8a8d5317e66125bc51618b751ebbf3e75ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:44:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e479527c0eb5429ebcf60387fffe7c088338f590","revisions":{"326e094c79c79b70ca565442c7662bcfd8c757d3":{"kind":"REWORK","_number":1,"created":"2025-10-23 17:18:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/964715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/15/964715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/15/964715/1"}}},"commit":{"parents":[{"commit":"91157285f676b43a5c6e349a95b53bd573aaf01b","subject":"Rename setup.cfg parsing utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/91157285f676b43a5c6e349a95b53bd573aaf01b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:18:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 17:14:29.000000000","tz":60},"subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","message":"Merge pbr.core, pbr.utils into pbr.setupcfg\n\nThis module now contains all functionality related to parsing the\nini-style setup.cfg files.\n\nChange-Id: I6591bf05d01cc72cbcf64a43fc7bef5a9d708cb4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/326e094c79c79b70ca565442c7662bcfd8c757d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/326e094c79c79b70ca565442c7662bcfd8c757d3"}]},"branch":"refs/heads/master"},"86813a9b60a1a4f706be1ea2040485ac277e28a4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-24 10:25:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/964715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/15/964715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/15/964715/2"}}},"commit":{"parents":[{"commit":"18647a4bc391f155bb1ba5dd4c7c2c029a0bde97","subject":"Rename setup.cfg parsing utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/18647a4bc391f155bb1ba5dd4c7c2c029a0bde97"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:18:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 10:25:15.000000000","tz":60},"subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","message":"Merge pbr.core, pbr.utils into pbr.setupcfg\n\nThis module now contains all functionality related to parsing the\nini-style setup.cfg files.\n\nChange-Id: I6591bf05d01cc72cbcf64a43fc7bef5a9d708cb4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/86813a9b60a1a4f706be1ea2040485ac277e28a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/86813a9b60a1a4f706be1ea2040485ac277e28a4"}]},"branch":"refs/heads/master"},"dac861b54c900e0567fdd64afe816a2635dd45eb":{"kind":"REWORK","_number":3,"created":"2025-10-24 10:33:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/964715/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/15/964715/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/15/964715/3"}}},"commit":{"parents":[{"commit":"5b33bebaa8ab689300ae45a8ba204fa0590defbc","subject":"Rename setup.cfg parsing utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5b33bebaa8ab689300ae45a8ba204fa0590defbc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:18:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 10:32:00.000000000","tz":60},"subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","message":"Merge pbr.core, pbr.utils into pbr.setupcfg\n\nThis module now contains all functionality related to parsing the\nini-style setup.cfg files.\n\nChange-Id: I6591bf05d01cc72cbcf64a43fc7bef5a9d708cb4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dac861b54c900e0567fdd64afe816a2635dd45eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dac861b54c900e0567fdd64afe816a2635dd45eb"}]},"branch":"refs/heads/master"},"7349aad5e97febd73c8112f1732bf7b819b1c419":{"kind":"REWORK","_number":4,"created":"2025-10-24 11:10:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/964715/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/15/964715/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/15/964715/4"}}},"commit":{"parents":[{"commit":"dcb32acf40b7ffa674a686c95c35c52adface93b","subject":"Rename setup.cfg parsing utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dcb32acf40b7ffa674a686c95c35c52adface93b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:18:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 11:10:35.000000000","tz":60},"subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","message":"Merge pbr.core, pbr.utils into pbr.setupcfg\n\nThis module now contains all functionality related to parsing the\nini-style setup.cfg files.\n\nChange-Id: I6591bf05d01cc72cbcf64a43fc7bef5a9d708cb4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7349aad5e97febd73c8112f1732bf7b819b1c419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7349aad5e97febd73c8112f1732bf7b819b1c419"}]},"branch":"refs/heads/master"},"870d3ad89507984580baacd9b27ad1e2d8bc41b4":{"kind":"REWORK","_number":5,"created":"2025-10-24 11:42:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/964715/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/15/964715/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/15/964715/5"}}},"commit":{"parents":[{"commit":"dcb32acf40b7ffa674a686c95c35c52adface93b","subject":"Rename setup.cfg parsing utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dcb32acf40b7ffa674a686c95c35c52adface93b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:18:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 11:42:02.000000000","tz":60},"subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","message":"Merge pbr.core, pbr.utils into pbr.setupcfg\n\nThis module now contains all functionality related to parsing the\nini-style setup.cfg files.\n\nChange-Id: I6591bf05d01cc72cbcf64a43fc7bef5a9d708cb4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/870d3ad89507984580baacd9b27ad1e2d8bc41b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/870d3ad89507984580baacd9b27ad1e2d8bc41b4"}]},"branch":"refs/heads/master"},"ed2738933b0967877386d0406502deaa2eb05ee0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-24 11:58:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/964715/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/15/964715/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/15/964715/6"}}},"commit":{"parents":[{"commit":"609bf18285b661fd9b5109ddfbcd4e878ea5a04d","subject":"Rename setup.cfg parsing utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/609bf18285b661fd9b5109ddfbcd4e878ea5a04d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:18:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 11:58:51.000000000","tz":60},"subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","message":"Merge pbr.core, pbr.utils into pbr.setupcfg\n\nThis module now contains all functionality related to parsing the\nini-style setup.cfg files.\n\nChange-Id: I6591bf05d01cc72cbcf64a43fc7bef5a9d708cb4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ed2738933b0967877386d0406502deaa2eb05ee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ed2738933b0967877386d0406502deaa2eb05ee0"}]},"branch":"refs/heads/master"},"e479527c0eb5429ebcf60387fffe7c088338f590":{"kind":"REWORK","_number":7,"created":"2025-10-24 14:48:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/964715/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/15/964715/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/15/964715/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/15/964715/7"}}},"commit":{"parents":[{"commit":"a3cf42d4941938e57c9581c0338cd79a5d795e96","subject":"Rename setup.cfg parsing utilities","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a3cf42d4941938e57c9581c0338cd79a5d795e96"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 16:18:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 14:48:40.000000000","tz":60},"subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","message":"Merge pbr.core, pbr.utils into pbr.setupcfg\n\nThis module now contains all functionality related to parsing the\nini-style setup.cfg files.\n\nChange-Id: I6591bf05d01cc72cbcf64a43fc7bef5a9d708cb4\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e479527c0eb5429ebcf60387fffe7c088338f590"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e479527c0eb5429ebcf60387fffe7c088338f590"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
