)]}'
{"id":"openstack%2Fpbr~964716","triplet_id":"openstack%2Fpbr~master~Icae5d61d59135a7269816014e82d18aaf96d6ea7","project":"openstack/pbr","branch":"master","topic":"pep517","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-06-25 12:13:02.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-25 12:13:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icae5d61d59135a7269816014e82d18aaf96d6ea7","subject":"WIP: Add initial pyproject.toml hook","status":"ABANDONED","created":"2025-10-23 17:18:19.000000000","updated":"2026-06-25 12:13:02.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"27da6cebf9939867ddcb79cd6dc7f56483c52980","_number":964716,"virtual_id_number":964716,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-28 18:11:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-23 17:41:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-23 18:05:00.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"}],"messages":[{"id":"1104c3b9f7e2342f1ef0e033c4f342e7f93fd603","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-23 17:18:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e4ef34580fbc78813bb445a58441e4ef123cd69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 17:41:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b45723fb85c94cd18a1fd51250b5bd33\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/52ae484f1f254392894742ed594124ba : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83526942351647f99ed0513437445376 : FAILURE in 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e438ea5721545b3a813c213b912637c : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e30314caaa141a48b9260afbe6c9195 : FAILURE in 2m 49s\n- build-python-release https://zuul.opendev.org/t/openstack/build/e2915dad1a1542ba95f4c1f2a059bf3e : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8debe832d53a4693a8341567b5196475 : FAILURE in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7bf7ab60eb54fdabf6470c0b51cf632 : FAILURE in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d48dad0212a74846ada6b98b3e58cb14 : FAILURE in 2m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8d5aa4eb1045482eab0d0ae2fcd0afcd : FAILURE in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd0899f982454f19ab25940c309547fd : FAILURE in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43f8efc4b9884659a0229bae180630d1 : FAILURE in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86c093d94c6b4732be45f0790f167127 : FAILURE in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7eb1b38e0fba40d0af14d7a08e38e9c5 : FAILURE in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e52b9e8f8d24435832d93d67c6f71f9 : FAILURE in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4c5c159975a4d04952549441121a088 : FAILURE in 7m 28s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/f853a5d29df7479eb171a53eef0eb406 : FAILURE in 11m 09s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/2bbf5ae514ae4f2a9bd454b88bc662c9 : FAILURE in 12m 41s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/72aadfefe9ab4675a25a0aeebf015ea8 : FAILURE in 14m 00s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/3fb2fbd95e5d4ce4a2fbbcfd5ce169f1 : FAILURE in 9m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a9078000027372770070339386c756057f985824","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-23 18:05:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f9407e7d71da3c3ac3e0fc4b4411a7bae45009a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:00:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5bdbd2995b254a9c30143a934486b7b7e8741274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:25:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"504aceba52a9e6d4ed914b29e48520ac1f21f5b0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:33:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4d8713c0a0800754d4ed36f12ce461bcac422ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 10:33:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0bb7da9df4745666d47cec246543b6df079c66ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 10:56:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e539536ca57445ca4359dbbeccd6617\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d6bcfbf7be214bf683e3122749c68354 : FAILURE in 1m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dc8dfb1e4934ff18d925df3e8454c97 : FAILURE in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f7a676895d4e9f8ac4c178100d94d5 : FAILURE in 1m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdd946131fc24067ad4934558ec88927 : FAILURE in 3m 14s\n- build-python-release https://zuul.opendev.org/t/openstack/build/d2a174d0065b49db9ceedaa8cb4607ac : FAILURE in 1m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2250e3f675ad4ebaab37af7d4d0fbc6e : FAILURE in 2m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0036155af8064ad1818682000a2202dd : FAILURE in 2m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ed3b65291b84d25b07bbb3683a5d7f2 : FAILURE in 1m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f64a7eb403584a2492ea6734c7546f1d : FAILURE in 1m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bf4043853874918b2f8b7e8c6ddcba1 : FAILURE in 1m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bb16cfbea9248b7b5e88d0170825496 : FAILURE in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea469ade0c724461a6e74fe631f7eb5f : FAILURE in 2m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa378f76ce2d47caa0788e9cb2af70db : FAILURE in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72d9dcba59c1414fb7f1cd3324fe6471 : FAILURE in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f2366681cee843e59b0f4013a5c0755a : FAILURE in 2m 52s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/64eeb4f550924bdfbb8f453cd9a03e32 : FAILURE in 9m 21s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/12974767d9ed45b4b17809a332a463cd : FAILURE in 11m 31s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9154cf8e2f5746fa8b2bbecfe417eb66 : FAILURE in 12m 13s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c5cd1579f1744ff1af44362630344eef : FAILURE in 18m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c06826fe81e96edb86d893d127e919085acb411b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 11:10:48.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"38b1e0fbfd0012a537e5214ac3b4bdedac94141a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 11:32:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b35b7c1253434990a27e295ba130b32b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d4c264b1f89840aa93359afc0bc00bc9 : SUCCESS in 2m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d43f180fb1e4737b45e3636feb749e4 : FAILURE in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390d9713bba9442ab61954c27f1ebf62 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bef59836bfe4ed5a91197c1072a0611 : SUCCESS in 2m 34s\n- build-python-release https://zuul.opendev.org/t/openstack/build/fdb619c492764505be0417c5929b5974 : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3771c6b180af4701ba0dcb4dcfcbfc7c : FAILURE in 7m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/356590027a2f418691d9ec3f4de79664 : FAILURE in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efe21bb3501749be92bb5c10adee74f4 : FAILURE in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/865d34e9b0e747dda5be012ca15d3c80 : FAILURE in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6d2f4e433c24d619605f4ec35b0fb3b : FAILURE in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aff7376e297f4fd18fad07f98f1f7ebe : FAILURE in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e56b14ac96444397b4062baf06ae6d8b : FAILURE in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33d86d3381bf4833883b17c4816a6e37 : FAILURE in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c8616a239c7435081076b4f6b489fc2 : FAILURE in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b851646a96a4e98821e311066cde19f : FAILURE in 7m 38s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/784e345613674457bba6a65382b412c7 : FAILURE in 11m 55s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/c24a502a5ee247d9bc1021b881d66004 : FAILURE in 12m 32s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/f95e722d2ecd478da27510484d155694 : FAILURE in 14m 31s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/5e3c842c7b2b436ebdc13d0de7b5e5d5 : FAILURE in 14m 19s","accounts_in_message":[],"_revision_number":4},{"id":"0754b9f357a9402d0a57e732d0cb3017a3436c95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 11:42:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2437b7edccc266473879b7304f1e55cbcc7b7fb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 11:58:57.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"bb0a998c0fd18e777baf399f8d960a082143c37e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 12:20:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6cc5ac5866d40b0b8e86bd1ee8cd2c9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f2f57a8aa6834b4bb35d3389ce8bfc0a : FAILURE in 2m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb3ab0c0ca51495ca9919571cbb5a2eb : FAILURE in 10m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb665ecddef40dfad72bd1bfcfd0e5f : FAILURE in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09fcf1f1fd2946d4ab6f70e00c46a887 : SUCCESS in 2m 52s\n- build-python-release https://zuul.opendev.org/t/openstack/build/07fc060bf99140a1b43c65f336245ab9 : FAILURE in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13b4eee0d25c4cc9805a264140315853 : FAILURE in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1fb04310765046afa9e55a77de5bd7a2 : FAILURE in 1m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/232048446d2b49508220b38685a0ec8d : FAILURE in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/028c7f1b9fd24e51b67bc3ebb134c291 : FAILURE in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2343197a262745b1a5e8faf6218c0e70 : FAILURE in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88d47e267c8d4a9d9df533341459f325 : FAILURE in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7243b5c3a1d34f15aefe28230bf6fe76 : FAILURE in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7eacba0586a8462b8980371c2d53a130 : FAILURE in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/339629fab3124e01ac113904b1f15b4a : FAILURE in 7m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7e4ccfb969f4843adb08d49bfa0cc9c : FAILURE in 6m 05s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/01a87e8f5a9444a0b8d0af8737a890ef : FAILURE in 10m 57s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/601a78ba55834da7b39ba4cad3c14138 : FAILURE in 10m 27s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0b6c63b71ac64117b11754a6845e84b6 : FAILURE in 15m 09s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/765d68fec88b48908275802f9506f101 : FAILURE in 11m 09s","accounts_in_message":[],"_revision_number":6},{"id":"23ad65ee00d851679f91b0bd14c7e37af8509bf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 14:48:48.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"131ab392891d0e65f33d4d80ec5ea5dfab46044e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 15:32:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22e5a798fad4440e8da6f9aa4ec09c8c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63fbdd2c47fb49ba802485a57477fff9 : FAILURE in 2m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d47feb626f14b71a1c17f2675065c8f : FAILURE in 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb67dd81ccc342998590d22666f24830 : FAILURE in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417b5a60a032479993112fd8501ee3c5 : SUCCESS in 2m 55s\n- build-python-release https://zuul.opendev.org/t/openstack/build/5ea6bd0490fc4e3c94ad3e3c446d4273 : FAILURE in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aa06d6382af42f290b332703964e509 : FAILURE in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d4c37fd878e43229214e951a9bed373 : FAILURE in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77b29e0353fa4fa091ca132e360e9f75 : FAILURE in 2m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5234849965247d6b45cb71e45fbefcf : FAILURE in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/193076e66ffd46d3842600877bf5f419 : FAILURE in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b207fcdee974470a2b2dcaca0cf7c0a : FAILURE in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/189c7b9b24534b72aae5d37593eda816 : FAILURE in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54f6b11933ab4b56982aca169a795b86 : FAILURE in 8m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97b178492df94fd8804a78b479569aa3 : FAILURE in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ddd19e98bff9488ea4718366a10a17da : FAILURE in 7m 59s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/1d0ed47d75924c11bfdd9c77f5a71960 : FAILURE in 13m 43s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/96d5169c3f0c42d99bd95f3492d59ff2 : FAILURE in 14m 23s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/7fd1bfc7c4c34925bbc8118cc6abedb3 : FAILURE in 7m 15s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/0546ab42a0e14d4895e90bc35c789812 : FAILURE in 17m 32s","accounts_in_message":[],"_revision_number":7},{"id":"14f46ec310e223d5ee2e61eb6136aad430797cdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 16:53:31.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"c60dc321630590dc313a7e531a2caa6f103d06c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 17:14:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b25c61f898f2497d9a2fc56019216020\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/882faa9becba4ae38e3382841068a19b : FAILURE in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2627a016628e4ef7beca93fe1c3c4143 : FAILURE in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7e240a012d641a09c70a21722febc01 : FAILURE in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c493b0162dc44c9aafbb190895b68cdc : SUCCESS in 3m 03s\n- build-python-release https://zuul.opendev.org/t/openstack/build/801e075eda8f4a1589474974159f9dc6 : FAILURE in 3m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f46f76f2209342d890cf6462915a6f28 : FAILURE in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a6d2839758848b4bf883688fce60d63 : FAILURE in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fed6753f90714d12bd02bfb4191f3563 : FAILURE in 1m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47af52676f6d43a48d72405d0ea5a4ca : FAILURE in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a1cac207c52467389e30e0e91522b2e : FAILURE in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/260b46ad278a4b0c9941adbd424d2a8f : FAILURE in 8m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b59b72093ed747a3b5c7e6822e010963 : FAILURE in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9ecaed4569e495f94f230be5a1fc154 : FAILURE in 6m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/882c8850f0ea473e8dcc37acf09a10ca : FAILURE in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c982e8f89bcf4b9ebd8a79737d94a897 : FAILURE in 4m 02s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/7233caedbfc043e6834324a44430c772 : FAILURE in 10m 59s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/935f866aad994ac499fb40816f5c1a64 : FAILURE in 8m 39s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/5937cc7f9f24435a99e6130e3e5c3158 : FAILURE in 7m 09s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b4952c5016c7486fac10a4a415113f90 : FAILURE in 14m 25s","accounts_in_message":[],"_revision_number":8},{"id":"aac13cce77532ad6b2d178c8253a3233efdee845","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-24 17:38:12.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"54381bedfc77728d3cf119c0643c16806ee7f15b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 15:49:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"38812171f1004c0eeec964e1feb4aca0100ea003","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 15:55:01.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"a614895ada547206952fc1ad5abc7661c32c07d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 18:11:04.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e4d4c2115e443f9bb9270df2110319b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/880191c2d01f4ade8160bae66f8b8183 : FAILURE in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f95c59894124fc6b0f71c4622835daf : SUCCESS in 2h 02m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd2baa555cb4fe3a10c0b952f6d020d : FAILURE in 1m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b801ce3dd74d48108b7673fc071a491c : SUCCESS in 4m 17s\n- build-python-release https://zuul.opendev.org/t/openstack/build/437064eaa5ff4b9fabc93e9cf6d60bbb : FAILURE in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/204673450ffc4a008a5ba082d16b5dea : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ba96bb89ce145ecabafd76a376117ad : FAILURE in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d606e380735142dd9eeb5a098061c2e4 : FAILURE in 1m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea1d921cbf9040a6adade1113fdf3a5a : FAILURE in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb1f9beb58424652aa52090fab550bc2 : FAILURE in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/014f0ff98ec44435ac18eca984703547 : SUCCESS in 8m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/099336c809884767bbf2a8b9cd562750 : FAILURE in 8m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eca5ab12abde47d0abf711309cfb360b : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff364e708eaf4055899b6d17716250a1 : SUCCESS in 8m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/984ad281abfc4586878cbdd4e1a64e70 : SUCCESS in 4m 19s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/18c44df0a4284c7fb715b1d34f21d60b : FAILURE in 6m 39s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/0a8860f5c8fb48fc8b75ee9ed310c069 : FAILURE in 8m 10s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/bb4e4eeb0c6847eb823e4e669b2d7ee7 : FAILURE in 32m 57s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/d25b0dd8302a456a889f87fab1a0abb4 : FAILURE in 28m 31s","accounts_in_message":[],"_revision_number":10},{"id":"27da6cebf9939867ddcb79cd6dc7f56483c52980","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 12:13:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"466d2f739db39544b223ed6fe3b0cd09039e299e","revisions":{"c09ef7d4913abdbf2090feba060958f22304664b":{"kind":"REWORK","_number":1,"created":"2025-10-23 17:18:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/1"}}},"commit":{"parents":[{"commit":"326e094c79c79b70ca565442c7662bcfd8c757d3","subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/326e094c79c79b70ca565442c7662bcfd8c757d3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-23 17:18:01.000000000","tz":60},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c09ef7d4913abdbf2090feba060958f22304664b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c09ef7d4913abdbf2090feba060958f22304664b"}]},"branch":"refs/heads/master"},"8033776d8347e517d90a17d480c20f4049464482":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-24 10:25:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/2"}}},"commit":{"parents":[{"commit":"86813a9b60a1a4f706be1ea2040485ac277e28a4","subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/86813a9b60a1a4f706be1ea2040485ac277e28a4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 10:25:15.000000000","tz":60},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8033776d8347e517d90a17d480c20f4049464482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8033776d8347e517d90a17d480c20f4049464482"}]},"branch":"refs/heads/master"},"640ad2abe77c0ba8edbb966627b00a4b5dc97fbb":{"kind":"REWORK","_number":3,"created":"2025-10-24 10:33:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/3"}}},"commit":{"parents":[{"commit":"dac861b54c900e0567fdd64afe816a2635dd45eb","subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dac861b54c900e0567fdd64afe816a2635dd45eb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 10:32:00.000000000","tz":60},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/640ad2abe77c0ba8edbb966627b00a4b5dc97fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/640ad2abe77c0ba8edbb966627b00a4b5dc97fbb"}]},"branch":"refs/heads/master"},"5ddad59e1444da351cc107e2ebfefb08b3c57488":{"kind":"REWORK","_number":4,"created":"2025-10-24 11:10:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/4"}}},"commit":{"parents":[{"commit":"7349aad5e97febd73c8112f1732bf7b819b1c419","subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7349aad5e97febd73c8112f1732bf7b819b1c419"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 11:10:35.000000000","tz":60},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5ddad59e1444da351cc107e2ebfefb08b3c57488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5ddad59e1444da351cc107e2ebfefb08b3c57488"}]},"branch":"refs/heads/master"},"332119012ee992bb19efdfdb5ec6eb402a38bed8":{"kind":"REWORK","_number":5,"created":"2025-10-24 11:42:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/5"}}},"commit":{"parents":[{"commit":"870d3ad89507984580baacd9b27ad1e2d8bc41b4","subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/870d3ad89507984580baacd9b27ad1e2d8bc41b4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 11:42:03.000000000","tz":60},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nCurrently we only support setting the name and dependencies dynamically.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/332119012ee992bb19efdfdb5ec6eb402a38bed8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/332119012ee992bb19efdfdb5ec6eb402a38bed8"}]},"branch":"refs/heads/master"},"7e94426d50247bba751f1652bc19642aea9d4829":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-24 11:58:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/6"}}},"commit":{"parents":[{"commit":"ed2738933b0967877386d0406502deaa2eb05ee0","subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ed2738933b0967877386d0406502deaa2eb05ee0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 11:58:51.000000000","tz":60},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nCurrently we only support setting the name and dependencies dynamically.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7e94426d50247bba751f1652bc19642aea9d4829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7e94426d50247bba751f1652bc19642aea9d4829"}]},"branch":"refs/heads/master"},"7c134317bb49fbce89eb1a7b897f0985cd53a484":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-24 14:48:48.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/7"}}},"commit":{"parents":[{"commit":"e479527c0eb5429ebcf60387fffe7c088338f590","subject":"Merge pbr.core, pbr.utils into pbr.setupcfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e479527c0eb5429ebcf60387fffe7c088338f590"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 14:48:40.000000000","tz":60},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nCurrently we only support setting the name and dependencies dynamically.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7c134317bb49fbce89eb1a7b897f0985cd53a484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7c134317bb49fbce89eb1a7b897f0985cd53a484"}]},"branch":"refs/heads/master"},"768ad5afc83a0b0c713cfd52c6d937b17d736bc3":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-24 16:53:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/8"}}},"commit":{"parents":[{"commit":"4180700879e120c93d117a2d4b47503b655a7191","subject":"Remove use of pkg_resources from tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4180700879e120c93d117a2d4b47503b655a7191"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 16:53:25.000000000","tz":60},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nCurrently we only support setting the name and dependencies dynamically.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/768ad5afc83a0b0c713cfd52c6d937b17d736bc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/768ad5afc83a0b0c713cfd52c6d937b17d736bc3"}]},"branch":"refs/heads/master"},"4f8f2fba00522cbcedfcd5d486fc5fd6eaf592ac":{"kind":"REWORK","_number":9,"created":"2025-10-28 15:49:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/9"}}},"commit":{"parents":[{"commit":"4180700879e120c93d117a2d4b47503b655a7191","subject":"Remove use of pkg_resources from tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4180700879e120c93d117a2d4b47503b655a7191"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 20:59:25.000000000","tz":60},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nCurrently we only support setting the name and dependencies dynamically.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4f8f2fba00522cbcedfcd5d486fc5fd6eaf592ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4f8f2fba00522cbcedfcd5d486fc5fd6eaf592ac"}]},"branch":"refs/heads/master"},"466d2f739db39544b223ed6fe3b0cd09039e299e":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-10-28 15:55:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/964716/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/964716/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/964716/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/964716/10"}}},"commit":{"parents":[{"commit":"adbe935a4f3736f10f29a4021e4d65009b3bb8f2","subject":"Remove use of pkg_resources from tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/adbe935a4f3736f10f29a4021e4d65009b3bb8f2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-19 13:16:00.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 15:54:52.000000000","tz":0},"subject":"WIP: Add initial pyproject.toml hook","message":"WIP: Add initial pyproject.toml hook\n\nThis will allow us to remove the stub \u0027setup.cfg\u0027 and \u0027setup.py\u0027 files\nthat would otherwise be required if using \u0027pyproject.toml\u0027\nconfiguration.\n\nCurrently we only support setting the name and dependencies dynamically.\n\nWIP while we implement more hooks, add tests, docs, etc.\n\nChange-Id: Icae5d61d59135a7269816014e82d18aaf96d6ea7\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/466d2f739db39544b223ed6fe3b0cd09039e299e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/466d2f739db39544b223ed6fe3b0cd09039e299e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
