)]}'
{"id":"openstack%2Fpbr~965031","triplet_id":"openstack%2Fpbr~master~I91872688982110297a589d3cb2bec6edea37dea0","project":"openstack/pbr","branch":"master","topic":"fix-tests-for-latest-pip","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-10-28 17:21:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91872688982110297a589d3cb2bec6edea37dea0","subject":"Fix functional test for pip 25.3","status":"MERGED","created":"2025-10-27 22:22:39.000000000","updated":"2025-10-28 17:22:19.000000000","submitted":"2025-10-28 17:21:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965031-fix-tests-for-latest-pip","meta_rev_id":"a89ecf27647d2d3a0ed1b85fd1d5f50b990e4178","_number":965031,"virtual_id_number":965031,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-28 17:21:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-10-28 10:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-10-28 15:22:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 00:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-28 10:21:30.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f0b0f139c834bb2064c3ac7c0749c4e1d6f5cfcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-27 22:22:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2eacbcd7ebc7dd78da885d6dcac510db65e3fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 00:28:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4684478e6704a26b4c2ada63dfe112d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a21c13220db4e18ac41771733e97762 : SUCCESS in 1h 59m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c4572001daf46e798315266e64a42b2 : SUCCESS in 2m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b083b1b372642bebe45ed0cd2eb9bef : SUCCESS in 1m 39s\n- build-python-release https://zuul.opendev.org/t/openstack/build/5f9b84e12c1a4c038486f1870e641e69 : SUCCESS in 1m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc46682ca32e44d9a2c457a1e162f9b7 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b6c6a6edc764f42b3c524e7c20ec373 : SUCCESS in 2m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e81231091de14166a118f8cf302fedfa : SUCCESS in 7m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4850bb5116ec411992e6c3ade257d3c8 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2368b3ec0fca4fce8a55b9955cdf752c : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a2047f74e8948249afe8a2db22da425 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e59beb5cb624ad6834f9e8c7e374180 : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f4f503115aa4868978c35aae002e455 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ba1c09b96844299a153e11f7e42e84f : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b23ab86fe9164667afebacb6fc2bbb51 : POST_FAILURE in 7m 59s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/98bc838e930d4527aa943b986c9cd36a : SUCCESS in 16m 26s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b14a89d1f1ae498fad803dfb49ce0666 : SUCCESS in 16m 32s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6f1262b8c6974954b298dede6f12f7e0 : SUCCESS in 34m 59s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/674bae09cbbf4550a81feac9ae37d53f : SUCCESS in 32m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5fc5dd8970fb5b7cd793f54a63d4838619fa92a9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 10:21:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08291fcd7429d3ca72dbff2ab360b33b52643794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 12:22:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67c85d986c8456483c73582fe1e9d39\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc955206b3a04297b28068baa38550da : SUCCESS in 1h 54m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92244ea010424141960e381b965c6990 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cbb1b6da6f34c63a55e41eb180d2c3a : SUCCESS in 3m 59s\n- build-python-release https://zuul.opendev.org/t/openstack/build/5a23998206c64f15a982c02e8fc0fd47 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c78c715f8bf43be8ac73cbecfdbff80 : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d5ab6f2a9ac4773a6d16cff3c48d00e : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a678e1e3959482880f1976dac4f6c15 : SUCCESS in 2m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7685c2ab45124d168f91970382fdf1c8 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fa2eff4ae604d4d90b78cf8827d53c5 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8a3fa4dd9dc44aeb24ee42ca3206e0d : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c82176f2d1b94cdfac5c139ff432a1a8 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf1a77905f634a069d4961f760b98a0e : SUCCESS in 9m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e319e4090b44c2d8bad2d5663ad834c : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dad536060474de5aafdab62192f3f4f : SUCCESS in 4m 05s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/731fa0eb1f2b4c07b2d4a31859b2b939 : SUCCESS in 15m 02s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/1c6305fb4d58422f85236f96c2205280 : SUCCESS in 24m 32s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/44970e1005b743979cd1b9397e10f964 : SUCCESS in 37m 25s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/22919dd40a854cffb466d7664ae6bd1a : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":1},{"id":"44afaf71ef5b26ef7796f78973540d3144289dcd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 15:22:14.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53a5b9af86c2e9103a15a4429f3fa3a22b4b268c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 15:23:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a24c7bffddd320f1f7d44dc328e114c97e471c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 17:21:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5af38d7158142c6afde27219bb879ed\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/768539b5892a407c906a8785ca45ad88 : SUCCESS in 1h 53m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5726955682742ccabd8d75d98466802 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10bdc262b6154e978c798b14ca3ca21c : SUCCESS in 1m 49s\n- build-python-release https://zuul.opendev.org/t/openstack/build/52a3bab7550e46798669294d3e3754a6 : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d8c7cc60d3a4f868460222cf7938e9c : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7699f1f5945c4295963aedbd0b60fc1e : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1a4c3768fe44972abf1b819d5ae11db : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8c60df431304fc987c2cc44cae14bdb : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77440c9b1f0543c6b3cb5515f76bbaf1 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98020c3dc91b432abd905560bd7f5389 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27c63014adcd40e29ab80e0cd4603bb2 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/443c595998454edea8e475a09f8ff55f : SUCCESS in 6m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02c5e4f1e2e04fd1b3eb0f284cc5b29c : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d41b806a1164073afab026bea4e95e3 : SUCCESS in 12m 00s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/024e18fbcfcd42ed8df16b522ed539a6 : SUCCESS in 29m 05s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/5f321bd9c6a7432e9f2cacd4ee7461a6 : SUCCESS in 29m 38s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0deeed1bb0964412b79db0b1b8808ec4 : SUCCESS in 39m 43s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/32ffd27c85704353a733855a7312f307 : SUCCESS in 35m 36s","accounts_in_message":[],"_revision_number":1},{"id":"660844a6e61af458ab611e7f113427598021ff9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 17:21:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a89ecf27647d2d3a0ed1b85fd1d5f50b990e4178","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 17:22:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72f626fe57944f6c884b1c6095d8877e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f8e9685f194049bd1bd5a4f5ec2563 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"452944d220afd0b96f59bd93ccd03c50c31d3cb2","revisions":{"452944d220afd0b96f59bd93ccd03c50c31d3cb2":{"kind":"REWORK","_number":1,"created":"2025-10-27 22:22:39.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/31/965031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/31/965031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/965031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/965031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/31/965031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/31/965031/1"}}},"commit":{"parents":[{"commit":"7181a4aa517c783256c6cdef19d31da02df4a18b","subject":"Pin zuul ansible, pyperclip versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7181a4aa517c783256c6cdef19d31da02df4a18b"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-10-27 22:20:00.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-10-27 22:22:28.000000000","tz":-420},"subject":"Fix functional test for pip 25.3","message":"Fix functional test for pip 25.3\n\nIt appears that pip applies the --no-index configuration to isolated\nbuild environments leading to errors like this when we try to build our\ntest packages:\n\n  Processing /tmp/tmpdhkm1rkk/tmpaik8vv54/repo/test_markers-0.0.tar.gz\n    Installing build dependencies: started\n    Installing build dependencies: finished with status \u0027error\u0027\n  STDERR:\n  error: subprocess-exited-with-error\n\n    Ã� installing build dependencies for test_markers did not run successfully.\n    â�� exit code: 1\n    â�°â��\u003e [3 lines of output]\n        Looking in links: /tmp/tmpdhkm1rkk/tmpaik8vv54/repo\n        ERROR: Could not find a version that satisfies the requirement setuptools\u003e\u003d40.8.0 (from versions: none)\n        ERROR: No matching distribution found for setuptools\u003e\u003d40.8.0\n        [end of output]\n\n    note: This error originates from a subprocess, and is likely not a problem with pip.\n  ERROR: Failed to build \u0027test_markers\u0027 when installing build dependencies for test_markers\n\nAddress this by passing --no-build-isolation to the test cases. This may\nintroduce additional problems but we\u0027re already running the pip install\nout of a dedicated venv so they should be fairly isolated to this small\ntest case setup.\n\nChange-Id: I91872688982110297a589d3cb2bec6edea37dea0\nSigned-off-by: Clark Boylan \u003cclark.boylan@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/452944d220afd0b96f59bd93ccd03c50c31d3cb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/452944d220afd0b96f59bd93ccd03c50c31d3cb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
