)]}'
{"id":"openstack%2Fpbr~965146","triplet_id":"openstack%2Fpbr~master~I500de96ad0ee5d9d2e81deaf3158b0b7f1fbef18","project":"openstack/pbr","branch":"master","topic":"pep517","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-25 12:13:00.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I500de96ad0ee5d9d2e81deaf3158b0b7f1fbef18","subject":"WIP: Alternative approach","status":"ABANDONED","created":"2025-10-28 15:49:34.000000000","updated":"2026-06-25 12:13:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6b550787d295f9bc7b7f1a5cc083b847eacdfcbc","_number":965146,"virtual_id_number":965146,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-10-28 17:20:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 17:20:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9197dd1352decaed5d28ad6579e0c437c15145da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 15:49:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a541bcfb2f35b7b6d503362bd5a3035a3d78b701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 15:55:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e5ee5fdffe77c8c3b02feabdf658cf666195a8a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 17:20:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50d142bcf38d4fea944e58fa1fe3d0d2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00512f8915c64dc1918109fa265a8891 : SUCCESS in 1h 22m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73769b86772946feb020035b169f832a : FAILURE in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99d495481f8d404d889f96cf4d8f6d39 : SUCCESS in 3m 05s\n- build-python-release https://zuul.opendev.org/t/openstack/build/81398a827f3a4eb1a201c336e2500412 : FAILURE in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24ca9e972f5d43fe996337c237d05ec6 : FAILURE in 8m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7fdbc83db5894f7785d0746ff79bd776 : FAILURE in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faffec4f43bc4809877a4350139f9437 : FAILURE in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33b14872e64148dfb8f479f29b4d9346 : FAILURE in 17m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a1e89e63c0841a291ef418b0cc2f105 : FAILURE in 8m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87dc638702dd49bbbd1f3b401442a506 : FAILURE in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c74d2a7508c349f7818c531e57c58b7d : FAILURE in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc4b06e95e334a748d29d8fbe80cc7ce : FAILURE in 9m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba8c84803461468692f13ba88a30a59e : FAILURE in 8m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a4ee880c8d74c6d9ccebf2e160ff73e : FAILURE in 11m 12s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/065f235d59d14fb185c226c75116ef7e : FAILURE in 4m 21s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/32c474eceb534b0f9cbd521f8fc480ba : FAILURE in 4m 17s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/02f085e649ab4b6f807d0f39b9ebb677 : FAILURE in 14m 21s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/0cc456955ea6425eb8d5c68eafc66673 : NODE_FAILURE Node(set) request f595df8283a246569c7e1a30a6a403b0 failed in 0s","accounts_in_message":[],"_revision_number":2},{"id":"6b550787d295f9bc7b7f1a5cc083b847eacdfcbc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 12:13:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1bbc170537648c6cc8795c58eac4830ff19e47a4","revisions":{"35694beac57e39ce05bf7253d62ae41ae9e3cc71":{"kind":"REWORK","_number":1,"created":"2025-10-28 15:49:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/965146/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/46/965146/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/46/965146/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/46/965146/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/46/965146/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/46/965146/1"}}},"commit":{"parents":[{"commit":"4f8f2fba00522cbcedfcd5d486fc5fd6eaf592ac","subject":"WIP: Add initial pyproject.toml hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4f8f2fba00522cbcedfcd5d486fc5fd6eaf592ac"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 21:00:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 15:49:25.000000000","tz":0},"subject":"WIP: Alternative approach","message":"WIP: Alternative approach\n\nChange-Id: I500de96ad0ee5d9d2e81deaf3158b0b7f1fbef18\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/35694beac57e39ce05bf7253d62ae41ae9e3cc71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/35694beac57e39ce05bf7253d62ae41ae9e3cc71"}]},"branch":"refs/heads/master"},"1bbc170537648c6cc8795c58eac4830ff19e47a4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-10-28 15:55:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/965146/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/46/965146/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/46/965146/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/46/965146/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/46/965146/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/46/965146/2"}}},"commit":{"parents":[{"commit":"466d2f739db39544b223ed6fe3b0cd09039e299e","subject":"WIP: Add initial pyproject.toml hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/466d2f739db39544b223ed6fe3b0cd09039e299e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-24 21:00:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 15:54:52.000000000","tz":0},"subject":"WIP: Alternative approach","message":"WIP: Alternative approach\n\nChange-Id: I500de96ad0ee5d9d2e81deaf3158b0b7f1fbef18\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1bbc170537648c6cc8795c58eac4830ff19e47a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1bbc170537648c6cc8795c58eac4830ff19e47a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
