)]}'
{"id":"openstack%2Fpbr~965314","triplet_id":"openstack%2Fpbr~master~Id18b97e2d88f263d2fcfed4f69b3388d3bfec110","project":"openstack/pbr","branch":"master","topic":"test-cleanup","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-14 13:27:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Id18b97e2d88f263d2fcfed4f69b3388d3bfec110","subject":"tests: Modernize TestInstallWithoutPbr","status":"NEW","created":"2025-10-29 18:25:37.000000000","updated":"2026-08-14 13:27:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4ed2af04c8fb19fd2d134906ff6131930c5d247b","_number":965314,"virtual_id_number":965314,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-14 11:46:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-29 20:10:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd19579a64c4e5597a5fe4f734e11869b43e2052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-29 18:25:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21e5c28df1f891514e293f3f60cb64d8e6ecc79b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-29 18:51:11.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b71d605288b0c2e5595e72bf1a108f75568aef15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 20:10:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1efe85792ad74342977feb676a7cf393\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/111ca97e319e42fa8f7b02c320c62597 : SUCCESS in 1h 41m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a28253b2a1646ed83487d1af33a202e : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e253c5ff7a8468fa0ead595ca41c228 : SUCCESS in 4m 19s\n- build-python-release https://zuul.opendev.org/t/openstack/build/98fbaa52d51749a6b347218d0f4b73a6 : SUCCESS in 2m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c47dd66369044d0996e1585f91a467e : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0088ff270c3e4966921947a7a7a29371 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df9b69fdb288447cb5fa3e9fbbe36520 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b48e09a552a47209b90bd1a1002bf96 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b88b64773134fc09bd48617533f2918 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4888035f6c3045e5b5464b72dfa9d0c8 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b27e2ad0d6842859b889b7764c865d5 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03208e1ff7384faa8bc951cc14b23027 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/757a822e9fa54b299649a8f3adbb7ab3 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40eecf02f13b4c11911fa7ec6bc34fd6 : SUCCESS in 7m 10s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/630de6a2c61748de9a11f48af2f53b72 : SUCCESS in 24m 45s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/8e6d9e5e254d42e493c04f8653b7f3f6 : SUCCESS in 26m 53s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/1323f7fd37f542d79aef78483d2e94f3 : SUCCESS in 32m 36s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b683ceb728534a4a935ef68867b8dacc : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f7206d7f2b7e5ddfa47c9facb32744c62891a3b2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-13 13:16:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"4f97a2310bb0846c5043654f63ab003a72da4275","tag":"autogenerated:gerrit:restore","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-13 14:10:44.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"114cb02c9dd292b0bf93365255fd18b027df1146","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 14:11:09.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/pbr for 965314,1","accounts_in_message":[],"_revision_number":1},{"id":"7e6dfcafc4b66692648a8c16cb109c4ff0540f2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-14 09:59:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"13d50b1e52b34fde71cb4a038812bbe84755db41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 11:46:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7596a55e60e4e43ac233274e977c280\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6ae65d7977243daaf2e3d70321ac703 : SUCCESS in 1h 39m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a88f103002ef4545b194857cc4e29919 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70d512f667f74104ae9f4861b88e410d : SUCCESS in 2m 23s\n- build-python-release https://zuul.opendev.org/t/openstack/build/9b8d3beeabbf40d080b2e5b0dcad7bf6 : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/097dd18b0ebf4adeb3b103860be4f1b1 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8c6c65318dd4634b9c5a9efb9ef43c1 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d49cdc94ad74b0c913b3dc0084d17ef : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86350ade1cfa40a59c88b90223fab792 : SUCCESS in 8m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9726fd215dd4b18a1d1cf30084769bb : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a9c6ec39d264d91b54a442f1171624a : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e1ddae04b6d40d098fe51075a9f7812 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df2ea8e54bd44881a9196d1d72f8ece1 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5dc4451db0c2495ea985ee3dd8040681 : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5817614df0c64f1c947608aa93759859 : SUCCESS in 5m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a266b2c838641faa90e2ae4b0250e4d : SUCCESS in 6m 57s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/8892a328e03847ca9e083415a0773ba9 : SUCCESS in 8m 09s (non-voting)\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/e18e072b687c4c388234cf2ed899040d : SUCCESS in 23m 07s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/5d7d653910f8478bb25d6fd023c1fa2c : SUCCESS in 37m 00s\n- pbr-installation-openstack-resolute https://zuul.opendev.org/t/openstack/build/2bab8c5578bf4dfd99975ae7e6fabfdf : SUCCESS in 19m 46s\n- pbr-installation-openstack-pip-dev-resolute https://zuul.opendev.org/t/openstack/build/f2b1f0bb54bd45189cd45e6d097db6b5 : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":2},{"id":"4ed2af04c8fb19fd2d134906ff6131930c5d247b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-14 13:27:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d1a013e087f9f384d4a6e7e0b2f901c157bd99f4","revisions":{"05ecbe1fc7679f063563c44922b415fb8b51d163":{"kind":"REWORK","_number":1,"created":"2025-10-29 18:25:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/965314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/965314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/965314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/965314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/965314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/965314/1"}}},"commit":{"parents":[{"commit":"bd86961a3ee0ae39b397ba60b6239c6425e9cb8f","subject":"Misc fixes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/bd86961a3ee0ae39b397ba60b6239c6425e9cb8f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-29 12:40:57.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-29 18:25:23.000000000","tz":0},"subject":"tests: Rework integration tests","message":"tests: Rework integration tests\n\nEnsure we always install pbr from the local repo and combine two mostly\nidentical tests.\n\nChange-Id: Id18b97e2d88f263d2fcfed4f69b3388d3bfec110\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/05ecbe1fc7679f063563c44922b415fb8b51d163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/05ecbe1fc7679f063563c44922b415fb8b51d163"}]},"branch":"refs/heads/master"},"d1a013e087f9f384d4a6e7e0b2f901c157bd99f4":{"kind":"REWORK","_number":2,"created":"2026-08-14 09:59:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/965314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/14/965314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/965314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/965314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/14/965314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/14/965314/2"}}},"commit":{"parents":[{"commit":"1eedb43b0bd90ac7807dd11e21a90c80a05c9207","subject":"tests: Combine two tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1eedb43b0bd90ac7807dd11e21a90c80a05c9207"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-13 14:25:03.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-14 09:58:53.000000000","tz":60},"subject":"tests: Modernize TestInstallWithoutPbr","message":"tests: Modernize TestInstallWithoutPbr\n\nReplace use of setup.py with build and pip.\n\nChange-Id: Id18b97e2d88f263d2fcfed4f69b3388d3bfec110\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d1a013e087f9f384d4a6e7e0b2f901c157bd99f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/d1a013e087f9f384d4a6e7e0b2f901c157bd99f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
