)]}'
{"id":"openstack%2Fpbr~965397","triplet_id":"openstack%2Fpbr~master~I7f2774a8497c847211ce496c9496748f607bd2d9","project":"openstack/pbr","branch":"master","topic":"jammy-py27","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-11-04 17:57:19.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}},"hashtags":[],"change_id":"I7f2774a8497c847211ce496c9496748f607bd2d9","subject":"Run py27 job with default ansible","status":"MERGED","created":"2025-10-30 15:10:58.000000000","updated":"2025-11-04 21:42:24.000000000","submitted":"2025-11-04 21:40:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965397-jammy-py27","meta_rev_id":"9c5f71f98a0f170b1ab8612f0b2ac8a9662b6e20","_number":965397,"virtual_id_number":965397,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-04 21:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2025-10-30 15:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2025-11-04 17:57:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2025-10-31 15:10:59.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-30 15:11:26.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-10-30 15:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca2f93ed6be45d7b40d7dcf74012853b31b640f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-30 15:10:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60bfa80939c09ce7fee1c19304f434e95cd53f73","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-30 15:11:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c771da34bc955f73591b502c65ff14c9c6fcec53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 15:30:04.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/965396 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"99c1d3522ecfc3ccd10ee93af23d91f2a11649d5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-30 15:30:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55cb3d37857b3c1cd2e4daf3b7955f9533248117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 15:44:15.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/965396 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"f63babc9a5527fa07be0cdb68ae9e8fde3ca68bc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-30 15:44:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3a48b74d3c0610d0d8bb6dafc626473647cece1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 15:46:02.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/965396 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"29bd488e2f7cb66a6a1d4790b85992ff912d628c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-30 15:46:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f79b63084ce9ba89d4786d6d46bc95181f2f9769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 16:45:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09f73e08a7ae4ff092239f6aa4d8bace\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b8ce8f14d90414997c521dc07a6be97 : SUCCESS in 56m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb8c00c3afd47d58210a10b03c9b35a : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fbac3c6da734618bf31965a1765f6ca : SUCCESS in 3m 19s\n- build-python-release https://zuul.opendev.org/t/openstack/build/3dec1f23dfc6486cafbf2e4d950493aa : SUCCESS in 1m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54001deef8af4eaaa4708230735e2d9b : SUCCESS in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58b4f80454754ec3b547e384aa5c06ff : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/099e095e60b34308a5c61a11955ee39b : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ab65319e1b24745ae31b27fceb11eb4 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2caa78b2034461d944cdf5e412eed3a : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/730da98ea4ce4baa84c5799dbeca3b55 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2066a53897c460dbb2958c1b6d5ba5e : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6bbf3ea3fc59497398df19c6423f9384 : SUCCESS in 7m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43810390633349708a5cf65b8a335944 : SUCCESS in 8m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/809e6dc039f14602bc658de8ab2898f7 : SUCCESS in 4m 19s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/e2f43c1626624aa28e7548f54467934b : SUCCESS in 28m 38s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/d17373af50fa4177afa0e4049b0d65a0 : SUCCESS in 16m 04s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/daa584529f4544b0b68aad3604382d77 : SUCCESS in 38m 32s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ab7526b6d6f6419da669d0daef347365 : SUCCESS in 43m 48s","accounts_in_message":[],"_revision_number":1},{"id":"70bc38e9476e0210378b1e57e9f96b1b66acf205","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 11:29:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c35796ed15d7db3222ee53dede695e535e216f5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 11:29:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14abd3b7b6b24490bed0fcff5c610652\n\n\nWarning:\n  Change 965396,4 in project openstack/openstack-zuul-jobs does not share a change queue with 965397 in project openstack/pbr","accounts_in_message":[],"_revision_number":1},{"id":"8e7cd7a46f7d42afe9b5ca231a4d15487065fe4a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 15:10:59.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"cb5d0d249089776fc6cca65a0815374d81fb7944","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-11-04 17:57:19.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"496b4e71f78ce24320f092ac7efb9d27d187fa36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 20:14:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f95e79348724bd18b987e0d7a57022b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d45f6fe7f94747208bbe44ac71c35096 : SUCCESS in 2h 02m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f63a45ea55b458ba5a483a39c3d1473 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bad4f2dc14ae4f7ea43c8f91408de9fb : SUCCESS in 3m 53s\n- build-python-release https://zuul.opendev.org/t/openstack/build/c23f383bbcf846b4b9e5ec9707577b39 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/903571af46664847b4fea6a0f4381ce8 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6cda3c7eca424f7a835f92031f883148 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33c3e9e33cc7447cb73ecab19ef95555 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26250a3a64c140218064331caaae8222 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68287759cc7144e780a79cf06b87d130 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab6845ae51e549218f7f075455cf5946 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cce1daee9e24c0d89ee10f71b5bab77 : SUCCESS in 9m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3712d930bca4f839acd85954fb10e54 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9faa7cf2d1694acfaf173b6f0ec7301e : SUCCESS in 5m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95b15ba69eb0446d9b613284a963786d : SUCCESS in 8m 10s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/2b1fab90ff04470a80965930e1f79cd4 : SUCCESS in 29m 57s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/8d3599d6ec044fefb50e14c7357b24fa : SUCCESS in 24m 49s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/1bd34fcac46c40d7a8781f94b32d6c2a : SUCCESS in 38m 06s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/299c6c15ee7545a4a324006f1623ad64 : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"078fda8d23fa6237706caf0d1152e3e110e8d3f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 20:15:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"772923ca21a1cb625c1e30af4bd0c0dc74fe89e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 21:40:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f09d231524d4fd89096e96b6fe56171\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e77d0f5abe074d6881a7ec5acede266b : SUCCESS in 1h 24m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12bac096ce13424ab5030554e760615d : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d0fd0d8ccd447488264ac653678b1b : SUCCESS in 4m 01s\n- build-python-release https://zuul.opendev.org/t/openstack/build/733bb1a6e3f64e259f3d2869249bf6cc : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7c4ce54b9104da8974c71f7395649b0 : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3327e4281b2243e9bc88f625bb8268e4 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a39ad3f16c24e9fb1a0ab5d23cd923d : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a692967d2e724937b8b6be139ed6e76e : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a4a61f91f25490fb7285c2a70ab0474 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/768034e47e774a79b31cbd902b139184 : SUCCESS in 8m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6253f4bae41447959c3f266d591d0829 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63e10ce3382741b38a99d4e09de13b42 : SUCCESS in 4m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2845f55bc0745279b042a4c95552540 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00a14885571a44ccbbe059588425962e : SUCCESS in 8m 19s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/fec0f28361284d7189fbb3a11a1289ac : SUCCESS in 26m 14s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b4c12746e944420a986149ee1f3a662e : SUCCESS in 24m 40s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/37bfedf7ea8c4ab69800643251a7d639 : SUCCESS in 32m 19s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/febb6f44d07f413bb1412d8522a54abf : SUCCESS in 39m 27s","accounts_in_message":[],"_revision_number":1},{"id":"292527df486c27f447306d7af672557cb213867b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 21:40:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9c5f71f98a0f170b1ab8612f0b2ac8a9662b6e20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 21:42:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/963e67ff1c084188aaf9d5573a1b56fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d315c86adfb84a4091ff0cba6380dc6f : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"963722ab86c9fa55a4e29d840f45ffd516c509b1","revisions":{"963722ab86c9fa55a4e29d840f45ffd516c509b1":{"kind":"REWORK","_number":1,"created":"2025-10-30 15:10:58.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/97/965397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/97/965397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/965397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/965397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/97/965397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/97/965397/1"}}},"commit":{"parents":[{"commit":"60521a9d79655527370b7e9e393d41822ac86280","subject":"Merge \"Remove use of pkg_resources from tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/60521a9d79655527370b7e9e393d41822ac86280"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-10-30 15:09:58.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-10-30 15:10:45.000000000","tz":-420},"subject":"Run py27 job with default ansible","message":"Run py27 job with default ansible\n\nThe depends on change updates the py27 job to run on Jammy which can be\nused with default Zuul ansible. Drop the ansible version override as a\nresult.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/965396\nSigned-off-by: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nChange-Id: I7f2774a8497c847211ce496c9496748f607bd2d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/963722ab86c9fa55a4e29d840f45ffd516c509b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/963722ab86c9fa55a4e29d840f45ffd516c509b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
