)]}'
{"id":"openstack%2Fpbr~965470","triplet_id":"openstack%2Fpbr~master~I083d6a4ee5bf5a488f3e4e909e01baf3a6c28847","project":"openstack/pbr","branch":"master","topic":"bug/2121859","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-31 16:47:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I083d6a4ee5bf5a488f3e4e909e01baf3a6c28847","subject":"Fix handling of long_description","status":"MERGED","created":"2025-10-31 10:59:54.000000000","updated":"2025-10-31 16:49:04.000000000","submitted":"2025-10-31 16:47:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965470-bug/2121859","meta_rev_id":"8739e8f6d3d953ed01b41ce7e23cda6e72c63abe","_number":965470,"virtual_id_number":965470,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-31 16:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-10-31 14:47:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-31 13:21:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2025-10-31 14:47:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-31 12:54:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-31 13:21:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-10-31 14:47:25.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"31d791b59cb0d673ebdb81de5afec8d106ac2432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 10:59:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45ba066d2de1609cc7ddd3ab430ef9f5d3c9cc15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-31 11:07:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b679828333e86a15ab68cf46d8f7cc4b139f78be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 12:54:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/344ae57513b94177a9c2cdcabdf1b9d6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79466aaae6224768af7a667a1e77161e : SUCCESS in 1h 42m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec741a91d61426c859dd94fbd0d86c7 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38af9f094c1c4fe8a068b82529a513db : SUCCESS in 2m 33s\n- build-python-release https://zuul.opendev.org/t/openstack/build/06370ed8eeb94979af105b547e423296 : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a2dbf25a5a44805aa8e9be1aa2519e9 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/118cd43bcb1a4f878e22c2cd6ac2ff4d : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5503ff6559c543b7a02375d514ed5102 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7bc6773e86a45e1a23698a040410787 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75c8affb596546d58ac60b757bdbafaa : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03434eb9d61a45b280fecf6d65076166 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d353c8921a5d437888d45e177a1a7021 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d528748dec2345fc965cee446683d4b4 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb1a11c221fd4c60941f34f6a34e1997 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13c9dfb88e9047079388df893c21eaca : SUCCESS in 8m 53s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/0bcd2066c4224f7bb348261b1a08aa36 : SUCCESS in 17m 34s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/f2a89ad0845249c8961f43be7c1632d1 : SUCCESS in 20m 44s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/72fa2229f57641e796c2d7c67050c2ea : SUCCESS in 21m 34s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/dbc9f53a4c0543218bc622d5fc8ab3bf : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":2},{"id":"602687fc2eeaca46d7fd684f082478a75e97b3b1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-10-31 13:21:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9db28ce73be29bbae3115a707e6a15df312bc60d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-31 14:47:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"43d5d516f19519b77eedec2aab0fc7e183d96402","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 14:48:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"69d1ecf1821c06710ebd27fe9d36d7cd8eb57238","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:47:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41c2a1f6b7ba401ca61887466de1b77b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb40f1fc75a2480ca0cb81b5b8f7c2ee : SUCCESS in 1h 54m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2057f9c878f403ca0259be37b11044b : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ffdd9f73f9f4b00b6b12177b2e1d8d3 : SUCCESS in 2m 44s\n- build-python-release https://zuul.opendev.org/t/openstack/build/9d90262694c44dfa823568f97be3fc48 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f73a97a9edb84644a0a38cb93be478ad : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b873935e122a4f4280c004b7d274fa4d : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f72729381024203a7de792a2a7f0df5 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24eba2e91289469a8fdb6bfaf5cbf5b6 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdd66e625d044cc68aefb69f91c740e0 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a161106b110a4ad2bea7a26bbf5a04d4 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c58106a88cc0423e81d8349ffc0b8064 : SUCCESS in 8m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/223f78de16a34280a82bca4dcbd4925b : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1cdad11b82eb4307a437f8ab3c925779 : SUCCESS in 7m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c170cc0588544d8c90cf26a716d3b6c0 : SUCCESS in 3m 56s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/42dc753d2e3f4e1cbe2a484076b566e7 : SUCCESS in 33m 48s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/638ebf307ace4b2ab3edfd4186feb318 : SUCCESS in 28m 03s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6ae5c5016ffe4578b45fd03625889b0a : SUCCESS in 17m 20s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c3b008ff90cc40dfa76da2f46711193d : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b9e893727a6dca5d92feb487f58f9c88e21e1858","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:47:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8739e8f6d3d953ed01b41ce7e23cda6e72c63abe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-31 16:49:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7a9fcf8053c4ff9bf3a47bd6aa39865\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2ba0422306450393ffdb9dc5723d65 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14829502221fc356f7ed76674956781040878e00","revisions":{"c22eaa4f820f9897eb8a816faa7a8b9d26588912":{"kind":"REWORK","_number":1,"created":"2025-10-31 10:59:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/70/965470/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/70/965470/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/965470/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/965470/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/965470/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/70/965470/1"}}},"commit":{"parents":[{"commit":"60521a9d79655527370b7e9e393d41822ac86280","subject":"Merge \"Remove use of pkg_resources from tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/60521a9d79655527370b7e9e393d41822ac86280"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-31 10:55:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-31 10:59:30.000000000","tz":0},"subject":"Fix handling of long_description","message":"Fix handling of long_description\n\nIt turns out passing e.g. \u0027long_description \u003d file: README\u0027 will not\ntrigger setuptools\u0027 automatic handling of this format.\n\nChange-Id: I083d6a4ee5bf5a488f3e4e909e01baf3a6c28847\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nPartial-bug: #2121859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c22eaa4f820f9897eb8a816faa7a8b9d26588912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c22eaa4f820f9897eb8a816faa7a8b9d26588912"}]},"branch":"refs/heads/master"},"14829502221fc356f7ed76674956781040878e00":{"kind":"REWORK","_number":2,"created":"2025-10-31 11:07:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/70/965470/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/70/965470/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/965470/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/965470/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/70/965470/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/70/965470/2"}}},"commit":{"parents":[{"commit":"60521a9d79655527370b7e9e393d41822ac86280","subject":"Merge \"Remove use of pkg_resources from tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/60521a9d79655527370b7e9e393d41822ac86280"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-31 10:55:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-31 10:55:48.000000000","tz":0},"subject":"Fix handling of long_description","message":"Fix handling of long_description\n\nIt turns out passing e.g. \u0027long_description \u003d file: README\u0027 will not\ntrigger setuptools\u0027 automatic handling of this format.\n\nChange-Id: I083d6a4ee5bf5a488f3e4e909e01baf3a6c28847\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nPartial-bug: #2121859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/14829502221fc356f7ed76674956781040878e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/14829502221fc356f7ed76674956781040878e00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
