)]}'
{"id":"openstack%2Fpbr~965916","triplet_id":"openstack%2Fpbr~master~Ie5f7f1c6752056ff246dbe3880b1d012f8fe9e33","project":"openstack/pbr","branch":"master","topic":"trivial","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-03 16:53:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5f7f1c6752056ff246dbe3880b1d012f8fe9e33","subject":"Add classifiers for latest Python","status":"MERGED","created":"2025-11-03 12:16:33.000000000","updated":"2025-11-03 16:54:16.000000000","submitted":"2025-11-03 16:53:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965916-trivial","meta_rev_id":"d7ae0dfe111fafff59596d98b7f68fcdadd9a7b5","_number":965916,"virtual_id_number":965916,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-03 16:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-11-03 15:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2025-11-03 12:31:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2025-11-03 15:51:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-03 12:31:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-11-03 13:58:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-03 15:51:34.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"e6514c68e3bc5f9e35c7d5b67060b3609aec7fca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-03 12:16:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eab253a6501dc4c0138c7a00385ac4b7b175fa6e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-11-03 12:31:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04a9525143d402f06848bbe7db35b9d4352c57b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 13:58:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea6feaabfff64be7a9c9da5ad63ad67b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb50b4b49b3847b696caa00d3fb2e7bf : SUCCESS in 1m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7effd43dc4c541dc9feaa9aaf6c9f5b6 : SUCCESS in 1h 37m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463b130bbb634473a4a054d274f68061 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580e3b3c0e46432dad5051f4a381d90d : SUCCESS in 2m 42s\n- build-python-release https://zuul.opendev.org/t/openstack/build/3802fb543c2b40b78be99c90700bc589 : SUCCESS in 2m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db864993ef58492395789a0f091bb7ca : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/540856aee9264aa1871c3b2bbd7774d0 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f97ae0b6d2df459bb48de851af07fd3b : SUCCESS in 2m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60c8e4b8344045aa9b1725ff304e6f3d : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b539c10b49d411ead02ae9cba42f874 : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a4bf045c2d940ce8fff5d2585f118bb : SUCCESS in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4af40d260504da593320f1bba589ee0 : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/501cf87faf2e42b080bcf2c96e1eddc2 : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a669226be20d4b88a47714c947f2ff17 : SUCCESS in 7m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66bb6db2eb744da9bfaf327e040916c0 : SUCCESS in 7m 38s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/d0183de0aee04772bd7bac1f1414c7ea : SUCCESS in 27m 25s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/4415ff8f2fdb47ea83713c7ab3e26830 : SUCCESS in 14m 26s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/c124e128dbfd4e8e99d7c723853ff46a : SUCCESS in 16m 44s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ae001f92550e4c738be41432f61edfd4 : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"df156ee6d531700fe6261d69583a9af0760656a3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-11-03 15:51:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"102e5c047a8cc4ec90da6cb821437d7970cd519e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 15:52:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92371a57ec07868150290f5a5311abeccebc74f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:53:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb4891a3c3054a46ab6a4903c47b5718\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eea7c6ab0cda454cb0fe9ac86f22a987 : SUCCESS in 3m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33cf1c0f4d434407bd17529afd037121 : SUCCESS in 53m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310527937b9a414993a028071caa15e6 : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74ae056a19b40cda2ab533e693442e2 : SUCCESS in 2m 37s\n- build-python-release https://zuul.opendev.org/t/openstack/build/aa2c930abc4b40b4bfb9d5213b586e7f : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/006f531cf0a24889b52d4ed267cb0dc8 : SUCCESS in 8m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3bfeaeacc1e4b399883ee22cc426c0a : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c60e2613b6f4f22a6a7704e817f833f : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/754f6a5dacf54a24a50eeba7c3d0d11d : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52c97d65647c4db7bb549bfd1c3cce0b : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1da67e75eced41d3894e29c177622056 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84bd52a8b95946888eeeeb2c7b829e9b : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ff08082a2f04d38bb847d5db3759cdc : SUCCESS in 3m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/272594adc3824f3f885625d3c8ec71cf : SUCCESS in 5m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f38a21a07694ded9f38146855ecf7e7 : SUCCESS in 7m 48s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/bee564785beb41769237c2a530189f28 : SUCCESS in 34m 41s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/b4886cabcc8d41818a95c361548a852e : SUCCESS in 25m 15s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/9c5cada4c12d41e189268995f48924ca : SUCCESS in 38m 24s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/b34a91462e8c480bb257ebbb42407fe3 : SUCCESS in 36m 42s","accounts_in_message":[],"_revision_number":1},{"id":"e354ed51f26c56a6855167e6d6935c770bca910a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:53:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d7ae0dfe111fafff59596d98b7f68fcdadd9a7b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-03 16:54:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4b70825636449c2a559c3f1eb5bff09\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/264b1edbbd7d4198aef50241f74a3b55 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1d6f41ae3afe1850242d5abd55c384fd71d7c22","revisions":{"f1d6f41ae3afe1850242d5abd55c384fd71d7c22":{"kind":"REWORK","_number":1,"created":"2025-11-03 12:16:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/16/965916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/16/965916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/965916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/965916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/16/965916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/16/965916/1"}}},"commit":{"parents":[{"commit":"f4bc8350f2cbddd9d631e65327965579779ec454","subject":"Merge \"Don\u0027t wrap develop on newer setuptools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f4bc8350f2cbddd9d631e65327965579779ec454"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-03 12:16:27.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-03 12:16:27.000000000","tz":0},"subject":"Add classifiers for latest Python","message":"Add classifiers for latest Python\n\nChange-Id: Ie5f7f1c6752056ff246dbe3880b1d012f8fe9e33\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f1d6f41ae3afe1850242d5abd55c384fd71d7c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f1d6f41ae3afe1850242d5abd55c384fd71d7c22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
