)]}'
{"id":"openstack%2Fpbr~988753","triplet_id":"openstack%2Fpbr~master~I83a2f73f2afbb79faee06b8c1364cef1adaa5206","project":"openstack/pbr","branch":"master","topic":"drop-python-2","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-06-26 13:44:51.000000000","reason":"Change was marked work in progress"},"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-06-25 23:34:17.000000000","reason":"\u003cGERRIT_ACCOUNT_2\u003e replied on the change","reason_account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-26 13:44:51.000000000","reason":"Change was marked work in progress"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-06-26 13:44:51.000000000","reason":"Change was marked work in progress"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2026-06-26 13:44:51.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I83a2f73f2afbb79faee06b8c1364cef1adaa5206","subject":"Drop Python \u003c 3.8","status":"ABANDONED","created":"2026-05-15 11:32:00.000000000","updated":"2026-08-13 13:15:53.000000000","total_comment_count":21,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"1a2b4799d53006a95128095c9e1dcc558a315ff1","_number":988753,"virtual_id_number":988753,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-25 15:33:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"date":"2026-06-25 23:34:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2026-06-25 13:32:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-06-25 16:44:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-25 22:48:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":-1,"date":"2026-06-25 23:34:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-15 13:13:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-15 15:13:33.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-05-20 07:57:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2026-06-25 13:29:21.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2026-06-25 16:44:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-06-25 22:48:50.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"0c6e324660e963003bb0358e61affd896e4fbade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-15 11:32:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecc71e22d8d28a729b1e2ca162ac84c3a382635b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-15 11:34:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21b4232f9fb05a84745c6ea058fffafcf6e395f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 13:13:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c6b4ffb6a5b4fa59aab02902165ea68\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3cdc64d8943042039b8170d2945e4d59 : SUCCESS in 3m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f13b350f08041d5b4f7b89a683bec45 : SUCCESS in 1h 34m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c8009128834b6ba5baf05ec0b60e15 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/148b397205094e01ae8748708eef3019 : SUCCESS in 5m 12s\n- build-python-release https://zuul.opendev.org/t/openstack/build/2dde03147e39402baf0dd3f3cc54e8df : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f5a6f16c74346a785b23b5cbe189d43 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf62b077adc348168732cbf113886167 : FAILURE in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be663e4eeec04e728f7eb207ef56b942 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/789ea941784c478fa21a84be9c1359e9 : SUCCESS in 5m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc108d4a568a45aabcc78c551572f891 : SUCCESS in 5m 03s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/07082ab393714cc5ae718cecdb5969c6 : FAILURE in 6m 51s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/cbcecd13fbf6420895a4be02e9a2bcaa : FAILURE in 6m 38s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/999ce6d403474daa87c986739139365b : SUCCESS in 33m 28s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/3f4ba244db3e436b9a19f66917537db7 : SUCCESS in 32m 18s","accounts_in_message":[],"_revision_number":2},{"id":"cee296a481d1bef23ec2c49abf1298367205ab62","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-05-15 15:13:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aa4af6fe3fddec66aa5a76f8cce8fe85909753be","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-05-15 15:14:58.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8376230188629d06c97df231f432589e428f7342","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-05-15 15:16:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5ec9735264ca399048b0fbc6e5e18b24cc47e9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-15 15:34:20.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0aa5ca3328a65a13d25a557a2b8ed67064ca1519","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-05-15 15:39:08.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b3bf0ab44dbad7d7899d99148a1824dceb3e3d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 16:31:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75c5f1069f83455bb333dc95c9842500\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b112a6e193a1459faa206480a24ac7e1 : SUCCESS in 1m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db299b476d9a4bbab3f40c445e4014a6 : SUCCESS in 54m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6db021c3c1f492480fe29d43f8e4111 : SUCCESS in 2m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6a64e3cd43451b8a3487368ade31a0 : SUCCESS in 1m 51s\n- build-python-release https://zuul.opendev.org/t/openstack/build/5c88813af84b4d1fa07c9b60402fdb9c : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0119955f537e4e75af9e7f9e6d19784c : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f379205080e4420a2e2bafee693eab5 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/437d99755d8145b8a8543c06781b2b9b : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8f977caf7be48b8b22cddb73f508b6d : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9c2ca0733764c57a1622c92a1166bdd : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fff365914a8042ae94005b99474a02d0 : SUCCESS in 6m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/835bc1cd15ab42d180e68b0673ebac81 : SUCCESS in 6m 34s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/70648da2f92e4865a31039e81370d8a9 : FAILURE in 14m 53s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/0dbd49518e024be4bd42f3d520893f0d : FAILURE in 10m 14s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/f5f3baac52d0435d86ccef38e6449d53 : SUCCESS in 33m 39s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/3cfcdf60b7854b85afbf065d8d6bea71 : SUCCESS in 17m 48s","accounts_in_message":[],"_revision_number":3},{"id":"bd61b26017decdcbc2174fa358ba4789138a2b44","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-18 12:21:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba9b6268f4e3922b88c26864522a8f5acc8bd30e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2026-05-20 07:57:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22a6a5054cd70a0db3e35a0c75a52e8e9519e3a1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-05-27 14:46:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5d3792d8564315a97616f5c4eb4e4ea53f51c079","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 12:11:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b43071a8e359c0cdb3ef0322c64c2f025d06829b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 13:22:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"b1181ec38bb9914dcc8908ae5143a9b7b437dd80","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 13:29:21.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a0aae80b9d6a514e09532737bc24d0c9f61ff99d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 13:32:08.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d4ba42536a02d016cfc3d4414f09f6f81765d106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 15:33:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c606340b33410395055c5f532c3f4e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7abad6902e6e4d4aacbfd255697b4b6b : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32e8c8cfa355453ebba62a4639629657 : SUCCESS in 2h 03m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/376e962b3f034778bf572bb4b965c9e6 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3898b8abd40546da833ea8f916844edf : SUCCESS in 3m 40s\n- build-python-release https://zuul.opendev.org/t/openstack/build/0271f0ab63b94eefb42bd507d9706da5 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd7112c1cdbc488dbce27b3a93b3a28a : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdd33ae8c3b446e094a868ddd2fc50e0 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f3fd603a0ab46cab18739826a1a0d59 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7430308fe5fe44d39c35ef2009204455 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0047c3393814b0e955843e5a1baddd9 : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5708d1c8da7b4690ae6f6db065de0479 : SUCCESS in 7m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70cf4052d6b142569b0c85c51c64fbcc : SUCCESS in 7m 21s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/c9069f470b14483ea638bb29465df5e9 : SUCCESS in 32m 33s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/6bf8d95c39c6461faa83acdbbc78fc08 : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":4},{"id":"5d67fcc4732e3a46bafcc106f6945bbb7b1a22f4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-06-25 16:44:50.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c3168bb787ff58b5d396c2d4d21325e5ba32dacc","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 16:57:01.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"5baf37f4502eca7ce3c4f29c70a84e5ee6284753","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 17:20:47.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3a7885a88a50f8b42b8ead262fbbaccd3743fe5c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-25 22:48:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"aa4a7866bdfbf22df5b8208e7933f7a284ef03cf","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 23:34:17.000000000","message":"Patch Set 4: -Code-Review Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"883806b71257f1c46d3ce14fc3da52638d0916f3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-26 13:44:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"1a2b4799d53006a95128095c9e1dcc558a315ff1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-13 13:15:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a0975303c31f335832f6963c4eca9280b274e3fe","revisions":{"dc7232127788a16f22be828a3c958b32d8836d6f":{"kind":"REWORK","_number":1,"created":"2026-05-15 11:32:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/988753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/988753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/988753/1"}}},"commit":{"parents":[{"commit":"21e60a77ca64abe2886138b60da12c11c4f51f3c","subject":"Merge \"Run py27 job with default ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/21e60a77ca64abe2886138b60da12c11c4f51f3c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-15 11:18:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-15 11:31:46.000000000","tz":60},"subject":"Drop Python \u003c 3.11","message":"Drop Python \u003c 3.11\n\nThe time has come. Python 2 is no longer available in any distro that I\nam aware of, the last version of OpenStack to support Python 2 was Train\n[1] (which is EOL for over 2 years [2]), and even Swift has long since\ndropped Python 2 support [3]. Additionally, none of the build isolation,\nPEP-517 build frontend/backend stuff ever made it back to Python 2, so\nusers still operating these systems (who?!) will continue to use their\ndistro-provided python-pbr packages for building.\n\nWith all that said, this may cause things to explode and we won\u0027t know\nuntil it\u0027s out in the wild. We should be prepared to yank the release\nand revert this change if necessary. However, given the above I believe\nthis is very unlikely at this point.\n\nNote that we opt for Python 3.10, despite the fact that we no longer\nsupport it in Hibiscus, since pbr is used by more projects than\nOpenStack and should maintain a wider support matrix.\n\nAlso note that we introduce some Python 3-only syntax to intentionally\nbreak environments where `python_requires` is not being respected. This\nshould prevent more insidious issues popping up later on.\n\n[1] https://superuser.openinfra.org/articles/openstack-ussuri-is-python3-only-upgrade-impact/\n[2] https://releases.openstack.org/\n[3] https://opendev.org/openstack/swift/commit/7367907c58933944aaed52f0659be6203791a047\n\nChange-Id: I83a2f73f2afbb79faee06b8c1364cef1adaa5206\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dc7232127788a16f22be828a3c958b32d8836d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dc7232127788a16f22be828a3c958b32d8836d6f"}]},"branch":"refs/heads/master"},"61e01b8f5ae714576c6fccaff2a25e0cdba6db06":{"kind":"REWORK","_number":2,"created":"2026-05-15 11:34:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/988753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/988753/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/988753/2"}}},"commit":{"parents":[{"commit":"21e60a77ca64abe2886138b60da12c11c4f51f3c","subject":"Merge \"Run py27 job with default ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/21e60a77ca64abe2886138b60da12c11c4f51f3c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-15 11:18:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-15 11:34:44.000000000","tz":60},"subject":"Drop Python \u003c 3.11","message":"Drop Python \u003c 3.11\n\nThe time has come. Python 2 is no longer available in any distro that I\nam aware of, the last version of OpenStack to support Python 2 was Train\n[1] (which is EOL for over 2 years [2]), and even Swift has long since\ndropped Python 2 support [3]. Additionally, none of the build isolation,\nPEP-517 build frontend/backend stuff ever made it back to Python 2, so\nusers still operating these systems (who?!) will continue to use their\ndistro-provided python-pbr packages for building.\n\nWith all that said, this may cause things to explode and we won\u0027t know\nuntil it\u0027s out in the wild. We should be prepared to yank the release\nand revert this change if necessary. However, given the above I believe\nthis is very unlikely at this point.\n\nNote that we opt for Python 3.10, despite the fact that we no longer\nsupport it in Hibiscus, since pbr is used by more projects than\nOpenStack and should maintain a wider support matrix.\n\nAlso note that we introduce some Python 3-only syntax to intentionally\nbreak environments where `python_requires` is not being respected. This\nshould prevent more insidious issues popping up later on.\n\n[1] https://superuser.openinfra.org/articles/openstack-ussuri-is-python3-only-upgrade-impact/\n[2] https://releases.openstack.org/\n[3] https://opendev.org/openstack/swift/commit/7367907c58933944aaed52f0659be6203791a047\n\nChange-Id: I83a2f73f2afbb79faee06b8c1364cef1adaa5206\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/61e01b8f5ae714576c6fccaff2a25e0cdba6db06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/61e01b8f5ae714576c6fccaff2a25e0cdba6db06"}]},"branch":"refs/heads/master"},"89dacf6832a25fb3a1ee912fe13b269c67701dfe":{"kind":"REWORK","_number":3,"created":"2026-05-15 15:34:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/988753/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/988753/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/988753/3"}}},"commit":{"parents":[{"commit":"21e60a77ca64abe2886138b60da12c11c4f51f3c","subject":"Merge \"Run py27 job with default ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/21e60a77ca64abe2886138b60da12c11c4f51f3c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-15 11:18:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-15 15:32:06.000000000","tz":60},"subject":"Drop Python \u003c 3.8","message":"Drop Python \u003c 3.8\n\nThe time has come. Python 2 is no longer available in any distro that I\nam aware of, the last version of OpenStack to support Python 2 was Train\n[1] (which is EOL for over 2 years [2]), and even Swift has long since\ndropped Python 2 support [3]. Additionally, none of the build isolation,\nPEP-517 build frontend/backend stuff ever made it back to Python 2, so\nusers still operating these systems (who?!) will continue to use their\ndistro-provided python-pbr packages for building.\n\nWith all that said, this may cause things to explode and we won\u0027t know\nuntil it\u0027s out in the wild. We should be prepared to yank the release\nand revert this change if necessary. However, given the above I believe\nthis is very unlikely at this point.\n\nNote that we opt for Python 3.8, despite the fact that we no longer\nsupport it in Hibiscus, since (a) pbr is used by more projects than\nOpenStack and should maintain a wider support matrix and (b) we are\nstill unsure of the implications of dropping older versions for\nenvironments that use the old setuptools `setup_requires` code paths.\n\nAlso note that we introduce some Python 3-only syntax (specifically\nPython 3.6) to intentionally break environments where `python_requires`\nis not being respected. This should prevent more insidious issues\npopping up later on.\n\n[1] https://superuser.openinfra.org/articles/openstack-ussuri-is-python3-only-upgrade-impact/\n[2] https://releases.openstack.org/\n[3] https://opendev.org/openstack/swift/commit/7367907c58933944aaed52f0659be6203791a047\n\nChange-Id: I83a2f73f2afbb79faee06b8c1364cef1adaa5206\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/89dacf6832a25fb3a1ee912fe13b269c67701dfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/89dacf6832a25fb3a1ee912fe13b269c67701dfe"}]},"branch":"refs/heads/master"},"a0975303c31f335832f6963c4eca9280b274e3fe":{"kind":"REWORK","_number":4,"created":"2026-06-25 13:22:37.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/53/988753/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/53/988753/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/53/988753/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/53/988753/4"}}},"commit":{"parents":[{"commit":"12cb33cedda6773eaebd621c1e08e0c24f4009ca","subject":"Drop Jammy testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/12cb33cedda6773eaebd621c1e08e0c24f4009ca"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-15 11:18:26.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 13:20:48.000000000","tz":0},"subject":"Drop Python \u003c 3.8","message":"Drop Python \u003c 3.8\n\nThe time has come. Python 2 is no longer available in any distro that I\nam aware of, the last version of OpenStack to support Python 2 was Train\n[1] (which is EOL for over 2 years [2]), and even Swift has long since\ndropped Python 2 support [3]. Additionally, none of the build isolation,\nPEP-517 build frontend/backend stuff ever made it back to Python 2, so\nusers still operating these systems (who?!) will continue to use their\ndistro-provided python-pbr packages for building.\n\nWith all that said, this may cause things to explode and we won\u0027t know\nuntil it\u0027s out in the wild. We should be prepared to yank the release\nand revert this change if necessary. However, given the above I believe\nthis is very unlikely at this point.\n\nNote that we opt for Python 3.8, despite the fact that we no longer\nsupport it in Hibiscus, since (a) pbr is used by more projects than\nOpenStack and should maintain a wider support matrix and (b) we are\nstill unsure of the implications of dropping older versions for\nenvironments that use the old setuptools `setup_requires` code paths.\n\nAlso note that we introduce some Python 3-only syntax (specifically\nPython 3.6) to intentionally break environments where `python_requires`\nis not being respected. This should prevent more insidious issues\npopping up later on.\n\nAlso, fix a few tests at the same time that now don\u0027t work even on\nmodern python. Without explicitly defining a project section and\nmarking our things as dynamic, plain old tox tests straight up fail.\nAnd the key in pyproject.toml is tool.setuptools, not tools.setuptools.\n\n[1] https://superuser.openinfra.org/articles/openstack-ussuri-is-python3-only-upgrade-impact/\n[2] https://releases.openstack.org/\n[3] https://opendev.org/openstack/swift/commit/7367907c58933944aaed52f0659be6203791a047\n\nChange-Id: I83a2f73f2afbb79faee06b8c1364cef1adaa5206\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a0975303c31f335832f6963c4eca9280b274e3fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a0975303c31f335832f6963c4eca9280b274e3fe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
