)]}'
{"id":"openstack%2Fpbr~989389","triplet_id":"openstack%2Fpbr~master~I77831644ca6775ddba2632f7a4c025a04b1bad9f","project":"openstack/pbr","branch":"master","topic":"pep518-support","attention_set":{},"removed_from_attention_set":{"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-06-26 16:05:01.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-25 23:19:12.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I77831644ca6775ddba2632f7a4c025a04b1bad9f","subject":"Remove more unused code","status":"MERGED","created":"2026-05-20 14:00:49.000000000","updated":"2026-06-26 16:06:13.000000000","submitted":"2026-06-26 16:05:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989389-pep518-support","meta_rev_id":"3fe1d1a0b84a44e825c39eefaaae73351552b594","_number":989389,"virtual_id_number":989389,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-26 16:05:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2026-06-25 22:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2026-06-25 22:32:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-25 23:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2026-06-25 23:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-20 16:10:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 22:58:16.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2026-06-25 22:32:24.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-06-25 23:01:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"f8803a4591598b58fbc03581223e2e0405df3daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-20 14:00:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ef2306d6770185023e84b1b703e11fbdc50c529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 16:10:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ebefb53f6754825a8f79c43fa408ef9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53320a8805734c5baca444ae3c9825b5 : SUCCESS in 2h 04m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a000b483df4a86a6b354fad6e2235d : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a85a9eb2f5dc45d2aadef31e72538b0d : SUCCESS in 4m 43s\n- build-python-release https://zuul.opendev.org/t/openstack/build/5564390ee1ea48bc8e81d9ab7360f345 : SUCCESS in 1m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7984748bf5c542e98f8885a882c6a0d9 : SUCCESS in 7m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/879a5e22e67940b3bbc3996e08843bd1 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b4d594ad07943ad88c5220971e4b984 : SUCCESS in 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe000ea424fa4782ba270962126c53d2 : SUCCESS in 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/963bd521ae9043a29c9fdd5ed5e752f9 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8ea2254cb7d47af8d8918844d4a9fb5 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33507c09321e427382e7d00af797ad86 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc5ea298de004dd7b451156216d336c6 : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/412efaf8b10c40a28ba40ffcf1825f52 : SUCCESS in 6m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1e44574caa1547c5b3a85fe4813145bd : SUCCESS in 7m 56s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/1346e7995f7942b199db9c2854ba4ceb : FAILURE in 8m 50s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/e6ff7a3b57ef4b8ba94f71c30676d425 : FAILURE in 10m 06s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/2229307eb0ec4e9399813272be863add : SUCCESS in 27m 22s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a42566b597434ce6b6484de64fa9371f : SUCCESS in 22m 18s","accounts_in_message":[],"_revision_number":1},{"id":"10e85148d8fa71c77d69078663d01bade55aca75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-10 22:54:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"e88577823d20a0ed581daf10997e1de8a1cfc2c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 00:25:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad296d3117214f5daee15d611efa4b96\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a90464027344e9a88acc03d625c708d : SUCCESS in 1h 27m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a85fe04851a4e40b28a92fd715b08f2 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0814d1e05f4ddbb3f75f0939c32787 : SUCCESS in 3m 16s\n- build-python-release https://zuul.opendev.org/t/openstack/build/379bf69e88084eaba29836d0e779eafb : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0900b4d44a514c5a919d6b3063ed7b5b : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8de5e50534594f6cab46ad90052261fb : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3db3d72be276463e9c05e7975ea6ec67 : SUCCESS in 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c1f59958c8b4a9d91d05bf0895da9a5 : SUCCESS in 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e91c36e3f364289b9bb99ba7f46e914 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e9649a6ca364e84a02bfa9b2c8e0ac7 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daf5064452b848919c557b05283f5779 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/389b0b203e7f461ca81c87467427daf0 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd42639c7e19470bbd5b3ade7643e078 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de849d4c834b48c4b074b7cda029036d : SUCCESS in 4m 29s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/88bacebb89ce443da973736b4a1b0ce3 : SUCCESS in 23m 56s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c3ebc379d00a4ae4b746406aa3e1c97e : SUCCESS in 34m 55s","accounts_in_message":[],"_revision_number":2},{"id":"1435a7d7a541e1ae67bec32de59c199ba53e7490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-24 22:22:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"8bd4be9e9c8f5fc87c4c7814dae282d1b768d2cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-24 22:44:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"73745244f180509e4598687166a3d0b20a585f88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-24 22:55:19.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"2bb93772dbfb78d44cc1555a3a0b11f84d07a548","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-24 22:58:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9b9aa9235755fa7c00d6b23acec0a9cf92c76e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 01:01:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6545c6b43c5f48bc919f164e14f1ed82\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a49a9f1e2d25467387250dfe45df4711 : SUCCESS in 1h 58m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2376770c3460447dae3d180182d1e190 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf75fba2dc7b4a09a26f0044e0cb06ea : SUCCESS in 3m 08s\n- build-python-release https://zuul.opendev.org/t/openstack/build/71b798af296b4b618fc1b110d447ff9f : SUCCESS in 3m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2357fb8157c24dc48df370ad1c87bb91 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/125809b0175c446481da6d0f114dce92 : SUCCESS in 9m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2f75cbe388f4964ae9ecdef16ef5310 : RETRY_LIMIT in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d34fb7f7026049e69783ebf307412b74 : SUCCESS in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/520c284e79d94473975389fd12bbdddc : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9e88b06085a43b1ae1229855ef2da85 : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6abdf0c7512d420cb6ba75bf653fc320 : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a029b70f07ec47229db120beac46fddc : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/810d7944d92641fe9fe0bc4514e774fb : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/246f09f5199c4a06bb2f02ff62983e4b : SUCCESS in 4m 46s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/f3b4f19a6c5c4793b84387f68850c75d : SUCCESS in 29m 19s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/80e718575b5648e5a4d7df51c272d0d7 : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":5},{"id":"c06d099962b836e3ef661f3b05a72324aa36b77e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 13:38:19.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"1d145f65a5607d4cec179b004398ede52d332ee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 13:41:07.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"089ccc5b35f9edd907d07d2101aa9960eb1e4f23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 14:05:41.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"d05fdff91a13bd2d8244c272882584fcf202038a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 14:06:29.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"d9ff8a85880f33535321bd118c97b6621b480952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 16:01:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83bb9abfa4334590aad28115d247d1f4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d2507bb5d5c4393a7f4e49ebc61c4d5 : SUCCESS in 1h 42m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee14c80f1f27459481e1c27a28ddb692 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/134dd59dc4464480911340cdae80bba6 : SUCCESS in 4m 51s\n- build-python-release https://zuul.opendev.org/t/openstack/build/d364f913cd2340abac4d4d72887f79da : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b9f0009412f42c1ace172f68168d0b6 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c22e8c05063041089748dec4c9292853 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4715d091f084d1a9d0bf84a752ee51c : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63ead982181447d48652c1e3846d0605 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86416b11f71c4a00aff984b45a7ce21d : SUCCESS in 6m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4aefb7292d06474397b159a5607ce35b : SUCCESS in 6m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d856777a903f430d84664e18801504c7 : SUCCESS in 5m 09s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/6bdfd9dab99b4fc5836991aee7e48e41 : SUCCESS in 38m 28s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/26b64ee23ece424f99c19a4e1291caea : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":9},{"id":"8be21c0c1770aa129a1b600f39ac4f8ccfdfab6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 20:46:28.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"2f09497459fa451ea45321f1e9c0e535a3c25fa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 21:45:05.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":11},{"id":"762f0ad293dfa3d406546ddcc4f8506de187ef70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 22:05:11.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"f6a9f0f0259d5dc72a6b72c205c6b2e6d277e729","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-06-25 22:32:24.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"04ca729ddf89b1a459086b61c402b0355d34ac65","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-25 23:01:57.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"13895a3a74e5d57240975f2c6decac86f34ea63f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 23:19:12.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"72d82e50af9cbabf18d1e326347b891202def4dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 00:07:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe47d32b3534c978e97f2c4fe95dd34\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd1639f1ce9e4f11bc09572142c5deb7 : SUCCESS in 1h 51m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb478b2231ab405b92116f8b3f33e850 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a3167664a54a52ad28032f3970e368 : SUCCESS in 2m 50s\n- build-python-release https://zuul.opendev.org/t/openstack/build/e08abda99b76413eab1f2249248b7fbb : SUCCESS in 3m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55b38b56ec1b4ebdae51505f53f71659 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/81c5f311776441539af763b345c7adb9 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a66c82b157049238e0fe189f865043a : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38160d9a68c6453b9e122e162509b3b0 : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0ac9c9524404e1b92b95137f6991a0c : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a9e24fc89b4494087155f3dc2f194ab : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd02fb7be83444ff8d7caf9796e4ab22 : SUCCESS in 6m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cccffe1933542b7970a17d60adcbf65 : SUCCESS in 6m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/802f35e968684dda90c353559e0bb835 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4b5de616fe9420d9b6ab7efa1a4d3ed : SUCCESS in 6m 26s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/4bac6385adfa4d69949ca4bd204265d3 : SUCCESS in 40m 00s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/a5459767a9cb4bb8a1e05524458019e8 : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":12},{"id":"e55027ab213c0a258c981c8bd293a734748c57b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 13:48:35.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"7e81bf4b2389e0ed999b968fea47787b2388ed94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 16:05:01.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/920f154eaaa14c12b9d537e26dddf57d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c479ef19fa4b4cf7a30cdf37945c4434 : SUCCESS in 2h 11m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4abb02a36045708de62c22e182b627 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/058f797c2f2941cc96595dac7bcd10da : SUCCESS in 3m 05s\n- build-python-release https://zuul.opendev.org/t/openstack/build/9f66c23e98da4da696a2b0284d1c6b7d : SUCCESS in 2m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6be39bbd4da47429a7cac43ad89c2d2 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f5a90bc125c341c98d14ae0e901ece67 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c58328cdfe440b4b956f66f6fd167d5 : SUCCESS in 8m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12acff6414344d8487260b82cecb085a : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1e81e164ff047e385e882efa89c54dd : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c8bc450f7d24a05917468fcf43b3893 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c635464449442c2847a48f92538d97c : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/542d10243bb24338a826c2f98452af31 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5d8cdbc7cfe413b8c2a123a7f93a341 : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bd9d6f357404c4889d238405d5be6a2 : SUCCESS in 4m 50s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/83d6714b7571474aaf09710f5f76be7b : SUCCESS in 31m 36s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/fb4b61043f8f4fe5a8ff5615cf2a98c5 : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":12},{"id":"1654e4cc88ffedcd9dacd68d3cfcab6a096fd8ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 16:05:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"3fe1d1a0b84a44e825c39eefaaae73351552b594","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 16:06:13.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/728982eeab0947dea3491d08bc4c9073\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b94f0cbe8764b90af5ef6cb462bba78 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"1e0ef19e4dcff6496aaffd5996dbceeb48910a61","revisions":{"df359fb9f7f2e989d713b6ad3d29aecfdf926d32":{"kind":"REWORK","_number":1,"created":"2026-05-20 14:00:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/989389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/1"}}},"commit":{"parents":[{"commit":"5d7896a8f1d45e0428c8becad211b00f8b6f4d5a","subject":"Add classifiers for latest Python","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5d7896a8f1d45e0428c8becad211b00f8b6f4d5a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 14:00:31.000000000","tz":60},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/df359fb9f7f2e989d713b6ad3d29aecfdf926d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/df359fb9f7f2e989d713b6ad3d29aecfdf926d32"}]},"branch":"refs/heads/master"},"996c25b24d0f105a6e2f7f17d66982e09d741345":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-10 22:54:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/989389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/2"}}},"commit":{"parents":[{"commit":"292d1eeef469758f30ba22b10046ae4c9b6285c7","subject":"Add classifiers for latest Python","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/292d1eeef469758f30ba22b10046ae4c9b6285c7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 22:53:39.000000000","tz":60},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/996c25b24d0f105a6e2f7f17d66982e09d741345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/996c25b24d0f105a6e2f7f17d66982e09d741345"}]},"branch":"refs/heads/master"},"6f639b25f890c40a09b22b980165f532f94651dd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-24 22:22:27.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/89/989389/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/3"}}},"commit":{"parents":[{"commit":"e458c5c5f609aee6e1576b6ea7a5b30764db7510","subject":"Add classifiers for latest Python","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e458c5c5f609aee6e1576b6ea7a5b30764db7510"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-24 22:22:23.000000000","tz":0},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6f639b25f890c40a09b22b980165f532f94651dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6f639b25f890c40a09b22b980165f532f94651dd"}]},"branch":"refs/heads/master"},"90bb892d5915f900ae6accaa5962389539a99942":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-24 22:44:56.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/89/989389/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/4"}}},"commit":{"parents":[{"commit":"9bb6ddf71c05b5a846933128e808f70bbaf17adc","subject":"Add classifiers for latest Python","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9bb6ddf71c05b5a846933128e808f70bbaf17adc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-24 22:44:31.000000000","tz":0},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/90bb892d5915f900ae6accaa5962389539a99942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/90bb892d5915f900ae6accaa5962389539a99942"}]},"branch":"refs/heads/master"},"a519dda2fab67140edbe78533a10873b1fbbc86f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-06-24 22:55:19.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/89/989389/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/5"}}},"commit":{"parents":[{"commit":"188c7d81de36d6f43c9e1dda9d655d06e8830c94","subject":"Add classifiers for latest Python","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/188c7d81de36d6f43c9e1dda9d655d06e8830c94"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-24 22:54:59.000000000","tz":0},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a519dda2fab67140edbe78533a10873b1fbbc86f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a519dda2fab67140edbe78533a10873b1fbbc86f"}]},"branch":"refs/heads/master"},"f5e8dde8ff38021f20de4f1adfb5756331389aa1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-06-25 13:38:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/989389/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/6"}}},"commit":{"parents":[{"commit":"fba91586245b533c9312718ff23f7ee5a048079d","subject":"Add testing on Python 3.14, 3.15 (dev)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/fba91586245b533c9312718ff23f7ee5a048079d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-25 13:38:07.000000000","tz":60},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f5e8dde8ff38021f20de4f1adfb5756331389aa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f5e8dde8ff38021f20de4f1adfb5756331389aa1"}]},"branch":"refs/heads/master"},"5cb00539c4afdb038df79f0216c72504a701eaa9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-25 13:41:07.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/89/989389/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/7"}}},"commit":{"parents":[{"commit":"70daa4ff23be29facbb9997678f30efe1a5f5ad1","subject":"Add Resolute, py314 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/70daa4ff23be29facbb9997678f30efe1a5f5ad1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 13:36:15.000000000","tz":0},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5cb00539c4afdb038df79f0216c72504a701eaa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5cb00539c4afdb038df79f0216c72504a701eaa9"}]},"branch":"refs/heads/master"},"3e3442b09e8d20a0a670cb610491951fcd03ad01":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-06-25 14:05:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/989389/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/8"}}},"commit":{"parents":[{"commit":"94d619c9973bcb8ada82f94044785f77c3e9df77","subject":"Add testing on Python 3.14, 3.15 (dev)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/94d619c9973bcb8ada82f94044785f77c3e9df77"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-25 14:05:00.000000000","tz":60},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3e3442b09e8d20a0a670cb610491951fcd03ad01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3e3442b09e8d20a0a670cb610491951fcd03ad01"}]},"branch":"refs/heads/master"},"1396bd29a9b639e13c7f8e32d422b3963f915291":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-25 14:06:29.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/89/989389/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/9"}}},"commit":{"parents":[{"commit":"8e5f2a95f7a7eb5cde1dbe209ac2579a31d400eb","subject":"Avoid warning about setup.py deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8e5f2a95f7a7eb5cde1dbe209ac2579a31d400eb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 14:06:08.000000000","tz":0},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1396bd29a9b639e13c7f8e32d422b3963f915291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1396bd29a9b639e13c7f8e32d422b3963f915291"}]},"branch":"refs/heads/master"},"da2055c760dced21132878ce58ffea0a0e951399":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-06-25 20:46:28.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/89/989389/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/10"}}},"commit":{"parents":[{"commit":"07ec6ba8c98b6b67e0c59ebc08854b7e76b11064","subject":"Avoid warning about setup.py deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/07ec6ba8c98b6b67e0c59ebc08854b7e76b11064"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 19:10:28.000000000","tz":0},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/da2055c760dced21132878ce58ffea0a0e951399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/da2055c760dced21132878ce58ffea0a0e951399"}]},"branch":"refs/heads/master"},"8e679578433aaeaa1447fe5e71d0c39be440317d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-25 21:45:05.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/89/989389/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/11"}}},"commit":{"parents":[{"commit":"8d04363b1f67a3f6765b40a488723ee502cd48a6","subject":"Avoid warning about setup.py deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8d04363b1f67a3f6765b40a488723ee502cd48a6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 21:45:00.000000000","tz":0},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8e679578433aaeaa1447fe5e71d0c39be440317d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8e679578433aaeaa1447fe5e71d0c39be440317d"}]},"branch":"refs/heads/master"},"1e0ef19e4dcff6496aaffd5996dbceeb48910a61":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-06-25 22:05:11.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/89/989389/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/89/989389/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/89/989389/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/89/989389/12"}}},"commit":{"parents":[{"commit":"87f96e26e93a100661a12b845f0ece5ac1a008bd","subject":"Avoid warning about setup.py deprecation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/87f96e26e93a100661a12b845f0ece5ac1a008bd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-19 23:17:04.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 22:05:06.000000000","tz":0},"subject":"Remove more unused code","message":"Remove more unused code\n\nChange I1fcec0f08408f2f81718484c9df46c43cdf018e4 should have\nremoved this.\n\nChange-Id: I77831644ca6775ddba2632f7a4c025a04b1bad9f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e0ef19e4dcff6496aaffd5996dbceeb48910a61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e0ef19e4dcff6496aaffd5996dbceeb48910a61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
