)]}'
{"id":"openstack%2Fpbr~989390","triplet_id":"openstack%2Fpbr~master~Iacbb7237870cc3ae529703a6fb434812d51129cb","project":"openstack/pbr","branch":"master","topic":"pep518-support","attention_set":{},"removed_from_attention_set":{"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-06-26 21:57:10.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-26 21:57:10.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-06-26 19:51:56.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Iacbb7237870cc3ae529703a6fb434812d51129cb","subject":"Remove need for setup.cfg, setup.py","status":"MERGED","created":"2026-05-20 14:00:49.000000000","updated":"2026-08-10 14:34:48.000000000","submitted":"2026-06-26 21:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989390-pep518-support","meta_rev_id":"df383a59b321acd492bde106ad5dbb053071afd0","_number":989390,"virtual_id_number":989390,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-26 21:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2026-06-26 17:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-26 19:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-26 19:51:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-20 15:50:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-25 13:42:08.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2026-06-25 23:09:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"050eec37867ea74e0f557a588a3b97b61559ac55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-20 14:00:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09cdf3df903b47f41d2efc6a141386d0350da940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:50:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/053c2e255c0b4be18a5666dadc713227\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6ae57b29d344b6dbbbd483df6eb728b : SUCCESS in 2m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30354710759b4285afae49f6fcfb3de2 : SUCCESS in 1h 45m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a76d2e3d46a435d97dae2cc6d86ba4d : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd0dfe9f867d4c6b832428e0c98534dc : SUCCESS in 2m 28s\n- build-python-release https://zuul.opendev.org/t/openstack/build/132818dd171d484fa0c2a8488659d7e4 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32a5e2b1af4846ff973172ac561f281e : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9f96aa97cee04f2197002bed534a3f8f : FAILURE in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaacd56a8cac40babf08e73800f5f309 : SUCCESS in 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f3ac5d0b5a4744079dc3731d834bc2b9 : SUCCESS in 1m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc20dacb1dc7488bbbd4f50472b39f1f : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27d3ae8483ed4f69bee8cf006e8088a8 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f227c1669b4a46da9bd96e5dcbde3073 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/002c739a4c0747f3a3b2af56baf0d9ea : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3fbac0619884da2a100574582719d5a : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7d27fca2bde4c858c1f0769a0622e05 : SUCCESS in 5m 13s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/97fd5cef1a8f4529b7386fbf048c719c : FAILURE in 9m 24s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/2eed44e9bca74cfc98d74365ab8a0f6e : FAILURE in 5m 53s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/db9cab3780ac4e9b82206ce6eae4c497 : SUCCESS in 27m 24s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/74dbb80c9cfd4846a2d10a4aefca3df0 : SUCCESS in 33m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4ca0f17b7b951226350184952982c3f385e5f73d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-10 22:54:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"88fda6ad2874b3f1099c8be1c95dc4b2bae76277","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 00:45:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbf0a2ad094c497393203252bdc75f15\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51c7abc6e94d49eb8faa2a3cacd9f03f : FAILURE in 4m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3611120103b476c97c22a32d3764aa8 : SUCCESS in 1h 45m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/563ffdfabaac401aab2a2aa548731b77 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcad177323784c7793f2ed78f60f9afd : SUCCESS in 2m 22s\n- build-python-release https://zuul.opendev.org/t/openstack/build/13657bbfe33345fbb6d311689a3c22a1 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea2a01a2b97c4d24b1e9ef48db7458c5 : SUCCESS in 2m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a8559dbed2df4bfaa8422cc5eae41683 : FAILURE in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf6b0861088b4fb685af39258f8eaec3 : SUCCESS in 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00e5e13498a14a6a89f95cb01d95e076 : SUCCESS in 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b0f280d89d248bda041ce65d821166f : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2c470cc21b04c9fa29b82928f9d7038 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8312948584844154bd92faa15a123ae0 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3de32da3ba7344ab9c32a27aa6b5c6cb : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46bf31ce29954eb69b0bf8b7a17db5f3 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6dc697032234ee0b3889498dcef0810 : SUCCESS in 4m 47s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/8a3a81174a5a4f9390a6a80f8a2d751e : SUCCESS in 32m 21s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c0b7e0acdaa24243bdc1e63cdd25f2e8 : SUCCESS in 42m 52s","accounts_in_message":[],"_revision_number":2},{"id":"b54dc6f238cbdefb38b13616daec0aa891703c88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-24 22:22:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"0a222997395f1ece1817f477ad6bc3705bed30bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-24 22:44:56.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a28d5b353d364fc6e65bdf82e75a49c06b0b2acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-24 22:55:19.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"72249b3a1227c42ed1175185ccdcb7ef45b596de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 00:57:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fde4ec76a854ea1a2218fbb11d91a77\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/695dbfc5d1494ce9a69a315812688537 : SUCCESS in 3m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20c6872be6eb4ea0a9357828ece8591d : SUCCESS in 1h 55m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b6ceffff44435da9a314336e05f8f2 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48be875b169345c6b062315bf31ec892 : SUCCESS in 3m 25s\n- build-python-release https://zuul.opendev.org/t/openstack/build/4daf6b3deab34b3498c1f9085d65d5ad : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd36cef5efc4417086eb85c55733e732 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c199b119f71413f81ba3f762b671df7 : FAILURE in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/259824edc1304235b6f04a85765131f2 : RETRY_LIMIT in 2m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec69b375b5bb4b3393f7cf0d09f76797 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9526e352b55e44ffb85ed5a61ed98fff : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ec6d7d5bbe04e43948981e3e69e193b : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7ffac3b1ce44a46aa648097d8f49163 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89d7ca1bb65a4740a9f181fc03f1c1fe : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5906ce92c07e40f38267a984964f54fd : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e0a3364e2cf4144ac2d4198cfd0ff5d : SUCCESS in 5m 02s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/8f91336ef1e6443fb1a9b1f8523479d5 : SUCCESS in 29m 52s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/e765f56ba0bd49e7a88fef939266e41a : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":5},{"id":"e4ebebb7816c764a2a8ce46b55d6680a17a54b14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 13:38:19.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"6fb66f641c7a240d802883953eee5f87ea55b51b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 13:41:07.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"17e517047e72067fdc187b2045d26e294744d3fa","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 13:42:08.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"4399c9510d715a16f0cd5541bf2129af019fd475","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 14:05:41.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"a85320458761ffa5b840ef3037638afb540ea4db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 14:06:29.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"91cc42b889c9f0899cf086eac872d4188ceea0ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 15:24:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2715e3c5a7224485bd7c8f76247d19c4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3baca8126ad84b1e914e75e0d520299e : SUCCESS in 2m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f6a3c87d20e4078ba31e79cf2d7047c : SUCCESS in 1h 05m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/743db3e6c79e4f2aa2faaceb8b4d8f3f : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec80c72524974ecc8200f83fa0420254 : SUCCESS in 3m 59s\n- build-python-release https://zuul.opendev.org/t/openstack/build/7076b4080ae940ef8019d4af96e8269f : SUCCESS in 1m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1762bada36e4ef9ba320bc4cda495df : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fe5418b8c0e48c5ab3a64bc0c885665 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/523f31304df0457294e8512169080726 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5a828d8c277449c837cf2c5b87ad52d : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53f3a4a39efb4f84af0763fe72600acc : SUCCESS in 5m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/31d3744a165649a4b05b39f89fc963f0 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3281026478374cfebc70f8ea9fc087cd : SUCCESS in 6m 20s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/ff06d87fa04d42be99a2b3034dd0b99d : SUCCESS in 39m 34s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/6dd074aca6c040bdaa3bd8d8f1d113d7 : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":9},{"id":"5074b2215f542ba5a4e3e3173319677fb7ac8eb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 20:46:28.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"0b4ec90f9ed60a28763826e4fb38bbf9e9f5ef79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 21:45:05.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"854b04fb50cc421a41a6c78b151d84bcd06963a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 22:05:11.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"edfecffebb1b798b568b3fbc5c59b3ff8cd39467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 22:25:14.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3e817035079157292b773cd60834d9086f0782e7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-25 23:09:17.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"f0216d55d59d368185073765a9ac221e059baf6d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 23:20:26.000000000","message":"Patch Set 13: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"897c7775ddbb9b056be7f62876f1ae0e5a9ebed5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 23:23:10.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d2fd2831cd1d2033e20b1c57f6e2bc2736b8c907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 00:17:32.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60d4e1e451e14e838b3824ac342490b3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/557b3a60a5d746fdb1578a5d7d3ff39f : SUCCESS in 2m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2e70eaab2334157ae2487ec93408c93 : SUCCESS in 1h 44m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab80e729e5cd4facb385e924cad901bd : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce1ac00f69c4a19b41994cfd438d2e3 : SUCCESS in 3m 12s\n- build-python-release https://zuul.opendev.org/t/openstack/build/e1d96a45bedc45cab6dc3b092dbbe263 : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89710abbe4c744d2a69f3e4af54f7f30 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51673ec2229246c0b36a411c2ac74afd : FAILURE in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f474356cf879444085bb21876933bc4a : FAILURE in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b354871575c24313aa62db2dae6d09a0 : SUCCESS in 10m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1156168b69274d8b8c8c7aea1de0ed39 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22232754f2ce48f4932231f5cfbea49b : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91db7a1c2de34dfcbc7b1246c3fabeb4 : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bb7c8be1b604efd9b7ae466e2367161 : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93cf94631fa14405b05d52decb8abe57 : SUCCESS in 4m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dffbcd35d5e5415f9a9ed3df010d1422 : SUCCESS in 7m 34s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/a97fa418d4b9404cac69c3c3dc8009f3 : SUCCESS in 15m 40s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/555e82566bbe4588a721adcdca060352 : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":13},{"id":"8cec21181c61c8cee9bb5ca64c60fed7e5f52bbd","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 04:59:37.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"55cd68909f0a09dcdd711c6264b47e8e51d91a30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 05:51:16.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"3f5cde6b25351bc1f46b759c778657929d91b5b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 07:47:40.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8369f42941d34a63b7b85ca03d516af4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bb69373867c549b685e30e8c80fe21b5 : SUCCESS in 2m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/482b8573cfb9424e9e033ec6e768052b : SUCCESS in 1h 51m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412411ece0894c3d87ed6676dfc80cfe : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2144a0345a435baf6631c4a06943e5 : SUCCESS in 2m 13s\n- build-python-release https://zuul.opendev.org/t/openstack/build/d800d0b642f94e5d903ed1f4b1007d40 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59a11b023cef482e963a849dfc7c2172 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6922b23ba58149e5b79d2fdc04a0661b : FAILURE in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9802cfa946ee4f1eb39f49475ac7b91a : FAILURE in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fe1cad90dd74814b96144b150712d65 : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/528ebbbfd5834b3791e43394c5f40a0f : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f50181a0620a4c4c9fa88f0369aef4ba : SUCCESS in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf30eaf71c194eb793b07cae932356a1 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66b1b4f50a714ca4859b4cc1e9897f84 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe11f802d0ed4f499bc681046c5f404e : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73107b3bb80c479987dadb6672817578 : SUCCESS in 4m 42s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/69e065b7c5f748e0b766a247324ce5c5 : SUCCESS in 16m 57s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/d0788721c1a2405c981cdb3781dd8ad1 : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":14},{"id":"3c033d6d124b16127b0670fe2d56c971a91faf0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 13:54:39.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"e3d4a94571029447bb2f1f7fbd19270c18e9b4e8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 13:55:53.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"d8231df1822668d39e90d5c95d51ed6f458bd001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 15:25:02.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da495c976eca42e8ba76696482eed8b9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3481db3ab4b74c6db863c53fa3f169e5 : SUCCESS in 3m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/115d54a8fccf4ba4ab4034ee4f0c80fb : SUCCESS in 1h 19m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe56bd96d69437880d0bde5cd3991d0 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e036d42fbde4259a2e5f3d6608f82ca : SUCCESS in 4m 35s\n- build-python-release https://zuul.opendev.org/t/openstack/build/cb98d5072a384cc8ab46cd1944b94530 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70b8e39b6bec4de58aa54a188fc63eab : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23b333de69a14b8fad66d4c4c0833cc2 : FAILURE in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0ed27db990442d49c33d81e1cad5258 : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/396fadf04f854c67823c841d110c0480 : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/816b363e471d4713a43b686b173af02c : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03a35060c5ed4c2fb8ab9b1525cfbf53 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5603abb850341068b97315fb1662e9c : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20539504fd274c6cb6872364cdb73164 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e863e8967f9f4b598198efcfd51f9026 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa5408776ed64c01b3d85e548e38825d : SUCCESS in 6m 45s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/97cd8711dc47408abab3b9d9b553a6b6 : SUCCESS in 30m 28s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/ad70e8a3876c42c0be3a8b0cf5b543b1 : SUCCESS in 16m 02s","accounts_in_message":[],"_revision_number":15},{"id":"ea893c41ada52cf7449f67273fc1ea152989e963","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 17:12:12.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"89ec0cedb7cd45724739e1416f047be2fbfa4bc1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 17:38:35.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"b4796b7d3669baba2168d677f9a2002646dc79dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 19:20:00.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/778bae20e83d4e63b9404841f332825b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/977da0f5ee2f425eaab768ff0a3634a7 : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/caea39dd1984468e9f5490dd6c4325de : SUCCESS in 2h 01m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/354d8d72a9b6417cadd59520cf5d3a00 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06adf204a6f0401b98a536203c09b535 : SUCCESS in 2m 45s\n- build-python-release https://zuul.opendev.org/t/openstack/build/6b3548f69d104939904972c49c7fd305 : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72c3fd1df61c4429924aa789324fb393 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a90873661ce74fbb9014e951b6210353 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/616511a019ad4675988754404a39f191 : SUCCESS in 7m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99377c8cd4824f029426f2a13e17373c : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b48f546baeb641198b12b3584b08f73e : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb0a003552344edba9cbaf5c44f33c6b : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a6df5336caa404e929ca59f8f4094ca : SUCCESS in 6m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83a46a1378bd420b94228f4efec532d1 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3ca4015ab354a63b63d4a63da12981e : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/daf7aceb02794e7ca938717403c5bf69 : SUCCESS in 4m 11s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/034c61f4f4594f33b4cfd8f668cca465 : SUCCESS in 34m 35s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/573ae47e30cf4b1ba478fb8248bc5bfc : SUCCESS in 22m 42s","accounts_in_message":[],"_revision_number":16},{"id":"b255cf621fc9072f814606791cbd35a5fd57aa22","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-26 19:51:56.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"790b5b1e92c69e16feb3b10defffa749a1fda667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 19:53:38.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"cd5a27576ef4385bde15c2d3490a679e3a0bb601","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 21:57:10.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62333cb9f5c048558b0841404a34bbdd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c3cb56dea889476d89d582a820698718 : SUCCESS in 1m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a42bed3872f04106b1782245f7057a0b : SUCCESS in 1h 58m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0e51b2a47e145a58f9096cb36e0a78d : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd5755c48c0945218257fcd571c2a1e1 : SUCCESS in 3m 10s\n- build-python-release https://zuul.opendev.org/t/openstack/build/1e299752c72041a0990b21e269ced770 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6376a5abf0c4d8f90cba9f21fe89dc4 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb0311a7edf5434b9328b515a78ab194 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a4ed6a6a12c4937a4fc2de8251f6b24 : SUCCESS in 8m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e86ce2ad9de84c948a5135bac7f810c8 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68ef9ab2c0394807bbaf240df84a83e7 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/001066b26b82494d9837d3ca0f294661 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/048ddef8b78548aba0125909b84375f0 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b12fe3ffa0574bddbcd9581cc1957517 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52bd02fc0f194039b4ea905d12e199bd : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6eda091b45b04799994be16c3e3c38c0 : SUCCESS in 5m 23s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/af4efd5fd1f94df497c0c591d63f4fb0 : SUCCESS in 15m 37s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/34a0945806b94c0a8396870020429e85 : SUCCESS in 16m 15s","accounts_in_message":[],"_revision_number":16},{"id":"6736136197c2301185ccd86f8a3f32074cfe34c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 21:57:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"947c3c00782fc58fce4b64a4ad36e0d526e24ae4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 21:58:32.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/312ee6eb329543b182555a5c70c28d1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79fcc373e70540db94901ed3b4cb4b2f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":16},{"id":"df383a59b321acd492bde106ad5dbb053071afd0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-10 14:34:48.000000000","message":"Patch Set 16:\n\nIt just dawned on me that we missed the opportunity to include a release note and update the usage documentation. This change is a pretty big deal as far as projects using pyproject.toml are concerned, since they\u0027ll likely be interested in cleaning up their stub setup.cfg and setup.py files now.\n\nShould I push a change for that before we tag a release?","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"3e8f971ba85305b993e240d7dfd1461edb13b4f9","revisions":{"25a8972935447398c34b15c9e580ccca0f2f4cb1":{"kind":"REWORK","_number":1,"created":"2026-05-20 14:00:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/90/989390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/1"}}},"commit":{"parents":[{"commit":"df359fb9f7f2e989d713b6ad3d29aecfdf926d32","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/df359fb9f7f2e989d713b6ad3d29aecfdf926d32"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 14:00:31.000000000","tz":60},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/25a8972935447398c34b15c9e580ccca0f2f4cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/25a8972935447398c34b15c9e580ccca0f2f4cb1"}]},"branch":"refs/heads/master"},"5074132d0a036f30f796ccf431d59598f947445f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-10 22:54:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/90/989390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/2"}}},"commit":{"parents":[{"commit":"996c25b24d0f105a6e2f7f17d66982e09d741345","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/996c25b24d0f105a6e2f7f17d66982e09d741345"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 22:53:39.000000000","tz":60},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5074132d0a036f30f796ccf431d59598f947445f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5074132d0a036f30f796ccf431d59598f947445f"}]},"branch":"refs/heads/master"},"14f71ee5cb08926042cec968522669e54b4af7df":{"kind":"REWORK","_number":3,"created":"2026-06-24 22:22:27.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/3"}}},"commit":{"parents":[{"commit":"6f639b25f890c40a09b22b980165f532f94651dd","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/6f639b25f890c40a09b22b980165f532f94651dd"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-24 22:22:23.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/14f71ee5cb08926042cec968522669e54b4af7df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/14f71ee5cb08926042cec968522669e54b4af7df"}]},"branch":"refs/heads/master"},"97a5e19a0eb92bbdc56510f9c2261995919c0dba":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-24 22:44:56.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/4"}}},"commit":{"parents":[{"commit":"90bb892d5915f900ae6accaa5962389539a99942","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/90bb892d5915f900ae6accaa5962389539a99942"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-24 22:44:40.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/97a5e19a0eb92bbdc56510f9c2261995919c0dba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/97a5e19a0eb92bbdc56510f9c2261995919c0dba"}]},"branch":"refs/heads/master"},"c634200656c226396b3efb0d11844123ff564625":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-06-24 22:55:19.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/5"}}},"commit":{"parents":[{"commit":"a519dda2fab67140edbe78533a10873b1fbbc86f","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a519dda2fab67140edbe78533a10873b1fbbc86f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-24 22:54:59.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c634200656c226396b3efb0d11844123ff564625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c634200656c226396b3efb0d11844123ff564625"}]},"branch":"refs/heads/master"},"2913b916cff3b464c41e5fe610aaeb0b8b685584":{"kind":"REWORK","_number":6,"created":"2026-06-25 13:38:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/90/989390/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/6"}}},"commit":{"parents":[{"commit":"f5e8dde8ff38021f20de4f1adfb5756331389aa1","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f5e8dde8ff38021f20de4f1adfb5756331389aa1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-25 13:38:07.000000000","tz":60},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2913b916cff3b464c41e5fe610aaeb0b8b685584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2913b916cff3b464c41e5fe610aaeb0b8b685584"}]},"branch":"refs/heads/master"},"cd5a4ab50fa2f4873c6e722bdb7d4086e1f7fa11":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-25 13:41:07.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/7"}}},"commit":{"parents":[{"commit":"5cb00539c4afdb038df79f0216c72504a701eaa9","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5cb00539c4afdb038df79f0216c72504a701eaa9"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 13:41:01.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd5a4ab50fa2f4873c6e722bdb7d4086e1f7fa11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd5a4ab50fa2f4873c6e722bdb7d4086e1f7fa11"}]},"branch":"refs/heads/master"},"15e59f2042ace7cef98818a90e5f1f14890ae916":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-06-25 14:05:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/90/989390/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/8"}}},"commit":{"parents":[{"commit":"3e3442b09e8d20a0a670cb610491951fcd03ad01","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3e3442b09e8d20a0a670cb610491951fcd03ad01"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-25 14:05:00.000000000","tz":60},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/15e59f2042ace7cef98818a90e5f1f14890ae916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/15e59f2042ace7cef98818a90e5f1f14890ae916"}]},"branch":"refs/heads/master"},"798482b4b61772acb39a455a7a8e5ad80d524eff":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-25 14:06:29.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/9"}}},"commit":{"parents":[{"commit":"1396bd29a9b639e13c7f8e32d422b3963f915291","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1396bd29a9b639e13c7f8e32d422b3963f915291"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 14:06:08.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/798482b4b61772acb39a455a7a8e5ad80d524eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/798482b4b61772acb39a455a7a8e5ad80d524eff"}]},"branch":"refs/heads/master"},"7167f021056e882e59a37c2c0e369e9eb1e312bb":{"kind":"REWORK","_number":10,"created":"2026-06-25 20:46:28.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/10"}}},"commit":{"parents":[{"commit":"da2055c760dced21132878ce58ffea0a0e951399","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/da2055c760dced21132878ce58ffea0a0e951399"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 19:12:23.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7167f021056e882e59a37c2c0e369e9eb1e312bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7167f021056e882e59a37c2c0e369e9eb1e312bb"}]},"branch":"refs/heads/master"},"867dcb5501574e31c8f2347476239b07b3203e47":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-25 21:45:05.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/11"}}},"commit":{"parents":[{"commit":"8e679578433aaeaa1447fe5e71d0c39be440317d","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/8e679578433aaeaa1447fe5e71d0c39be440317d"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 21:45:00.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/867dcb5501574e31c8f2347476239b07b3203e47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/867dcb5501574e31c8f2347476239b07b3203e47"}]},"branch":"refs/heads/master"},"744a3227169b7c1dda10130d28b4d5f55000ecf5":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-06-25 22:05:11.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/12"}}},"commit":{"parents":[{"commit":"1e0ef19e4dcff6496aaffd5996dbceeb48910a61","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e0ef19e4dcff6496aaffd5996dbceeb48910a61"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 22:05:06.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/744a3227169b7c1dda10130d28b4d5f55000ecf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/744a3227169b7c1dda10130d28b4d5f55000ecf5"}]},"branch":"refs/heads/master"},"1186fb608105cc2eba6b9a06ea26e1c02caaad5d":{"kind":"REWORK","_number":13,"created":"2026-06-25 22:25:14.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/13"}}},"commit":{"parents":[{"commit":"1e0ef19e4dcff6496aaffd5996dbceeb48910a61","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e0ef19e4dcff6496aaffd5996dbceeb48910a61"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 22:24:51.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1186fb608105cc2eba6b9a06ea26e1c02caaad5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1186fb608105cc2eba6b9a06ea26e1c02caaad5d"}]},"branch":"refs/heads/master"},"462603b0a1c01d8698895edd539fcd9e6044d4b4":{"kind":"REWORK","_number":14,"created":"2026-06-26 05:51:16.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/14"}}},"commit":{"parents":[{"commit":"1e0ef19e4dcff6496aaffd5996dbceeb48910a61","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e0ef19e4dcff6496aaffd5996dbceeb48910a61"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-26 05:51:09.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/462603b0a1c01d8698895edd539fcd9e6044d4b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/462603b0a1c01d8698895edd539fcd9e6044d4b4"}]},"branch":"refs/heads/master"},"970b9275c0ccd1ea77f3c40264f8544b45cb2f25":{"kind":"REWORK","_number":15,"created":"2026-06-26 13:54:39.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/15"}}},"commit":{"parents":[{"commit":"1e0ef19e4dcff6496aaffd5996dbceeb48910a61","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e0ef19e4dcff6496aaffd5996dbceeb48910a61"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-26 13:52:09.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nAdd a guard blocking this from EOL pythons 2.7 and 3.6, because the\nway the setuptools for them works with easy_install in tox creates\npath conflict issues on entrypoint probes. Supporting 2.7 and 3.6\nfor continuity does not mean adding support for super modern features\nfor them.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/970b9275c0ccd1ea77f3c40264f8544b45cb2f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/970b9275c0ccd1ea77f3c40264f8544b45cb2f25"}]},"branch":"refs/heads/master"},"3e8f971ba85305b993e240d7dfd1461edb13b4f9":{"kind":"REWORK","_number":16,"created":"2026-06-26 17:12:12.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/90/989390/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/90/989390/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/90/989390/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/90/989390/16"}}},"commit":{"parents":[{"commit":"1e0ef19e4dcff6496aaffd5996dbceeb48910a61","subject":"Remove more unused code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1e0ef19e4dcff6496aaffd5996dbceeb48910a61"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-07-04 18:57:36.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-26 17:07:33.000000000","tz":0},"subject":"Remove need for setup.cfg, setup.py","message":"Remove need for setup.cfg, setup.py\n\nAdd a setuptools.finalize_distribution_options entrypoint, allowing us\nto plug in to setuptools\u0027 distribution generation and avoid having to\nkeep setup.cfg and setup.py stubs around.\n\nAdd a guard blocking this from EOL pythons 2.7 and 3.6, because the\nway the setuptools for them works with easy_install in tox creates\npath conflict issues on entrypoint probes. Supporting 2.7 and 3.6\nfor continuity does not mean adding support for super modern features\nfor them.\n\nChange-Id: Iacbb7237870cc3ae529703a6fb434812d51129cb\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nAssisted-By: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3e8f971ba85305b993e240d7dfd1461edb13b4f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/3e8f971ba85305b993e240d7dfd1461edb13b4f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
