)]}'
{"id":"openstack%2Fpbr~989391","triplet_id":"openstack%2Fpbr~master~I48447a58d88088516be135c2cc30287089620f28","project":"openstack/pbr","branch":"master","topic":"pep518-support","attention_set":{},"removed_from_attention_set":{"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-06-26 19:47:36.000000000","reason":"\u003cGERRIT_ACCOUNT_2\u003e replied on the change","reason_account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-26 22:35:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I48447a58d88088516be135c2cc30287089620f28","subject":"Hook in LocalSDist override","status":"MERGED","created":"2026-05-20 14:00:49.000000000","updated":"2026-06-26 22:37:15.000000000","submitted":"2026-06-26 22:35:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989391-pep518-support","meta_rev_id":"b3a682a2e355f86e935026b01716977bbae24711","_number":989391,"virtual_id_number":989391,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-26 22:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2026-06-26 19:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-26 18:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2026-06-26 19:47:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-20 15:53:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-25 13:42:03.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2026-06-25 23:09:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"c0d3b8e440c9d70c6fbcedfba9085ec9f572f122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-20 14:00:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e16440b51878b0dcf80655314084e488b32c5cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:53:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/32ad9189a7ed4eae9fbaef920c646441\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e25299c014f4379a1bb83e07e444ff6 : SUCCESS in 1h 48m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2742cab6b7ca4bf5ac47d71a535587d6 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e1b22eb39c84503b4543796a1f7f9bf : SUCCESS in 2m 45s\n- build-python-release https://zuul.opendev.org/t/openstack/build/3ae139b0258d4850aa93a41649068f75 : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec526686234e4a0d968446e7b66b6ba6 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a81df8e24994ec3ba30ca869f3f36a7 : FAILURE in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc80934d3874419898b5f6066c1dae0b : SUCCESS in 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f3839e274fc42c286d7d6959486f15c : SUCCESS in 1m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af9d8784a8a04b0897de5ec429cda64d : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e3d962a098d4fe3a10204e300147eb1 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ae96612231a4d00ade742f9beb01d1d : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c80406bbbf5e4ff4a94e4292dae7ce61 : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d30bb6e10f2f4f1296ab4e925d9d0075 : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/def0915895ac4f93b58f1cde6d4164d8 : SUCCESS in 4m 45s\n- pbr-installation-openstack-jammy https://zuul.opendev.org/t/openstack/build/eaa12ab6086442eba6b713472ec5fcdc : FAILURE in 11m 23s\n- pbr-installation-openstack-pip-dev-jammy https://zuul.opendev.org/t/openstack/build/e06865d66f4e42b09e524d057a900118 : FAILURE in 8m 59s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/95a9cab543794a76930732925dbbd7da : SUCCESS in 28m 31s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/8e0b9fbc20f144a89175bc0b7cc738a1 : SUCCESS in 27m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4f0b320acaf9144e10b9fe8a914513206153cef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-10 22:54:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"71d5c1299f749f657c04b8a856e4a05b1693c18e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-11 00:45:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7aa31799c46b4f68a478cdab44c7fa32\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d95989ca25b2414ebe8bc2fa76304799 : SUCCESS in 1h 47m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6948b1399c2948f785d2186de49da00c : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4334f3e6140a4d73907427ae8cc390f5 : SUCCESS in 3m 01s\n- build-python-release https://zuul.opendev.org/t/openstack/build/f73f49a558254bb29bdbf0ac6a4cb2f2 : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5738d82025354afe9f8745135c9ec20f : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21dcc26ec0b9484c9e1de7bb75699b30 : FAILURE in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02bf54dd8c1f45debadfe6740321d336 : SUCCESS in 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10c5d5694d8c43fb814b725f3efec5ee : SUCCESS in 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73ae6709c08542d5984a00f4d80729ed : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0457dcbd182a435bb6c3bd787c5bf502 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/169ca75c3c4e491da46b4598b7f07a63 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a78bb33be58438489658c16ee250321 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/678c05f1a78d482ea3dd78112d24f457 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/661387e95b87423fb112115bfca9dce8 : SUCCESS in 5m 08s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/31571bf6078741cdb81a6d7eb8597a30 : SUCCESS in 32m 13s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/9b2fd98ebdcc4bbb8e6baa67c078d4a2 : SUCCESS in 31m 59s","accounts_in_message":[],"_revision_number":2},{"id":"c9dc43ef5f9aa9b836e6f2b54428bfe7986abbc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 13:38:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a0c2056468a11c3949883ccbf1e6eacbb7021140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 13:41:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"87d03464c7f557c5b800d1b25c48e44f2d63b3e5","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 13:42:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"de8a2347991d1c9719ecfe94b7c3bf2c782640a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 14:05:41.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"0d6afa434c8212d3433380a621d28ee5dec318d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 14:06:29.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"d86d007f4f30d66fa874b03858dd8286d3abe4b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 15:13:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c4f170682d84f5884f4044690c23df9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41f6974a93434e2d967345cea56389cb : SUCCESS in 54m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658c415adf004bd388bcff5062cdf8b5 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e60e798a56874698af9603da18217e4e : SUCCESS in 2m 30s\n- build-python-release https://zuul.opendev.org/t/openstack/build/fdf32c6fe3574216a20091d773859e04 : SUCCESS in 1m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11f861c4d37b4ca78917698db2cec6f2 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/589990c669b84092a929696c6c7eb011 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaabf3da10514250aadfaff8ed96046c : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acd879d7b4e0426d86b1100a570bee84 : SUCCESS in 8m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97a97b02e81b4bcbab026c776dc11454 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2416736d12184a6193cbcfc682df5753 : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aead8910bd604aeabc226cde7075e719 : SUCCESS in 3m 45s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/0d3b15cef5c348969b269b0383abe04a : SUCCESS in 32m 27s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/4b18171b644b45a1aefb74a6ab472f8c : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":6},{"id":"98ffc0f727094bd6887b0f9d3f4cee0bcdf6944d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 20:46:28.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"f2fe4a74126bae212f61ba7167c709975b4c1b80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 21:45:05.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"2058c7f00de6e04eaedbc4281e0c9cce2c6e40ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 22:05:11.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"40a7a7fa9505abbe6885c52e4e00a464d585d238","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-25 22:25:14.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":10},{"id":"9c858d7bf29517f2abcf0df854a72fed0cf4d701","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-25 23:09:48.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"545221571ba9ecfddd477180503fcd79cbefc0e5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-25 23:20:33.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"caf82150536b91aaeec8667aaa696e97ff27e702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 23:45:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bf7c1cab90a42be8511acbd7a4c527e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a36c067cab742aca44ba2cbacaf83bf : SUCCESS in 1h 16m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ffd66d19e254d8e91a2b2e9db6d4688 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e420b58a337146ae8e2ff3b6c047870a : SUCCESS in 3m 04s\n- build-python-release https://zuul.opendev.org/t/openstack/build/2e437b59c2bc44c9881bb7fabddb99f4 : SUCCESS in 1m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1abccd103bbf4c07930ffa5a558179a5 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/966e9635c53e4db0a2e6009de646a960 : FAILURE in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4f8b8869ed2434696b77e479c60bb8b : FAILURE in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/305afe181b3d4080baad447d68b6882d : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3187fb2c48341a89d91e6430f114eab : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9a94e611d064958bbe5a7930e01476e : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a6e0939556446a2af9b43c27291f2e7 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/394e20555cfb4b51b885ca212594656c : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d187e522917414d8eadafce596fb914 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6efa6c5c0e74433183f2c5caed00091f : SUCCESS in 2m 38s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/d23012c5bccd431da3a52c7843e7f8bc : SUCCESS in 34m 22s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/4b6ffe388c85401db77308b7197331cb : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":10},{"id":"1ea79c6ddc9c06778ed248bf5b87ecce426119d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 05:53:09.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"dede834d5a5930dbce6ebf4df56dc0c02d188c91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 07:49:55.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb3a77cf8d404653b8017df7ddd8583b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a0048b703d249239ec12fd71af3791d : SUCCESS in 1h 53m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4079db650b62472ba7c7bc8c607b1e4d : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6ded868757430db3e416edad20c793 : SUCCESS in 2m 24s\n- build-python-release https://zuul.opendev.org/t/openstack/build/fed32d51901f4bc7a3f4d68096a3a752 : SUCCESS in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91e39bf626b3464ba34f566adf1358f8 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af52e02730d14b5dad0415e5c5f9228d : FAILURE in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73c227c99b114902ad2a24dd86d862c1 : FAILURE in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfc6b7401cce4d668d0a830aaac460bf : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a40ad1176e9341ab87610efbd775783c : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/685ff3d2bf654d1cb5862892c3bccbe0 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0650e501f57f41df97abd9e2e2af2a08 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8ef0aba41bd459fb6c45a999734e2c9 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89078a4be37744dd9b0866f09b5ce718 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f4c2657ffa3461cae90eb5a45272548 : SUCCESS in 4m 31s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/a8f324f9f3b745aba9099344b34a2e48 : SUCCESS in 18m 34s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c21757e237714c11a27dc7cf8d0a23aa : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":11},{"id":"bc922b5c1fa81eeda0955ed6f0a4d8b73f0bc464","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 13:54:39.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"dff6a93002cc445e77fb6ec1a259363f362b50d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 14:59:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b78f11ea0a147fabc0f9a3cdb5de9cc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84593081590c4468bdd23740e0f21918 : SUCCESS in 54m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51dc8ecbaba743d0a9a9f9f4bf1930fd : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b594e2df212348d78cbf3a38e59ba896 : SUCCESS in 3m 23s\n- build-python-release https://zuul.opendev.org/t/openstack/build/32786cc726d940a788d80694bb729150 : SUCCESS in 4m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99368ee09b6048d99dfd343d2b1ba666 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac1c81e4ba7a4bd2bae81037eb5e2ded : FAILURE in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd9d59c22abf4c5dbca38994b5d06d93 : SUCCESS in 8m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6237988ef074eca89afdb85837bc18e : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b5cc2f93d944f29bb7c6aec4e47de26 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb58ab0ae26345d2a8854462b71a1a5d : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5aad8e600ec4e90ad6022df47cdcf3b : SUCCESS in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5448109f643b424bbd03d6c4aa9ce73e : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5899220d8afa4e8bba3b8b93def7d0c5 : SUCCESS in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fca99bf14f74db4b8839d3ee3f17777 : SUCCESS in 4m 20s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/06fd31f8c5b942aea10a49734f596b5e : SUCCESS in 34m 56s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/09819eda42454172bec97b21c6744319 : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":12},{"id":"cbbaa9359f4a85b4273cd96640c8de7a295ba2ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 17:35:03.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"efae168bf5cccb03a40d0d02c0c77bd81927b2e8","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 17:37:59.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"8145fba04bc4ed96a943ae608c33eaf4a1485be5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 18:46:21.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"a27fa58b80cd245f3eb985f3d7d1a8b79692c042","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-26 19:47:36.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"f7b5b66a8302df59db29c3b223cab9e6cbe3a99c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 20:27:02.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0429b39e5d74f209f763c39ac083907\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9e33fea8b234d10bc0992e46356cf74 : SUCCESS in 1h 37m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/548e731df3a34e41b5fe52c46b3de49d : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/902b29332b3f4243aa455830b783bd75 : SUCCESS in 4m 44s\n- build-python-release https://zuul.opendev.org/t/openstack/build/66c6589bba5e4d0fbc334b0f9873e592 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bb0c86892254995b365826abe0bb062 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a4be04864434617a23477f6bfe04791 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a72e206670d4ead8498654add66a721 : SUCCESS in 7m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e993cc5c2e94a1fa00c24a0742e1628 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/707227c51318434a8fd19c80f9b36759 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68a554abfd4542f3ad85415687729c23 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef9e9649bff145ba818ec94f28bf99c9 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fd6550d281f4c54a1585aabd2728acd : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3086bb87f34742e5a0137308f8e92b17 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fe183fc1a3b4106ad3b9b0b5590a31a : SUCCESS in 5m 34s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/cab75dc1394b4876a3aae214418eb545 : SUCCESS in 32m 18s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/c99d6da0a2a7468b96b41bc5b14292fb : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":14},{"id":"ba06c4c150e7c25d6385de3ab6336002475304a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 20:40:04.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"f33f254fdbfe755d36cf60480cc5c7fdacfb3612","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 22:35:20.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f9e7275ca554d33a538d183149fc829\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1bdde1ed1fa247388a4d034681af7388 : SUCCESS in 1h 48m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/112e7f26812a4ab8899f1a88f700cf2d : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e60399e75d44449a7b3a405bd32538 : SUCCESS in 2m 55s\n- build-python-release https://zuul.opendev.org/t/openstack/build/1d232156c4f54f228cfa7db47eb64743 : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd6f8d6c73264c8a90d140f05c48c9d4 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2df15ae2b0a54b6d93c753d4cb1cdf59 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bc5044d0e254df888d8272376700bbc : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88d35587bd28488392c182f71c2ff732 : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0e0373a77c94da5a86946815d9e8d11 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d21eda0a8774dff878f8df23ada35be : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f69b5c2af116405ea5a0c040479f567c : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eef90432bab746728d62f8f790cead1f : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/35ac49972b954f4cbadf868e9f6182ec : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4303205981c4cba983bd34c4e8c1f3d : SUCCESS in 5m 20s\n- pbr-installation-openstack-noble https://zuul.opendev.org/t/openstack/build/91a18cb67a7e4196a648085f62be855c : SUCCESS in 18m 16s\n- pbr-installation-openstack-pip-dev-noble https://zuul.opendev.org/t/openstack/build/9e5874de64184ef387df238b6aa27849 : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":14},{"id":"f399f062d3a2dae5bf820244705547004817b31f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 22:35:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"b3a682a2e355f86e935026b01716977bbae24711","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 22:37:15.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abf374b5570d4ce5b705511bb864bd8d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a8f9352bc74122bf2801a2094d42ca : SUCCESS in 42s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"5f9299c83b9d70e373d27b87612968be1468dc5f","revisions":{"5afaea1840aa0d1d91a7882b8224ba795dba9cef":{"kind":"REWORK","_number":1,"created":"2026-05-20 14:00:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/91/989391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/1"}}},"commit":{"parents":[{"commit":"25a8972935447398c34b15c9e580ccca0f2f4cb1","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/25a8972935447398c34b15c9e580ccca0f2f4cb1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 14:00:31.000000000","tz":60},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5afaea1840aa0d1d91a7882b8224ba795dba9cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5afaea1840aa0d1d91a7882b8224ba795dba9cef"}]},"branch":"refs/heads/master"},"a225f2af519f75213b56ff539add2fc1e4cbc97b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-10 22:54:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/91/989391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/2"}}},"commit":{"parents":[{"commit":"5074132d0a036f30f796ccf431d59598f947445f","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5074132d0a036f30f796ccf431d59598f947445f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-10 22:53:39.000000000","tz":60},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a225f2af519f75213b56ff539add2fc1e4cbc97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a225f2af519f75213b56ff539add2fc1e4cbc97b"}]},"branch":"refs/heads/master"},"a038782f99bb2aab6af29bfbb8d9f34e6362ea85":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-25 13:38:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/91/989391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/3"}}},"commit":{"parents":[{"commit":"2913b916cff3b464c41e5fe610aaeb0b8b685584","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/2913b916cff3b464c41e5fe610aaeb0b8b685584"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-25 13:38:56.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a038782f99bb2aab6af29bfbb8d9f34e6362ea85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a038782f99bb2aab6af29bfbb8d9f34e6362ea85"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"5074132d0a036f30f796ccf431d59598f947445f","ours":"a225f2af519f75213b56ff539add2fc1e4cbc97b","theirs":"2913b916cff3b464c41e5fe610aaeb0b8b685584","merge_strategy":"recursive","contains_conflicts":false}},"a40794521ab2a7c4e95704e57a793420b9d3b96d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-25 13:41:45.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/4"}}},"commit":{"parents":[{"commit":"cd5a4ab50fa2f4873c6e722bdb7d4086e1f7fa11","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/cd5a4ab50fa2f4873c6e722bdb7d4086e1f7fa11"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 13:41:41.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a40794521ab2a7c4e95704e57a793420b9d3b96d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/a40794521ab2a7c4e95704e57a793420b9d3b96d"}]},"branch":"refs/heads/master"},"9be4852d11d038a2cc4811ae97488c384ed1b849":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-06-25 14:05:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/91/989391/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/5"}}},"commit":{"parents":[{"commit":"15e59f2042ace7cef98818a90e5f1f14890ae916","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/15e59f2042ace7cef98818a90e5f1f14890ae916"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-25 14:05:00.000000000","tz":60},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9be4852d11d038a2cc4811ae97488c384ed1b849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9be4852d11d038a2cc4811ae97488c384ed1b849"}]},"branch":"refs/heads/master"},"f195fb4210f2e5904b5dfa1990059bdc613d7fd1":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-06-25 14:06:29.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/6"}}},"commit":{"parents":[{"commit":"798482b4b61772acb39a455a7a8e5ad80d524eff","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/798482b4b61772acb39a455a7a8e5ad80d524eff"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 14:06:08.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f195fb4210f2e5904b5dfa1990059bdc613d7fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/f195fb4210f2e5904b5dfa1990059bdc613d7fd1"}]},"branch":"refs/heads/master"},"9345925eb56fd8618270b639ae5f15fe26eeeb50":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-06-25 20:46:28.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/7"}}},"commit":{"parents":[{"commit":"7167f021056e882e59a37c2c0e369e9eb1e312bb","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/7167f021056e882e59a37c2c0e369e9eb1e312bb"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 19:12:39.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9345925eb56fd8618270b639ae5f15fe26eeeb50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/9345925eb56fd8618270b639ae5f15fe26eeeb50"}]},"branch":"refs/heads/master"},"816ef72712518af23b5ca339ab30c65f5a9ad60c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-06-25 21:45:05.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/8"}}},"commit":{"parents":[{"commit":"867dcb5501574e31c8f2347476239b07b3203e47","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/867dcb5501574e31c8f2347476239b07b3203e47"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 21:45:00.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/816ef72712518af23b5ca339ab30c65f5a9ad60c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/816ef72712518af23b5ca339ab30c65f5a9ad60c"}]},"branch":"refs/heads/master"},"e4972ecaf804f372071f473ab70cbb6ed54c27bf":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-06-25 22:05:11.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/9"}}},"commit":{"parents":[{"commit":"744a3227169b7c1dda10130d28b4d5f55000ecf5","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/744a3227169b7c1dda10130d28b4d5f55000ecf5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 22:05:07.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e4972ecaf804f372071f473ab70cbb6ed54c27bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/e4972ecaf804f372071f473ab70cbb6ed54c27bf"}]},"branch":"refs/heads/master"},"4d1698fdb4cd6f054e202bc11c75a174c13b405c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-06-25 22:25:14.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/10"}}},"commit":{"parents":[{"commit":"1186fb608105cc2eba6b9a06ea26e1c02caaad5d","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/1186fb608105cc2eba6b9a06ea26e1c02caaad5d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-25 22:25:09.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4d1698fdb4cd6f054e202bc11c75a174c13b405c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/4d1698fdb4cd6f054e202bc11c75a174c13b405c"}]},"branch":"refs/heads/master"},"dd81469e7e0f34f8c14095208c0d04172a4bfdc5":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-06-26 05:53:09.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/11"}}},"commit":{"parents":[{"commit":"462603b0a1c01d8698895edd539fcd9e6044d4b4","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/462603b0a1c01d8698895edd539fcd9e6044d4b4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-26 05:51:22.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dd81469e7e0f34f8c14095208c0d04172a4bfdc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/dd81469e7e0f34f8c14095208c0d04172a4bfdc5"}]},"branch":"refs/heads/master"},"0e6f9043f1df82bdc831ead8cb674d579e267a80":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-06-26 13:54:39.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/12"}}},"commit":{"parents":[{"commit":"970b9275c0ccd1ea77f3c40264f8544b45cb2f25","subject":"Remove need for setup.cfg, setup.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/970b9275c0ccd1ea77f3c40264f8544b45cb2f25"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-26 13:53:59.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0e6f9043f1df82bdc831ead8cb674d579e267a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/0e6f9043f1df82bdc831ead8cb674d579e267a80"}]},"branch":"refs/heads/master"},"ac0f7c2cabae1799dcb81a3233065059f2b40910":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-06-26 17:35:03.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/13"}}},"commit":{"parents":[{"commit":"b18451d0a20b03fee6fdc44b4dd1f4055677b507","subject":"Convert CapturedSubprocess, Venv fixtures to context managers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/b18451d0a20b03fee6fdc44b4dd1f4055677b507"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-26 17:29:55.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ac0f7c2cabae1799dcb81a3233065059f2b40910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/ac0f7c2cabae1799dcb81a3233065059f2b40910"}]},"branch":"refs/heads/master"},"5f9299c83b9d70e373d27b87612968be1468dc5f":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-06-26 18:46:21.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/91/989391/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/pbr","ref":"refs/changes/91/989391/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/pbr refs/changes/91/989391/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/pbr refs/changes/91/989391/14"}}},"commit":{"parents":[{"commit":"c71e67e2962700629f80063d758b27af5713acdc","subject":"Convert CapturedSubprocess, Venv fixtures to context managers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/c71e67e2962700629f80063d758b27af5713acdc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-20 12:50:15.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-26 18:46:21.000000000","tz":0},"subject":"Hook in LocalSDist override","message":"Hook in LocalSDist override\n\nThis gives us the ability to e.g. call reno if it\u0027s installed.\n\nChange-Id: I48447a58d88088516be135c2cc30287089620f28\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5f9299c83b9d70e373d27b87612968be1468dc5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/pbr/commit/5f9299c83b9d70e373d27b87612968be1468dc5f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b18451d0a20b03fee6fdc44b4dd1f4055677b507","ours":"ac0f7c2cabae1799dcb81a3233065059f2b40910","theirs":"c71e67e2962700629f80063d758b27af5713acdc","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
