)]}'
{"id":"openstack%2Fplacement~1002526","triplet_id":"openstack%2Fplacement~master~I0452e63bd08414579ba2bf97b8b0db510fe73f2a","project":"openstack/placement","branch":"master","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-08-26 19:29:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0452e63bd08414579ba2bf97b8b0db510fe73f2a","subject":"bump min version of os-traits","status":"MERGED","created":"2026-08-26 16:05:58.000000000","updated":"2026-08-26 19:30:30.000000000","submitted":"2026-08-26 19:29:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002526","meta_rev_id":"b7c9d6d235fe3e65bdb9263915edbd40f467c3c9","_number":1002526,"virtual_id_number":1002526,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 19:29:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2026-08-26 16:22:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2026-08-26 16:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2026-08-26 16:22:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-26 16:08:36.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-08-26 16:22:08.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2026-08-26 18:03:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bd30f8a488e962ae49be6f9b9d626ecb45c01b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-26 16:05:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56de752a339e32d3d5c2bb81d8d91563c857b281","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-26 16:08:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4749ef83eed1a287e3544ab17225e0d7619acf8d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-26 16:22:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e947018ea7cb5a950fe002138f005790d1b0c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:03:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e007d3ffe75f4f62bc4b47c4e6a47f43\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8dcbd199636049c1b78d74a5e4df2b15 : SUCCESS in 4m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/ac98239aca3d49fd9156e6d0a6278f59 : SUCCESS in 1h 00m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2c30f85d444f4f8989139fbfb341dddb : SUCCESS in 32m 48s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/926106a7b46a4b618569736e983d60de : SUCCESS in 1h 31m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77feab275cbe441b915e00331b519cb8 : SUCCESS in 57m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31de1305388f48d8ac3aed0d0426cdf2 : SUCCESS in 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/548d4f4215574f6fa1b5a19e5ff53518 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d29244c15bd74cc5b7cafc3fd2c541c7 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/570063733c814561a35955b1e937a331 : SUCCESS in 6m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b7592bb07c194a4b8a8d42f0eb39d292 : FAILURE in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54bb0a34ecad466aafcad83b741e8a8d : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20dd6422f3d6491b9d73efc41944bfcd : SUCCESS in 11m 33s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e923347ce27e4112be572993f000bf4f : SUCCESS in 5m 26s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e4afe5b33d334ee2997ce5442faaccd6 : SUCCESS in 3m 21s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7c21a5d2572545e29b4930e02c4161f9 : SUCCESS in 8m 26s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a1ef907d74c54f37913048060758d957 : SUCCESS in 23m 08s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/7bc461639ad845379d5e8e3b15945165 : FAILURE in 3m 27s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/b1b3a20cd6dd496eb711dcf6880fe4e2 : FAILURE in 3m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d108496cbdf4676ab5e24e6c8906598 : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c464e150d6e1464dea067fd203ec5801b7d1467e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:04:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6a4bf3d1756e0e9c9f24678b1990ed7c9607c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:29:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c14280423a24644ac9f6af01a2ebcf4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8cf4777cf5ad43779dec42572db267e8 : SUCCESS in 3m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/dcee7d39ec244f9f9cd41cf2eaeab441 : SUCCESS in 32m 49s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/24e67ed78f5040a099f7551a4eaed046 : SUCCESS in 51m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b23e78d6805b48da90fc8041f6df2b80 : SUCCESS in 49m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf4ad8c1cd164c4eaf58041c3494a5e5 : SUCCESS in 57m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d15971833b14f77bdf2d4e3258e4f27 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/274b805ee84e443aa7606032ad4e290f : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da5ddf17d5fd462f85f81d157eba4c33 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170256eeb38a463cad6207571e621f18 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0d93ccf4aae4285a6e1c79e9e4eb863 : SUCCESS in 8m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/87860cce76f44b878fbbbf055295daf6 : SUCCESS in 5m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/260ff5147a2e43b587a9925ab06e2d68 : SUCCESS in 5m 33s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/7ad83e08b4d8481aa697f50a0b8de601 : SUCCESS in 8m 31s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/579bc45f4a6849498070b413782e14f6 : SUCCESS in 20m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d7a5c04209348d48590c1b460ed44df : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0579bf0086b800f7e7ce6f5e2f6ad10a7522f4f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:29:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b7c9d6d235fe3e65bdb9263915edbd40f467c3c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:30:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66595602566047d487cf20cb7fd09a80\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29efe5beb70843539a32a48e2ec2548c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d35f974d52fdbb457ff2b6b7bff3741165dd30d","revisions":{"7d35f974d52fdbb457ff2b6b7bff3741165dd30d":{"kind":"REWORK","_number":1,"created":"2026-08-26 16:05:58.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/26/1002526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/26/1002526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/26/1002526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/26/1002526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/26/1002526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/26/1002526/1"}}},"commit":{"parents":[{"commit":"0d503c6df70aa3049c29d4c2d8672703e068463e","subject":"Merge \"Check overall capacity before generating candidates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0d503c6df70aa3049c29d4c2d8672703e068463e"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-08-26 15:57:57.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-08-26 15:57:57.000000000","tz":60},"subject":"bump min version of os-traits","message":"bump min version of os-traits\n\nthis increase the min version of os-traits to\nthe latest version released during 2026.2\nto ensure all standard traits are present\n\nChange-Id: I0452e63bd08414579ba2bf97b8b0db510fe73f2a\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7d35f974d52fdbb457ff2b6b7bff3741165dd30d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7d35f974d52fdbb457ff2b6b7bff3741165dd30d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
