)]}'
{"id":"openstack%2Fplacement~1004418","triplet_id":"openstack%2Fplacement~master~Icc6b40abcc455952694004585f053886eeb81005","project":"openstack/placement","branch":"master","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-07 18:09:16.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 19:21:48.000000000","reason":"Change was submitted"},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2026-09-07 19:21:48.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-09-07 19:21:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc6b40abcc455952694004585f053886eeb81005","subject":"[pyproject] Update setuptools config","status":"MERGED","created":"2026-09-07 13:17:58.000000000","updated":"2026-09-07 19:23:13.000000000","submitted":"2026-09-07 19:21:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"1004418","meta_rev_id":"8f2c7e83e7aa6182220e1638db5dbbb31ca83b7a","_number":1004418,"virtual_id_number":1004418,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 19:21:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-09-07 18:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-09-07 13:50:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-09-07 18:09:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 13:18:59.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-09-07 13:18:59.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-09-07 13:18:59.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-07 15:10:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7960585fed36764365a39edebaa0bdea46bb7010","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-09-07 13:17:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8affa57b0715ccf27324261446227721c8e4798e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:19:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e085eab37efa0572029c6743b6a19fe133d00615","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-07 13:20:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ae7f844f12d7248250849097d23a3f280461efb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-07 13:20:35.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"58a26e8611ec756f38d31f7d4f11d4efcd5e3da7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:20:51.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"285f985fb7edfedd07573334dfc951ca6ce5d6b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:50:45.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f3738d8479087f320b9c5eadd2acce55acf9f1d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:10:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/808cd8de1d6b4bb29c8db098fa105a69\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fa5eba6166d14155bce24300faac76c1 : SUCCESS in 3m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0be5a1de0f89442ab3633d740018ca32 : SUCCESS in 4m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/5d4f06999f4f41c682bebf932c7e34a7 : SUCCESS in 29m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ad8f12ed5704aa7bad319f5bbf8d168 : SUCCESS in 35m 41s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f994bca2f8204dd0bf949b85d9d4cf0b : SUCCESS in 1h 17m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f56a2dd234474224864eb54c4be7c742 : SUCCESS in 33m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c53001b4809434e8dae947a85c92b7a : SUCCESS in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e953285c004848998c7e6da8b627c8b5 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac3cd98e966244749f0b12fd8c8cf7b8 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37fc3bafbfcb4147a5b26f14547e8cf1 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e35cebd43baa454b9c76c3e1bc8a6cdc : FAILURE in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a074e6a12e409abe1db54559c060d6 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78551e80691848f19c7be8dee15f6c51 : SUCCESS in 5m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/920ed54552df415eb2a04b40ea26c599 : SUCCESS in 6m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e13fc9b5db9b4f56bd629bfdc5aaa7c0 : SUCCESS in 3m 05s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/5e245c1712964e398bdb3d4a23ab9900 : SUCCESS in 9m 22s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a3832bbafd504bafa87db068c05106b7 : SUCCESS in 23m 38s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/ce93361bfed74a0391437faac0b59c71 : FAILURE in 3m 59s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/1122ea28ba714f96a09e3442c7c35bbe : FAILURE in 3m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52f523e2a9cf4faaa6c4a08d0c143c2d : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b231715fa57f10f244fff5a790c7ca8c86b20b97","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-07 18:09:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ba9e3b7199725301498180ffa7c2e2984d9e500c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 18:09:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5fc707d124dd64683d84dad64f6f9663b99f402c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 19:21:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f834f35083a04550930dcd3456bc417a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7557f72e3d5249bf8641e37ab7ca6e60 : SUCCESS in 2m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f64e7275446643c4848cf93986d17dd0 : SUCCESS in 3m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/9dc88e6460c44647a317c399c86236c2 : SUCCESS in 1h 07m 13s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/77b034d85496409ba6a86243f58cdbeb : SUCCESS in 47m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/61e2a8da605246ca82ae0a9add32f578 : SUCCESS in 49m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9297627e9f7c4c7cbcf828e071e62ee1 : SUCCESS in 38m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6c5eae8a1054effaa283969ada906ff : SUCCESS in 6m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e02250a49ff2443f925f01f1f857be6b : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cef346a4eee24935929b5df9085d459c : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e7f702af5e45ca997cec043f850976 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec09bf0b3cc84bfdb23f9538599b2dc6 : SUCCESS in 9m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/255f528ffa0a4be8bcb799810e1920a0 : SUCCESS in 4m 24s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b9d5922fd7c84b5eb8762757a1ec1f37 : SUCCESS in 6m 41s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/930573fa1e4043a5b1932daff3f5d25d : SUCCESS in 4m 57s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1731d8c655de481fb6173bad6f9c419e : SUCCESS in 22m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbacd10ffd184ec990bcf41ec174e5b9 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":2},{"id":"43f594ce6d20f824f2e0c13432625955fe3a25c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 19:21:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8f2c7e83e7aa6182220e1638db5dbbb31ca83b7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 19:23:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dbf7937bb654c70ab37c29636aa491f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd9ce368285460ab751b78b5ab05339 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"01ac96d93374a8e2e8c9a1b650fb009085caeb39","revisions":{"ea6a17b805d00f83a63fda0176970c718c45381d":{"kind":"REWORK","_number":1,"created":"2026-09-07 13:17:58.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/18/1004418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/18/1004418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/1004418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/1004418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/1004418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/18/1004418/1"}}},"commit":{"parents":[{"commit":"7d35f974d52fdbb457ff2b6b7bff3741165dd30d","subject":"bump min version of os-traits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7d35f974d52fdbb457ff2b6b7bff3741165dd30d"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-09-07 13:15:10.000000000","tz":-240},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-09-07 13:17:42.000000000","tz":-240},"subject":"[pyproject] Update setuptools config","message":"[pyproject] Update setuptools config\n\nUpdate the setuptools config to find all submodules that contain\n\"placement\", instead of just finding the top-level module.\n\nChange-Id: Icc6b40abcc455952694004585f053886eeb81005\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ea6a17b805d00f83a63fda0176970c718c45381d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ea6a17b805d00f83a63fda0176970c718c45381d"}]},"branch":"refs/heads/master"},"01ac96d93374a8e2e8c9a1b650fb009085caeb39":{"kind":"NO_CHANGE","_number":2,"created":"2026-09-07 13:50:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/1004418/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/18/1004418/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/1004418/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/1004418/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/18/1004418/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/18/1004418/2"}}},"commit":{"parents":[{"commit":"7d35f974d52fdbb457ff2b6b7bff3741165dd30d","subject":"bump min version of os-traits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7d35f974d52fdbb457ff2b6b7bff3741165dd30d"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2026-09-07 13:15:10.000000000","tz":-240},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:15:10.000000000","tz":-240},"subject":"[pyproject] Update setuptools config","message":"[pyproject] Update setuptools config\n\nUpdate the setuptools config to find all submodules that contain\n\"placement\", instead of just finding the top-level module.\n\nChange-Id: Icc6b40abcc455952694004585f053886eeb81005\nSigned-off-by: Emma Foley \u003cefoley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/01ac96d93374a8e2e8c9a1b650fb009085caeb39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/01ac96d93374a8e2e8c9a1b650fb009085caeb39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
