)]}'
{"id":"openstack%2Fplacement~1004422","triplet_id":"openstack%2Fplacement~master~I0e671da2ec15e6ef9f09ee11906728c452ba037f","project":"openstack/placement","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 14:32:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0e671da2ec15e6ef9f09ee11906728c452ba037f","subject":"Replace license classifier","status":"NEW","created":"2026-09-07 13:49:50.000000000","updated":"2026-09-11 04:38:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b758e65f0c9bd77e6f948b0c520fcd90fba63e9","_number":1004422,"virtual_id_number":1004422,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 14:53:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-09-11 04:38:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2026-09-09 13:15:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-09-09 13:15:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 14:32:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-09-07 15:21:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 17:00:08.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-09-11 04:38:10.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"}],"messages":[{"id":"16ea204115a038002ac037d75b62e04d98a133e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:49:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c1d6c2cf5464e7891a46bc08ca280261fc0f338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:50:45.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b6525dd2ed40f8df19fa78e491799d5b803e5fb9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-07 14:32:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0eb688b97b7fb1495c38da8816189167ed95b8bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:21:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f636a82365dc42639829249e424557ff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ad47c372f2e243bcaa12d97032fa79a6 : SUCCESS in 2m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd7749b231d64004ae7f7b154871d881 : SUCCESS in 3m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/f6e0e2fa4fb74c1a89b4a0a7bb8d52bc : SUCCESS in 1h 15m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/511511b135ea404497fd68094a99d75b : SUCCESS in 47m 59s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/a382bfd1803e40d1867c0847b66c5156 : SUCCESS in 1h 24m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd923981b6ae4bef92c7aca970988a19 : SUCCESS in 1h 01m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7cd0390ccb0445794605828672f773c : SUCCESS in 15m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5b7c78badc641309d5c0fcc599671b7 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d014a60c9014743adb957c70b1731ea : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/282e00623b10415b8b96feef53d737c6 : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/01ac59347cf24cd5848a9fc3278d2808 : FAILURE in 8m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59d9157df654cdea492a480eebdfc4d : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e18d9613c9247fcadd5945ac89968a4 : SUCCESS in 9m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1bc360fafdba47c9aefa54c38266d820 : SUCCESS in 5m 07s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fb6f8d2a776c4178ab40e3183e078f41 : SUCCESS in 6m 11s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/2acbeb782e394d799904661ee235057c : SUCCESS in 9m 00s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5f3fb6f2ff5d4cd29e3a6b0a4f2d3d2f : SUCCESS in 23m 22s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/1c538d874b1044bb8b70ab9198207f0b : FAILURE in 4m 24s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/a070f01e55d44e339b755e8ff9c9a63d : FAILURE in 3m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6978c91779324a8db7f78dfcfc3eb7bf : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":2},{"id":"618d8d244eb629c3356cbf306d06b8dd174e68ad","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-08 17:00:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"576bcb161190928b3a398efd0334e440b2ffa265","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-09 13:15:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"e053e3763c861dd1d0428a1ab2228c25198b236d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 14:53:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2ddae7466244b11b6ab69983e9f3203\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ee6d2dc001d841cc9af6bce79859463a : SUCCESS in 2m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c0e6ea08d85a4d57b93e794ae5a0c538 : SUCCESS in 2m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/e0d3bf359e284e11bd4ca22a81768246 : SUCCESS in 1h 05m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83a9ee928c8542f1b8fe92fbc358d58c : SUCCESS in 1h 06m 59s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/5deb0d6fd67440b1ad3305d3cb07d371 : SUCCESS in 1h 30m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8709cc3fa9c14f27b69677b249eb3f7c : SUCCESS in 1h 07m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/199a084397fa496c9675821bbc3f745b : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdbd0946c4145628bf5d6f2f6e0e5e8 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da61eee7a298458fa13f503a9518c682 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d7aa9779c0494f5f9f1653851b94cd07 : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff882024f2d44672b9dae87f4ba5b237 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df405c53b85748ca84e8ac8952b80c9f : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44cd0a951c414ea593b9625ac51bdabf : SUCCESS in 11m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6f0933c82c014975b62b7eccc866ba34 : SUCCESS in 5m 08s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cb31adf2e064428a865d07daad15ffe0 : SUCCESS in 5m 24s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/15a623b6dc84494e8671e8cae753f5ee : SUCCESS in 7m 59s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9a9e07c5b2c7429eb2e9e6b56262594b : SUCCESS in 22m 10s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/857b02f7be874d289f22f126c2bf1341 : FAILURE in 3m 21s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/52254ca8d22142bf8549dc8782affd6a : FAILURE in 4m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d471c6eee834efeb72405974722d78c : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":3},{"id":"8b758e65f0c9bd77e6f948b0c520fcd90fba63e9","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2026-09-11 04:38:10.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"64b27b0b0697a9a21689625fe205ce25f6aed43e","revisions":{"af1495d07d81de22cb0f9a663d8f97cff86b81dc":{"kind":"REWORK","_number":1,"created":"2026-09-07 13:49:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/22/1004422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/22/1004422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/22/1004422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/22/1004422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/22/1004422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/22/1004422/1"}}},"commit":{"parents":[{"commit":"ea6a17b805d00f83a63fda0176970c718c45381d","subject":"[pyproject] Update setuptools config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ea6a17b805d00f83a63fda0176970c718c45381d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:34:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:42:15.000000000","tz":60},"subject":"Replace license classifier","message":"Replace license classifier\n\nResolve some ugly warnings from setuptools.\n\nChange-Id: I0e671da2ec15e6ef9f09ee11906728c452ba037f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/af1495d07d81de22cb0f9a663d8f97cff86b81dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/af1495d07d81de22cb0f9a663d8f97cff86b81dc"}]},"branch":"refs/heads/master"},"d87a53bf91d48219803e7b512b4715274b4bd342":{"kind":"NO_CHANGE","_number":2,"created":"2026-09-07 13:50:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/22/1004422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/22/1004422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/22/1004422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/22/1004422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/22/1004422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/22/1004422/2"}}},"commit":{"parents":[{"commit":"01ac96d93374a8e2e8c9a1b650fb009085caeb39","subject":"[pyproject] Update setuptools config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/01ac96d93374a8e2e8c9a1b650fb009085caeb39"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:34:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:34:31.000000000","tz":60},"subject":"Replace license classifier","message":"Replace license classifier\n\nResolve some ugly warnings from setuptools.\n\nChange-Id: I0e671da2ec15e6ef9f09ee11906728c452ba037f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d87a53bf91d48219803e7b512b4715274b4bd342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d87a53bf91d48219803e7b512b4715274b4bd342"}]},"branch":"refs/heads/master"},"64b27b0b0697a9a21689625fe205ce25f6aed43e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-09 13:15:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/22/1004422/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/22/1004422/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/22/1004422/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/22/1004422/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/22/1004422/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/22/1004422/3"}}},"commit":{"parents":[{"commit":"109ea4af9561aba4e67df8e6075055b764993489","subject":"tox: Use constraints option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/109ea4af9561aba4e67df8e6075055b764993489"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:34:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-09 13:13:44.000000000","tz":60},"subject":"Replace license classifier","message":"Replace license classifier\n\nResolve some ugly warnings from setuptools.\n\nChange-Id: I0e671da2ec15e6ef9f09ee11906728c452ba037f\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/64b27b0b0697a9a21689625fe205ce25f6aed43e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/64b27b0b0697a9a21689625fe205ce25f6aed43e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
