)]}'
{"id":"openstack%2Fplacement~1004423","triplet_id":"openstack%2Fplacement~master~I841bf071aa2fecb0df7d5dfbf7a9f0974bc321b1","project":"openstack/placement","branch":"master","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-09 13:17:55.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-08 23:05:23.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I841bf071aa2fecb0df7d5dfbf7a9f0974bc321b1","subject":"Migrate requirements to pyproject.toml","status":"NEW","created":"2026-09-07 13:49:50.000000000","updated":"2026-09-09 14:18:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"56f14b77c322e9405e36887ae2026c67fc83dd1c","_number":1004423,"virtual_id_number":1004423,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-09 14:18:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 14:52:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-08 17:01:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"8706b08f0d7daa18960508601676bf7123073e2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:49:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d5769cded67803acded46b7f36b89245bebd611","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 13:50:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c996463567c4a8ac1c3098bfd3100d002b885b49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:52:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c56623e7cd21488faa9deb1ab51c0db2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8498538d16664895a545d54d51e288cb : FAILURE in 3m 16s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/711af07e6a2e4053875d084dd5ca9669 : FAILURE in 2m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a5bc9e066c804b66a0c5bfd49a109b1b : FAILURE in 2m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/5527342d1f064bd88c2e675bdffe0425 : FAILURE in 54m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f338e708ad2545048971d795631d8c3f : FAILURE in 39m 25s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/3e629d6ccce8422bbe9dc5963dc2cbda : FAILURE in 9m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c38e5fa98684daea1fb1d2727e4a543 : POST_FAILURE in 9m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4743b7b68999473e8c02d8adfad3cce3 : FAILURE in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b443cbaffb4a2da537364ad17eaa82 : FAILURE in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bafab7823f974fcd85970fb49c7abe8b : FAILURE in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f2730dd5cd8471d97f863fec00cb0de : FAILURE in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5f0f8b2434324ad087d3e331bfeb64d0 : FAILURE in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2297613f65d94900a72f927cee5cd2fd : FAILURE in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8963a7efc2e24cd7905cdc42cc009eeb : FAILURE in 3m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0345fd8f5e754d63829d3849ac8c0d8d : FAILURE in 3m 33s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/936144d24b00488e8d584593c8c40f6d : FAILURE in 3m 18s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/d3d898545aca430ebb10cc9bf3546111 : FAILURE in 6m 11s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/774b3b952596407992b9f1b35e987ff4 : FAILURE in 5m 02s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/e61e1fb47cf54e929023bb851a75b3d4 : FAILURE in 1m 55s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/f411f64bd2a24c56b2b3e4cb7d8cda7e : FAILURE in 2m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b6cc3b16c814b548d1f2c2241ae4687 : FAILURE in 18m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b312d37a98cf32d220f547cfbe1c23425855a608","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-08 17:01:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0474e340799cb9f4093e0e9ecb41be7321d26273","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-08 23:05:23.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"0fbd90e889c10b467c9126faec9715b2cb4053cb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-09 13:15:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5fe7c0ade24cb5a31299b6b0d99f406ae2a63917","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-09 13:17:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ffb602f560d20fa8e2745838729f6a08e9f6dfe2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-09 13:18:03.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"56f14b77c322e9405e36887ae2026c67fc83dd1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 14:18:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97091038649e4468935c7ce72ad04f43\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f40df8fa30084e8c8b115b03887f7133 : SUCCESS in 4m 32s\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1ffefa0755714dce8a30dd579f35456e : SUCCESS in 3m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/979cec085ee34d61a3e5070bb0ea764d : SUCCESS in 2m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/4a5747934fe345b5bda8a6cf01851229 : SUCCESS in 53m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8d7142048014482bd4df95e9c266509 : SUCCESS in 31m 43s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/12f4f0414b0845b38a131faf00985623 : SUCCESS in 59m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd9a403d66dd44aab79f0bb7be833a0c : SUCCESS in 29m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dd5418e24df430296dd505cbba3dfae : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0987d8e279c491fa11ccd6f1cafcaab : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a8f0a4e45bf4003980fdac427bcb577 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc7a6aa48f4745a896693a4f289f30ac : SUCCESS in 3m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0294a41853eb45648aa9826dcbe33b7b : FAILURE in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/869a9f75fe4f44d18e3a2cde2559c9e1 : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7acc9f376db42beb41b80fd2c5db159 : SUCCESS in 4m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/25d2648e1c754f728b4c90b2597285c2 : SUCCESS in 5m 37s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/29bd34206a3742238ae769df4a857b9f : SUCCESS in 3m 30s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/282180724bcd4fc3bb97c673614206fd : SUCCESS in 7m 59s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/027bcbdecb53423893334b4cba429a17 : SUCCESS in 22m 31s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/707c57ef8579451fa56007c4cca03540 : FAILURE in 3m 18s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/a633b551ca184efaa63a65e7918f57c8 : FAILURE in 3m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75d91f68ba714e989b3c130183f0d48d : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f7dc620c4a0c0b5125676dd2d21cedd8fdd6e92f","revisions":{"b3301a6c3d847b5ba1bb2e6cb3f2ace017a0bf1f":{"kind":"REWORK","_number":1,"created":"2026-09-07 13:49:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/1004423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/23/1004423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/1004423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/1004423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/1004423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/23/1004423/1"}}},"commit":{"parents":[{"commit":"51c5874d13e3c7b1d7b6733d5994b6e949ee5d26","subject":"tox: Use constraints option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/51c5874d13e3c7b1d7b6733d5994b6e949ee5d26"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:41:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:47:41.000000000","tz":60},"subject":"Migrate requirements to pyproject.toml","message":"Migrate requirements to pyproject.toml\n\nWhile here, we also fix some indentation and ordering in tox.ini and add\nhelp strings for some sections.\n\nChange-Id: I841bf071aa2fecb0df7d5dfbf7a9f0974bc321b1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b3301a6c3d847b5ba1bb2e6cb3f2ace017a0bf1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b3301a6c3d847b5ba1bb2e6cb3f2ace017a0bf1f"}]},"branch":"refs/heads/master"},"a47c728337f14b4906db43220975957e7e9b450c":{"kind":"REWORK","_number":2,"created":"2026-09-07 13:50:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/1004423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/23/1004423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/1004423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/1004423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/1004423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/23/1004423/2"}}},"commit":{"parents":[{"commit":"da17c2795d4e70438ca4c3a899cdf434ebddc54d","subject":"tox: Use constraints option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/da17c2795d4e70438ca4c3a899cdf434ebddc54d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:41:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:50:37.000000000","tz":60},"subject":"Migrate requirements to pyproject.toml","message":"Migrate requirements to pyproject.toml\n\nWhile here, we also fix some indentation and ordering in tox.ini and add\nhelp strings for some sections.\n\nChange-Id: I841bf071aa2fecb0df7d5dfbf7a9f0974bc321b1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a47c728337f14b4906db43220975957e7e9b450c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a47c728337f14b4906db43220975957e7e9b450c"}]},"branch":"refs/heads/master"},"f7dc620c4a0c0b5125676dd2d21cedd8fdd6e92f":{"kind":"REWORK","_number":3,"created":"2026-09-09 13:15:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/1004423/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/23/1004423/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/1004423/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/1004423/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/23/1004423/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/23/1004423/3"}}},"commit":{"parents":[{"commit":"64b27b0b0697a9a21689625fe205ce25f6aed43e","subject":"Replace license classifier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/64b27b0b0697a9a21689625fe205ce25f6aed43e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 13:41:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-09 13:15:32.000000000","tz":60},"subject":"Migrate requirements to pyproject.toml","message":"Migrate requirements to pyproject.toml\n\nWhile here, we also fix some indentation and ordering in tox.ini and add\nhelp strings for some sections.\n\nNote that we retain doc/requirements.txt for now, pending a fix for the\nreleasenotes job [1].\n\n[1] https://review.opendev.org/c/zuul/zuul-jobs/+/1002494\n\nChange-Id: I841bf071aa2fecb0df7d5dfbf7a9f0974bc321b1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f7dc620c4a0c0b5125676dd2d21cedd8fdd6e92f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f7dc620c4a0c0b5125676dd2d21cedd8fdd6e92f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
