)]}'
{"id":"openstack%2Fplacement~1005883","triplet_id":"openstack%2Fplacement~master~Id899d47411f4ac918736daffaf004abecc16f8ed","project":"openstack/placement","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-16 16:43:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id899d47411f4ac918736daffaf004abecc16f8ed","subject":"Fix tests on Python 3.14","status":"MERGED","created":"2026-09-16 13:46:55.000000000","updated":"2026-09-16 16:44:10.000000000","submitted":"2026-09-16 16:43:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1005883","meta_rev_id":"b6b71a82413ec934816c6047c375744231079d80","_number":1005883,"virtual_id_number":1005883,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 16:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-09-16 15:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-09-16 13:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-09-16 15:18:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 13:49:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-09-16 15:18:53.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-09-16 15:34:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a16f00cd53e90a14af3e52503f45b266206ccdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 13:46:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48bfca92732551280592e0c6a5b8e3c534855221","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-16 13:49:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f864f28a4c39914efacb05199048f53f99459ea8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 13:50:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f4f283303659b50af61efdab5bef4cb72699b20","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-16 15:18:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"166f9ae4b999121df103b8d1c04d743523cd5fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:34:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23d1bf9366464413b37838f0dbf64847\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d1cb2046d5348a8b5f0bd28f7169724 : SUCCESS in 1h 00m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89762fb15c2b43e4bc0393e50c7b3369 : SUCCESS in 53m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0209be45bdb4aaaa14504061f22c246 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95c255f20a7c4e1ea9c920f5990e26d2 : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1eddd7e9ac444e8894993c29273f6290 : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6abe7555d9a74e00a5a4dcb57acd4a23 : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c45601ba0c141d6a2b66b2afceefdff : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adea5cf4e6c740f1beaa809cf2111d66 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2210253dd23249d8975815b1a9b22d14 : SUCCESS in 8m 39s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3dd92a612baf4182ab3b04f6d596da7a : SUCCESS in 6m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2463a0f49a75459fa2f109ceb60e34bc : SUCCESS in 6m 01s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/75144e8e1bf24a53af7099d317b3d668 : SUCCESS in 7m 14s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8fbe41819e1d4d2ba2f4caf35dc3828f : SUCCESS in 23m 31s","accounts_in_message":[],"_revision_number":1},{"id":"18a51d1e9f57b822c9813fe0d278dc5f8615b37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 15:35:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fc00ce25f0aca57c1c830813fb577cb8920571f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:43:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77d2c5eba61f4da08ce0d219e3ae658d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/540c0c154142423ba3b33f50eece4650 : SUCCESS in 1h 05m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f50c372d23ae4910b8c323b5bb7ae4f1 : SUCCESS in 1h 03m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a52e6467f9a49efbb3e6089f5639e8a : SUCCESS in 7m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d1a22e26b4249f6bb604307648ae06f : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63c0d3bdb8aa4592aa46fa8c5a1d0652 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ddd7191cdec4fa391c63282130b0924 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7fedc0c8dea4d86b345608b5ed13d61 : SUCCESS in 7m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3b7c8a7f50fb436d99acf96c1c999c68 : SUCCESS in 6m 50s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/628713adc46447e984e502de53e0ec87 : SUCCESS in 5m 33s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/f40c3f789d1a413bba2d03c76a3405f2 : SUCCESS in 10m 18s (non-voting)\n- placement-nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9ff2654e385a42e28bd5de81092c32fa : SUCCESS in 22m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7c3a5325dcae700c9f7cf57d2ba4b5ed4e8a291f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:43:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b6b71a82413ec934816c6047c375744231079d80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:44:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f916f52d32314016afb336ded29b5cb1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff579fa6716444f89ebf6fe79cd0af4 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d3724ea24bf3e1950e638d5419460245d621523","revisions":{"3d3724ea24bf3e1950e638d5419460245d621523":{"kind":"REWORK","_number":1,"created":"2026-09-16 13:46:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/1005883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/83/1005883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/1005883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/1005883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/1005883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/83/1005883/1"}}},"commit":{"parents":[{"commit":"88a3de9c4651d425d04f7c491faa1a0f658a523b","subject":"Merge \"Commonize code for updating and creating allocations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/88a3de9c4651d425d04f7c491faa1a0f658a523b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 13:46:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-16 13:46:48.000000000","tz":60},"subject":"Fix tests on Python 3.14","message":"Fix tests on Python 3.14\n\nChange-Id: Id899d47411f4ac918736daffaf004abecc16f8ed\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3d3724ea24bf3e1950e638d5419460245d621523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3d3724ea24bf3e1950e638d5419460245d621523"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
