)]}'
{"id":"openstack%2Fplacement~1007283","triplet_id":"openstack%2Fplacement~master~Id7ee1bd778d33c8795935664771473157f28b7c5","project":"openstack/placement","branch":"master","topic":"py314","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-29 03:17:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7ee1bd778d33c8795935664771473157f28b7c5","subject":"Move upper python version job","status":"MERGED","created":"2026-09-24 17:41:31.000000000","updated":"2026-09-29 03:20:37.000000000","submitted":"2026-09-29 03:17:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007283-py314","meta_rev_id":"80be4aec1f7bc5bbfd16c076084bf0fcecc7d77a","_number":1007283,"virtual_id_number":1007283,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-29 03:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-28 16:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2026-09-28 16:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2026-09-29 01:40:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-24 17:41:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-28 16:35:45.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-09-28 16:36:19.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"05f4fe4380ad9e5078ca4cce07da78ce68df86f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 17:41:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12e85781cf808af7dec091d95f47733273a500cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 17:41:54.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/placement on branch master.  The\nproblem was:\n\n  Job nova-tox-functional-py314 not defined\n\nThe problem appears in the \"placement-nova-tox-functional-py314\" job stanza:\n\n  job:\n      name: placement-nova-tox-functional-py314\n      parent: nova-tox-functional-py314\n      description: |\n          Run the nova functional tests to confirm that we aren\u0027t breaking\n          the PlacementFixture.\n      vars:\n          # \u0027functional-without-sample-db-tests\u0027 tox env is defined in nova tox.ini\n          # to skip the api|notification _sample_tests and db-related tests.\n  ...\n\n  in \"openstack/placement/.zuul.yaml@master\", line 78","accounts_in_message":[],"_revision_number":1},{"id":"488086a57800e3f5a7782896d29bc6a93cb353c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 18:05:16.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"89c7efc9f08836976ba2449e0c022d30b9ce32c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-24 18:05:21.000000000","message":"Topic set to py314","accounts_in_message":[],"_revision_number":2},{"id":"b9470244dba14ac44ad7ea44abb060a98150fa05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 18:13:24.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/1007299 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"e5de6c38b82e090fe25ecb21da70cd8116e980ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-28 08:36:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c1f0c312270f278fe89efdfa3c3025cd6760ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 09:42:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a30f49ea18974411a1336a8d4c903a0d\n\n- grenade https://zuul.opendev.org/t/openstack/build/a47d944eae27402fb07ad516db3a4419 : SUCCESS in 46m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc4ebb7601ea4dfaa39df7944131288d : SUCCESS in 1h 03m 15s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/3e7f767239774a5999b530d162f10d58 : SUCCESS in 54m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26997b8ca54b4ad8b5003d8eb471f494 : SUCCESS in 29m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b196516f3ff406a883f926f17d786d1 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5673604e77468ea38be5729cdb4136 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ff6020fe45b4daa94a23dea80117328 : SUCCESS in 4m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9ad1c4ae8ad14c25bec2ab37bec35f39 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a77f32142b74d3dada190d419d777da : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5148f8f567d4926b221bf94409ee1af : SUCCESS in 10m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0b03a0666b56470abd196e0cd9c2f5dc : SUCCESS in 5m 12s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c38a18dd6f1249e9a8a592e56c87c28c : SUCCESS in 4m 34s\n- placement-nova-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/c08d774314d942f58acab8ce7c1fdf6e : SUCCESS in 25m 56s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/28b1a16b173e4ec89e16ff872b7bca30 : FAILURE in 7m 14s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/8a035676f309412e8e9473a733ce998f : FAILURE in 3m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6212d002ec434d30aea298c08a8962d3 : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":2},{"id":"778c2417fb35d5c0cc0744f7ca088e189c73d6a3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-28 16:35:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"59f0981d59f4821dbbe404542e25d789f8d3eb19","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-28 16:36:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6e3218699d14a5c4eaa478fe87bafbab49229fe9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-29 01:40:12.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6766fdb84ff07f994fa44f447de9e44ba977418d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 01:41:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3f67fffacc6cf20225b92e137e07be45b096d52b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 03:17:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99fae126f3fd47aca81159207878b76c\n\n- grenade https://zuul.opendev.org/t/openstack/build/4613fbf5ddbc46328c203fc1ac3a2a15 : SUCCESS in 46m 20s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/7f32f961d3a34df58661b3510c2cb4d0 : SUCCESS in 1h 31m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c62ed5aaf6044adab3f05f510e205a4f : SUCCESS in 1h 03m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/965f620c883d422ead46611ff5ac083d : SUCCESS in 35m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77179b8068e64341a69ff53832098445 : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/416bb94118ff4e00a3888f107ebaa0fc : SUCCESS in 2m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96f2b0357ac34690baa745227a4d5caf : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510171825add4c6ba11b75fabd300830 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b56c3c1f38504a0487a0caa61f1069f1 : SUCCESS in 10m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d43b6f050ca044738e99d1ee0b8a6299 : SUCCESS in 5m 54s\n- openstack-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/029af065d73949d6b65fe0a7bd6bd593 : SUCCESS in 9m 37s\n- placement-nova-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/6313f827071144dd891102fe85c3c3fa : SUCCESS in 24m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74c48b4808d04f52b6db01fcf8889df8 : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":2},{"id":"343beeb4f524e7dd98bad9678e5b9ad59efd7d76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 03:17:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"80be4aec1f7bc5bbfd16c076084bf0fcecc7d77a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 03:20:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78f688d2e11b4f579938d36bf5de0f80\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cfd62385e1b48c7b0d0dd7ccbd531a3 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0609658b480d3934a0d5aaf8620e2d31801661ea","revisions":{"623a3b812ae60324c53d65d8936e5596791be342":{"kind":"REWORK","_number":1,"created":"2026-09-24 17:41:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/1007283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/83/1007283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/1007283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/1007283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/1007283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/83/1007283/1"}}},"commit":{"parents":[{"commit":"d3821e93fbb9ac4bd4093df9527c48ad1af068dc","subject":"Merge \"pre-commit: Add codespell\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d3821e93fbb9ac4bd4093df9527c48ad1af068dc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 17:41:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 17:41:07.000000000","tz":540},"subject":"Move upper python version job","message":"Move upper python version job\n\nPython 3.14 is now part of the tested runtimes.\n\nChange-Id: Id7ee1bd778d33c8795935664771473157f28b7c5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/623a3b812ae60324c53d65d8936e5596791be342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/623a3b812ae60324c53d65d8936e5596791be342"}]},"branch":"refs/heads/master"},"0609658b480d3934a0d5aaf8620e2d31801661ea":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-24 18:05:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/1007283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/83/1007283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/1007283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/1007283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/1007283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/83/1007283/2"}}},"commit":{"parents":[{"commit":"d3821e93fbb9ac4bd4093df9527c48ad1af068dc","subject":"Merge \"pre-commit: Add codespell\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d3821e93fbb9ac4bd4093df9527c48ad1af068dc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 17:41:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-24 18:05:16.000000000","tz":0},"subject":"Move upper python version job","message":"Move upper python version job\n\nPython 3.14 is now part of the tested runtimes.\n\nDepends-on: https://review.opendev.org/c/openstack/nova/+/1007299\nChange-Id: Id7ee1bd778d33c8795935664771473157f28b7c5\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0609658b480d3934a0d5aaf8620e2d31801661ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0609658b480d3934a0d5aaf8620e2d31801661ea"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
