)]}'
{"id":"openstack%2Fplacement~637892","triplet_id":"openstack%2Fplacement~master~Ib78c326d79b1ad2393c42702f9081a7593097e3e","project":"openstack/placement","branch":"master","topic":"bug/1813147","hashtags":[],"change_id":"Ib78c326d79b1ad2393c42702f9081a7593097e3e","subject":"Downgrade os-traits/os-resource-classes sync log level to DEBUG","status":"MERGED","created":"2019-02-19 15:32:37.000000000","updated":"2019-02-19 22:04:11.000000000","submitted":"2019-02-19 22:04:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"637892-1550613851592-44fb25e9","meta_rev_id":"402a2ace54ce9ee3d55c920242a2ea2824b82fcf","_number":637892,"virtual_id_number":637892,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-02-19 21:59:36.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-02-19 22:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-02-19 20:24:54.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-02-19 15:38:12.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-02-19 20:24:54.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-19 15:38:12.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-02-19 20:24:54.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-02-19 22:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e075e6bba7e79c8dfd1bcfabd6a592a92103bcc3","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-02-19 15:32:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bc451bd02af8d61e4f200544391bf9a556f37ba","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-02-19 15:38:12.000000000","message":"Patch Set 1: Code-Review+2\n\nyeah, seems reasonable","accounts_in_message":[],"_revision_number":1},{"id":"684ea6cc07258828ac7440e87167bff3a417e7fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 18:37:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/92/637892/1/check/grenade-py3/fd83e26/ : SUCCESS in 1h 04m 51s\n- tempest-full-py3 http://logs.openstack.org/92/637892/1/check/tempest-full-py3/f5a0259/ : SUCCESS in 1h 32m 57s\n- openstack-tox-docs http://logs.openstack.org/92/637892/1/check/openstack-tox-docs/5f15110/html/ : SUCCESS in 5m 06s\n- openstack-tox-cover http://logs.openstack.org/92/637892/1/check/openstack-tox-cover/bbc77cd/cover/ : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/637892/1/check/openstack-tox-lower-constraints/f8592f2/ : SUCCESS in 5m 42s\n- openstack-tox-pep8 http://logs.openstack.org/92/637892/1/check/openstack-tox-pep8/b60f471/ : SUCCESS in 6m 21s\n- openstack-tox-py27 http://logs.openstack.org/92/637892/1/check/openstack-tox-py27/e72787b/ : SUCCESS in 6m 18s\n- openstack-tox-py35 http://logs.openstack.org/92/637892/1/check/openstack-tox-py35/2a284f6/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/92/637892/1/check/openstack-tox-py36/0732282/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/92/637892/1/check/openstack-tox-py37/bdc2e4d/ : SUCCESS in 5m 48s\n- openstack-tox-functional http://logs.openstack.org/92/637892/1/check/openstack-tox-functional/fd0b91c/ : SUCCESS in 6m 13s\n- openstack-tox-functional-py35 http://logs.openstack.org/92/637892/1/check/openstack-tox-functional-py35/bd8b881/ : SUCCESS in 5m 27s\n- openstack-tox-functional-py36 http://logs.openstack.org/92/637892/1/check/openstack-tox-functional-py36/28e499a/ : SUCCESS in 5m 24s\n- placement-perfload http://logs.openstack.org/92/637892/1/check/placement-perfload/3df2350/ : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15d71b0647aafbf23435a165444339e14ddc9248","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-19 20:24:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI concur. Should we do this in nova too?","accounts_in_message":[],"_revision_number":1},{"id":"3082d32ad9daedab0c986d3d81eb10636f5dc4c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 20:25:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"076ada9de2fd96f1b2158248b49ab96acc2ba5b8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-02-19 21:59:36.000000000","message":"Patch Set 1:\n\n\u003e Should we do this in nova too?\n\nWe can...\n\nMy main objective is related to the noise this puts in the nova functional job logs which contributes to tripping the subunit parser error and nova functional tests are using the placement fixture from the placement repo, so we don\u0027t *have* to mirror this in nova but it wouldn\u0027t hurt either.","accounts_in_message":[],"_revision_number":1},{"id":"eb3b730889acb6d07a0cef292c378ffa130c33de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 22:04:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/92/637892/1/gate/grenade-py3/e7adc8e/ : SUCCESS in 1h 07m 46s\n- tempest-full-py3 http://logs.openstack.org/92/637892/1/gate/tempest-full-py3/37d7296/ : SUCCESS in 1h 32m 17s\n- openstack-tox-docs http://logs.openstack.org/92/637892/1/gate/openstack-tox-docs/5b3d10e/html/ : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/92/637892/1/gate/openstack-tox-lower-constraints/861a876/ : SUCCESS in 6m 19s\n- openstack-tox-pep8 http://logs.openstack.org/92/637892/1/gate/openstack-tox-pep8/938d73f/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/92/637892/1/gate/openstack-tox-py27/8fb7f5b/ : SUCCESS in 5m 08s\n- openstack-tox-py35 http://logs.openstack.org/92/637892/1/gate/openstack-tox-py35/8dc34a8/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/92/637892/1/gate/openstack-tox-py36/800e0b7/ : SUCCESS in 5m 40s\n- openstack-tox-py37 http://logs.openstack.org/92/637892/1/gate/openstack-tox-py37/fcc7e1e/ : SUCCESS in 5m 12s\n- openstack-tox-functional http://logs.openstack.org/92/637892/1/gate/openstack-tox-functional/12dd424/ : SUCCESS in 6m 24s\n- openstack-tox-functional-py35 http://logs.openstack.org/92/637892/1/gate/openstack-tox-functional-py35/d44555e/ : SUCCESS in 5m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/92/637892/1/gate/openstack-tox-functional-py36/4b73da7/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9b1bdc3c3356ea434ab6143d52770f99dc7094c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 22:04:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3d0ffb743ba06bfc3611cf0d9dda07abb921d87","revisions":{"a3d0ffb743ba06bfc3611cf0d9dda07abb921d87":{"kind":"REWORK","_number":1,"created":"2019-02-19 15:32:37.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/92/637892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/92/637892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/92/637892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/92/637892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/92/637892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/92/637892/1"}}},"commit":{"parents":[{"commit":"530be3d3985f02e2eb7d45e6f2977b294eb9cc65","subject":"Trivial: return empty set instead of list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/530be3d3985f02e2eb7d45e6f2977b294eb9cc65"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-19 15:31:04.000000000","tz":-300},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-02-19 15:31:04.000000000","tz":-300},"subject":"Downgrade os-traits/os-resource-classes sync log level to DEBUG","message":"Downgrade os-traits/os-resource-classes sync log level to DEBUG\n\nThe current count of traits from the os-traits library is 120 so\nthe INFO log message to dump the traits that get synchronized into\nthe traits table on startup of the service is quite large.\n\nThis drops that logging from INFO to DEBUG and does the same for\nos-resource-classes which is admittedly smaller than os-traits but\nstill debug information more than informational.\n\nChange-Id: Ib78c326d79b1ad2393c42702f9081a7593097e3e\nRelated-Bug: #1813147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a3d0ffb743ba06bfc3611cf0d9dda07abb921d87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a3d0ffb743ba06bfc3611cf0d9dda07abb921d87"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"530be3d3985f02e2eb7d45e6f2977b294eb9cc65","is_merged_in_target_branch":true,"change_id":"Id463f0955231f1dfa3a363a738f9428078189a4a","change_number":636846,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
