)]}'
{"id":"openstack%2Fplacement~640909","triplet_id":"openstack%2Fplacement~master~Ia863977a154300ccf8da321c1828f83c951c9a19","project":"openstack/placement","branch":"master","topic":"trivial","hashtags":[],"change_id":"Ia863977a154300ccf8da321c1828f83c951c9a19","subject":"FUPs for improve-debug-log series","status":"MERGED","created":"2019-03-05 01:05:39.000000000","updated":"2019-03-05 17:50:19.000000000","submitted":"2019-03-05 17:50:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"640909-1551808219880-4f1731f4","meta_rev_id":"95655f2806e7c72e525b60eaa509c7a835e17d3c","_number":640909,"virtual_id_number":640909,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-03-05 01:08:19.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-03-05 17:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-03-05 01:59:42.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-03-05 15:41:17.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":1,"date":"2019-03-05 15:41:17.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-05 01:59:42.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-03-05 15:41:17.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-03-05 17:50:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ed91ea488417d9bcbe733942f6494b46394e4c2","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-05 01:05:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dfe9519326be98c4a06b62ae31067985bb6358b","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-03-05 01:08:19.000000000","message":"Patch Set 1:\n\nnote: renamed as follows\n    rp1    -\u003e ss1\n    rp[N] -\u003e rp[N-1] (for N \u003e\u003d2)","accounts_in_message":[],"_revision_number":1},{"id":"adcd30f03d23e33acb008836284967468627e745","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-05 01:59:42.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks Tetsuro.","accounts_in_message":[],"_revision_number":1},{"id":"49a8d5933d3395110199e7c052b0ddd87d84bc43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 02:19:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/09/640909/1/check/grenade-py3/17acf1e/ : SUCCESS in 1h 02m 54s\n- tempest-full-py3 http://logs.openstack.org/09/640909/1/check/tempest-full-py3/e21fed9/ : SUCCESS in 1h 12m 49s\n- openstack-tox-docs http://logs.openstack.org/09/640909/1/check/openstack-tox-docs/01e66b6/html/ : SUCCESS in 4m 25s\n- openstack-tox-cover http://logs.openstack.org/09/640909/1/check/openstack-tox-cover/f23326b/cover/ : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/640909/1/check/openstack-tox-lower-constraints/d5fe129/ : SUCCESS in 7m 07s\n- openstack-tox-pep8 http://logs.openstack.org/09/640909/1/check/openstack-tox-pep8/a62a251/ : SUCCESS in 6m 02s\n- openstack-tox-py27 http://logs.openstack.org/09/640909/1/check/openstack-tox-py27/6f5af7a/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/09/640909/1/check/openstack-tox-py35/a695028/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/09/640909/1/check/openstack-tox-py36/efee8aa/ : SUCCESS in 4m 18s\n- openstack-tox-py37 http://logs.openstack.org/09/640909/1/check/openstack-tox-py37/5d1c2e5/ : SUCCESS in 3m 58s\n- openstack-tox-functional http://logs.openstack.org/09/640909/1/check/openstack-tox-functional/a937ee4/ : SUCCESS in 6m 31s\n- openstack-tox-functional-py35 http://logs.openstack.org/09/640909/1/check/openstack-tox-functional-py35/7689b47/ : SUCCESS in 4m 22s\n- openstack-tox-functional-py36 http://logs.openstack.org/09/640909/1/check/openstack-tox-functional-py36/106f8b9/ : SUCCESS in 4m 46s\n- placement-perfload http://logs.openstack.org/09/640909/1/check/placement-perfload/5ceec45/ : SUCCESS in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d1c142a45c80f10f737259ea2a86deb2ab5652a","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-03-05 15:41:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"7e9ba0d1a071a3abbc9c7fda4daa7365d18adb9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 15:42:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3633db504e650b4d30d1ebf06be7c14689464c9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 17:50:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/09/640909/1/gate/grenade-py3/1e7a66a/ : SUCCESS in 1h 00m 16s\n- tempest-full-py3 http://logs.openstack.org/09/640909/1/gate/tempest-full-py3/cf350ea/ : SUCCESS in 1h 15m 52s\n- openstack-tox-docs http://logs.openstack.org/09/640909/1/gate/openstack-tox-docs/0fba3dc/html/ : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/09/640909/1/gate/openstack-tox-lower-constraints/e0ada6f/ : SUCCESS in 5m 15s\n- openstack-tox-pep8 http://logs.openstack.org/09/640909/1/gate/openstack-tox-pep8/fe79ccf/ : SUCCESS in 6m 47s\n- openstack-tox-py27 http://logs.openstack.org/09/640909/1/gate/openstack-tox-py27/1319175/ : SUCCESS in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/09/640909/1/gate/openstack-tox-py35/75385a1/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/09/640909/1/gate/openstack-tox-py36/aac0f5a/ : SUCCESS in 8m 19s\n- openstack-tox-py37 http://logs.openstack.org/09/640909/1/gate/openstack-tox-py37/170e9c8/ : SUCCESS in 4m 50s\n- openstack-tox-functional http://logs.openstack.org/09/640909/1/gate/openstack-tox-functional/68510d5/ : SUCCESS in 4m 58s\n- openstack-tox-functional-py35 http://logs.openstack.org/09/640909/1/gate/openstack-tox-functional-py35/b079dca/ : SUCCESS in 4m 54s\n- openstack-tox-functional-py36 http://logs.openstack.org/09/640909/1/gate/openstack-tox-functional-py36/386c99f/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"61cb71b6d00fc9146a03264d61cb5e09f521d335","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 17:50:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d0c47945713663a59e53a3b02b1d73f77c9a273","revisions":{"0d0c47945713663a59e53a3b02b1d73f77c9a273":{"kind":"REWORK","_number":1,"created":"2019-03-05 01:05:39.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/09/640909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/09/640909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/09/640909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/09/640909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/09/640909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/09/640909/1"}}},"commit":{"parents":[{"commit":"0123e8c2bd5bcb58cfc197ad81d9ebe0b8b18be1","subject":"Merge \"Optionally migrate database at service startup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0123e8c2bd5bcb58cfc197ad81d9ebe0b8b18be1"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-03-01 10:39:55.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2019-03-01 10:51:33.000000000","tz":0},"subject":"FUPs for improve-debug-log series","message":"FUPs for improve-debug-log series\n\nThe following issues are addressed:\n\n - https://review.openstack.org/#/c/639888/3/placement/tests/unit/objects/test_rp_candidates.py@24\n - https://review.openstack.org/#/c/639888/3/placement/objects/resource_provider.py@3697\n\nChange-Id: Ia863977a154300ccf8da321c1828f83c951c9a19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0d0c47945713663a59e53a3b02b1d73f77c9a273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0d0c47945713663a59e53a3b02b1d73f77c9a273"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0123e8c2bd5bcb58cfc197ad81d9ebe0b8b18be1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
