)]}'
{"id":"openstack%2Fplacement~654799","triplet_id":"openstack%2Fplacement~master~I79ce8f1b20c60b89fc1ed0ece8bcc321e9435bad","project":"openstack/placement","branch":"master","topic":"bp/support-consumer-types","hashtags":[],"change_id":"I79ce8f1b20c60b89fc1ed0ece8bcc321e9435bad","subject":"Spec: Support Consumer Types","status":"MERGED","created":"2019-04-23 06:15:32.000000000","updated":"2019-07-01 21:19:50.000000000","submitted":"2019-07-01 21:19:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":68,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654799-1562015990034-16fc3d50","meta_rev_id":"b1462bc1e1c71165585380792002e3fa18b519a9","_number":654799,"virtual_id_number":654799,"owner":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2019-07-01 21:19:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2019-07-01 15:50:48.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-01 15:52:04.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-01 15:52:04.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 02:47:45.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-06-27 04:06:03.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-06-27 14:23:30.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-06-28 07:33:41.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-07-01 15:50:48.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-01 15:52:04.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-07-01 21:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20dad2c4dfe6929ce76ea408d19b9f1ee4c250ef","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-04-23 06:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3cd7d85aa345145c10bc240e88d24b690810dd8","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-04-23 06:16:13.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"76d6a6f24f4bfcd77da3b43101213a6f3bdb4955","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 06:57:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/654799/1/check/openstack-tox-docs/a1f804d/html/ : SUCCESS in 4m 19s\n- openstack-tox-pep8 http://logs.openstack.org/99/654799/1/check/openstack-tox-pep8/802343f/ : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"72f4b56d23f31aa083af4e8494fce5e8dd0f7c6b","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-04-23 20:16:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56f6d42c067aa60dcc5062f63fab9e65c195c57a","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-04 17:52:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07ecc57defd9fdf7e2055bb2938293432e6f7127","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-06 16:24:09.000000000","message":"Patch Set 1:\n\nSee this thread http://lists.openstack.org/pipermail/openstack-discuss/2019-May/thread.html#5878 for a summary of discussion from the PTG, including a slight change in plan: the default type will be something like \"unknown\" or \"incomplete\".","accounts_in_message":[],"_revision_number":1},{"id":"7cf1d1c39be6a3e6460b0801738ce0e1714f2574","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-15 14:26:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"35f50b662236bc9cafb9dfb1abace4bccbb14269","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-15 14:28:14.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fa5ddc6aeda809a8f8a208e2e2610545c392f85b","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-15 14:30:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"03a248ee6fdbf7efae43f119bb0f4a805896d710","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-15 14:32:39.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"84384f19d0c53b3e9c2c02f398237b81f6ab955a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:55:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/654799/4/check/openstack-tox-docs/7402c74/html/ : SUCCESS in 6m 10s\n- openstack-tox-pep8 http://logs.openstack.org/99/654799/4/check/openstack-tox-pep8/09a918d/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":4},{"id":"d290364548c6b08e1b2c139a18df90d57e6f29dd","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-15 17:03:29.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nThis is looking good. Some suggestions and questions within.\n\nAlso: you might want to flesh out the parts that describe the situation with database schema and data migrations.\n\nBut mostly the idea is sound, and what remain is hashing out the details of the requests and responses.","accounts_in_message":[],"_revision_number":4},{"id":"9697bf2c17c8b26bab459bb102857b3aeef288f8","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-15 17:35:41.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nthanks for the review and your time :)\n\n \u003e (4 comments)\n \u003e \n \u003e This is looking good. Some suggestions and questions within.\n \u003e \n \u003e Also: you might want to flesh out the parts that describe the\n \u003e situation with database schema and data migrations.\n\nack, I\u0027ll add info on the migration in the placement side for the \"UNKNOWN\".\n\n \u003e \n \u003e But mostly the idea is sound, and what remain is hashing out the\n \u003e details of the requests and responses.\n\nyep.","accounts_in_message":[],"_revision_number":4},{"id":"1411a924b1ecf6c414419e569b896b8b24818522","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-31 16:06:49.000000000","message":"Patch Set 4: Code-Review-1\n\n(13 comments)\n\nMain issues:\n- /reshaper needs consumer_type in request \u0026 response payloads\n- Definitely want to see ?consumer_type\u003dXXX filters on 1-4 GET APIs\n- Include consumer_type for parity in responses that can be turned around into requests\n- Missing totals in usages","accounts_in_message":[],"_revision_number":4},{"id":"713fbbabc022f8a6586852f7bf3706089a87e2cb","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-06-04 06:38:49.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"21c6421b1d8b78e78d1ab8f3a63fcc30dade2979","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-06-26 13:32:35.000000000","message":"Patch Set 4:\n\n(14 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a1c69ef2e900d45aaa22444d3c83ee6678b17688","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-06-26 13:33:36.000000000","message":"Patch Set 4:\n\n\u003e (13 comments)\n \u003e \n \u003e Main issues:\n \u003e - /reshaper needs consumer_type in request \u0026 response payloads\n \u003e - Definitely want to see ?consumer_type\u003dXXX filters on 1-4 GET APIs\n \u003e - Include consumer_type for parity in responses that can be turned\n \u003e around into requests\n \u003e - Missing totals in usages\n\nthanks for the review, hopefully I have addressed the mentioned points in a satisfactory manner.","accounts_in_message":[],"_revision_number":4},{"id":"710ea40fd2e796abd878680ca6f84cfd248cd555","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-06-26 13:47:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4ac8624fdee5231c3c36e1b53853028a665a1371","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-26 16:54:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nThis looks good to me now. Minor mistake in the regex for consumer type, but the intent is clear from the context; this can be fixed up later.\n\nThanks Surya!","accounts_in_message":[],"_revision_number":5},{"id":"fef3fe76029fd501ac2795c0c6b4d182c5b045d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 23:29:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/654799/5/check/openstack-tox-docs/8239607/html/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/99/654799/5/check/openstack-tox-pep8/a9964d7/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":5},{"id":"55ae0662592c6f5e758a5ab3cb570a5a5b0d4507","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-06-27 02:35:21.000000000","message":"Patch Set 5:\n\n(4 comments)\n\nSorry for being really late to reviewing this. LGTM, just leaving comments from PS4 that I had already made.","accounts_in_message":[],"_revision_number":5},{"id":"186538f21e510888a47f7ee60a604becbaf89d40","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-06-27 02:47:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3fdb7e53c86bf1a3d574b75750b6f686354a75c6","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-06-27 04:06:03.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"76796eb7faa57518a3b899534569e42cdbb7973e","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-06-27 09:04:36.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8fd434c868f0fb17dbbcee4168801d78d3c5e93b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-27 14:23:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2370acd0d554fcbf71ab0a7323a08267de7354df","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-06-27 15:28:15.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9997bdd83c10dd96ba70a4babe8123c156d32269","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-06-27 15:31:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f4194e88e312874ff38c7b60e7801c711ab6b181","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-01 14:31:21.000000000","message":"Patch Set 5:\n\n(6 comments)\n\nSome questions within about placement-side migrations (to add tables and columns) so going to hold off on +W, but mostly I\u0027m on board and we\u0027re just at the clarifying things stage at this point.\n\nI think it is good and useful that we are decoupling the management of the UNKNOWN -\u003e $known state.","accounts_in_message":[],"_revision_number":5},{"id":"72ffca723bf5cdd600b7e1f88c16642b6b79721a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-01 15:20:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b50b9c45d7dddb239004f305222f5bf41c4331ab","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-07-01 15:24:05.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3f5d086d835c6204f1aa15325759d13009c0e80f","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-07-01 15:45:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1fd1bc9cf019321413649d9ebcfb7c374f37fda0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-01 15:48:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"42d4961b4346d04f8eed993bc127fdc0064fd7d7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-01 15:50:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a8bd57b2f0aa790db6dff7da3ce05d729e5465c9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-01 15:50:48.000000000","message":"Patch Set 7: Code-Review+2\n\nI just corrected the regex.","accounts_in_message":[],"_revision_number":7},{"id":"9ccf8ceb4b060d17a5d60b84ecba6e34a7a95ba0","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-01 15:52:04.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nyeah make it go, thanks for those updates","accounts_in_message":[],"_revision_number":7},{"id":"7c6651da2edf9af37771b000d6453ede6fb5cb0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 18:59:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/654799/7/check/openstack-tox-docs/2aa6a16/html/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/99/654799/7/check/openstack-tox-pep8/f3aa5b4/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":7},{"id":"e7f6448c58b6e4859fa00506e09f74f9a5d361fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 19:00:14.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"891d0bf5835d3ed37d4b162afa9e4fd364cfda92","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:19:49.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/654799/7/gate/openstack-tox-docs/4f25d85/html/ : SUCCESS in 4m 10s\n- openstack-tox-pep8 http://logs.openstack.org/99/654799/7/gate/openstack-tox-pep8/bbbbd40/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":7},{"id":"e4806604cc00b7f577db432cb54d655a2e3a8845","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:19:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ed4a9a1d7847e7f4661ce4cc531141efeab48972","revisions":{"5226420f77e1fad0d55993c93973eb2e1667c79e":{"kind":"REWORK","_number":1,"created":"2019-04-23 06:15:32.000000000","uploader":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"ref":"refs/changes/99/654799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/99/654799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/99/654799/1"}}},"commit":{"parents":[{"commit":"931a9e124251a0322550ff016ae1ad080cd472f3","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/931a9e124251a0322550ff016ae1ad080cd472f3"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-04-23 06:04:41.000000000","tz":120},"committer":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-04-23 06:08:08.000000000","tz":120},"subject":"[WIP] Spec: Support Consumer Types","message":"[WIP] Spec: Support Consumer Types\n\nThis spec aims at providing support for services to model ``consumer types`` in\nplacement. While placement defines a consumer to be an entity consuming\nresources from a provider it does not provide a way to identify similar \"types\"\nof consumers or group/query consumers from a particular service. This spec\nproposes to associate each consumer to a particular type defined by the service\nowning the consumer.\n\nChange-Id: I79ce8f1b20c60b89fc1ed0ece8bcc321e9435bad\nStory: 2005473\nTask: 30554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5226420f77e1fad0d55993c93973eb2e1667c79e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5226420f77e1fad0d55993c93973eb2e1667c79e"}]},"branch":"refs/heads/master"},"9025b8b0334f07514851c2409b97853fa4f1ddb7":{"kind":"REWORK","_number":2,"created":"2019-05-15 14:26:23.000000000","uploader":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"ref":"refs/changes/99/654799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/99/654799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/99/654799/2"}}},"commit":{"parents":[{"commit":"1281806c99ceb80fed78237b39aa34b9097a39e6","subject":"Merge \"Skip _exclude_nested_providers() if not nested\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1281806c99ceb80fed78237b39aa34b9097a39e6"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-04-23 06:04:41.000000000","tz":120},"committer":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-05-15 14:26:10.000000000","tz":120},"subject":"Spec: Support Consumer Types","message":"Spec: Support Consumer Types\n\nThis spec aims at providing support for services to model ``consumer types``\nin placement. While placement defines a consumer to be an entity consuming\nresources from a provider it does not provide a way to identify similar\n\"types\" of consumers and henceforth allow services to group/query them based\non their types. This spec proposes to associate each consumer to a particular\ntype defined by the service owning the consumer.\n\nChange-Id: I79ce8f1b20c60b89fc1ed0ece8bcc321e9435bad\nStory: 2005473\nTask: 30554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9025b8b0334f07514851c2409b97853fa4f1ddb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9025b8b0334f07514851c2409b97853fa4f1ddb7"}]},"branch":"refs/heads/master"},"709d36d5293f6843556fd7050267a8fb598dd045":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-15 14:28:14.000000000","uploader":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"ref":"refs/changes/99/654799/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/99/654799/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/99/654799/3"}}},"commit":{"parents":[{"commit":"1281806c99ceb80fed78237b39aa34b9097a39e6","subject":"Merge \"Skip _exclude_nested_providers() if not nested\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1281806c99ceb80fed78237b39aa34b9097a39e6"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-04-23 06:04:41.000000000","tz":120},"committer":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-05-15 14:27:07.000000000","tz":120},"subject":"Spec: Support Consumer Types","message":"Spec: Support Consumer Types\n\nThis spec aims at providing support for services to model\n``consumer types`` in placement. While placement defines a\nconsumer to be an entity consuming resources from a provider\nit does not provide a way to identify similar \"types\" of\nconsumers and henceforth allow services to group/query them\nbased on their types. This spec proposes to associate each\nconsumer to a particular type defined by the service owning\nthe consumer.\n\nChange-Id: I79ce8f1b20c60b89fc1ed0ece8bcc321e9435bad\nStory: 2005473\nTask: 30554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/709d36d5293f6843556fd7050267a8fb598dd045"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/709d36d5293f6843556fd7050267a8fb598dd045"}]},"branch":"refs/heads/master"},"4a27702f875e3e4ba1fb37184f6a5eb265376101":{"kind":"REWORK","_number":4,"created":"2019-05-15 14:30:10.000000000","uploader":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"ref":"refs/changes/99/654799/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/99/654799/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/99/654799/4"}}},"commit":{"parents":[{"commit":"1281806c99ceb80fed78237b39aa34b9097a39e6","subject":"Merge \"Skip _exclude_nested_providers() if not nested\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1281806c99ceb80fed78237b39aa34b9097a39e6"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-04-23 06:04:41.000000000","tz":120},"committer":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-05-15 14:29:51.000000000","tz":120},"subject":"Spec: Support Consumer Types","message":"Spec: Support Consumer Types\n\nThis spec aims at providing support for services to model\n``consumer types`` in placement. While placement defines a\nconsumer to be an entity consuming resources from a provider\nit does not provide a way to identify similar \"types\" of\nconsumers and henceforth allow services to group/query them\nbased on their types. This spec proposes to associate each\nconsumer to a particular type defined by the service owning\nthe consumer.\n\nChange-Id: I79ce8f1b20c60b89fc1ed0ece8bcc321e9435bad\nStory: 2005473\nTask: 30554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4a27702f875e3e4ba1fb37184f6a5eb265376101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4a27702f875e3e4ba1fb37184f6a5eb265376101"}]},"branch":"refs/heads/master"},"5fa4a172b7d20a105ce9020c2c6e9487c351b168":{"kind":"REWORK","_number":5,"created":"2019-06-26 13:47:58.000000000","uploader":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"ref":"refs/changes/99/654799/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/99/654799/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/99/654799/5"}}},"commit":{"parents":[{"commit":"1281806c99ceb80fed78237b39aa34b9097a39e6","subject":"Merge \"Skip _exclude_nested_providers() if not nested\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1281806c99ceb80fed78237b39aa34b9097a39e6"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-04-23 06:04:41.000000000","tz":120},"committer":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-06-26 13:47:28.000000000","tz":120},"subject":"Spec: Support Consumer Types","message":"Spec: Support Consumer Types\n\nThis spec aims at providing support for services to model\n``consumer types`` in placement. While placement defines a\nconsumer to be an entity consuming resources from a provider\nit does not provide a way to identify similar \"types\" of\nconsumers and henceforth allow services to group/query them\nbased on their types. This spec proposes to associate each\nconsumer to a particular type defined by the service owning\nthe consumer.\n\nChange-Id: I79ce8f1b20c60b89fc1ed0ece8bcc321e9435bad\nStory: 2005473\nTask: 30554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5fa4a172b7d20a105ce9020c2c6e9487c351b168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5fa4a172b7d20a105ce9020c2c6e9487c351b168"}]},"branch":"refs/heads/master"},"1f9433431056bc579cd13f81c2b2e122b19e8d47":{"kind":"REWORK","_number":6,"created":"2019-07-01 15:45:56.000000000","uploader":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"ref":"refs/changes/99/654799/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/99/654799/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/99/654799/6"}}},"commit":{"parents":[{"commit":"1281806c99ceb80fed78237b39aa34b9097a39e6","subject":"Merge \"Skip _exclude_nested_providers() if not nested\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1281806c99ceb80fed78237b39aa34b9097a39e6"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-04-23 06:04:41.000000000","tz":120},"committer":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-01 15:42:40.000000000","tz":120},"subject":"Spec: Support Consumer Types","message":"Spec: Support Consumer Types\n\nThis spec aims at providing support for services to model\n``consumer types`` in placement. While placement defines a\nconsumer to be an entity consuming resources from a provider\nit does not provide a way to identify similar \"types\" of\nconsumers and henceforth allow services to group/query them\nbased on their types. This spec proposes to associate each\nconsumer to a particular type defined by the service owning\nthe consumer.\n\nChange-Id: I79ce8f1b20c60b89fc1ed0ece8bcc321e9435bad\nStory: 2005473\nTask: 30554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1f9433431056bc579cd13f81c2b2e122b19e8d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1f9433431056bc579cd13f81c2b2e122b19e8d47"}]},"branch":"refs/heads/master"},"ed4a9a1d7847e7f4661ce4cc531141efeab48972":{"kind":"REWORK","_number":7,"created":"2019-07-01 15:50:34.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/99/654799/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/99/654799/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/99/654799/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/99/654799/7"}}},"commit":{"parents":[{"commit":"1281806c99ceb80fed78237b39aa34b9097a39e6","subject":"Merge \"Skip _exclude_nested_providers() if not nested\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1281806c99ceb80fed78237b39aa34b9097a39e6"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-04-23 06:04:41.000000000","tz":120},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-07-01 15:50:31.000000000","tz":-300},"subject":"Spec: Support Consumer Types","message":"Spec: Support Consumer Types\n\nThis spec aims at providing support for services to model\n``consumer types`` in placement. While placement defines a\nconsumer to be an entity consuming resources from a provider\nit does not provide a way to identify similar \"types\" of\nconsumers and henceforth allow services to group/query them\nbased on their types. This spec proposes to associate each\nconsumer to a particular type defined by the service owning\nthe consumer.\n\nChange-Id: I79ce8f1b20c60b89fc1ed0ece8bcc321e9435bad\nStory: 2005473\nTask: 30554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ed4a9a1d7847e7f4661ce4cc531141efeab48972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ed4a9a1d7847e7f4661ce4cc531141efeab48972"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
