)]}'
{"id":"openstack%2Fplacement~657077","triplet_id":"openstack%2Fplacement~master~I72ae7f552879be13777fea90fa7055e056d309a7","project":"openstack/placement","branch":"master","topic":"story/2005562","hashtags":[],"change_id":"I72ae7f552879be13777fea90fa7055e056d309a7","subject":"Run nova-tox-functional-py36 in the placement gate","status":"MERGED","created":"2019-05-03 22:29:48.000000000","updated":"2019-05-05 18:40:03.000000000","submitted":"2019-05-05 18:40:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657077-1557081603572-959ddcf6","meta_rev_id":"fb1d1d5083ab1f31095b38dc169b78e0fd48e937","_number":657077,"virtual_id_number":657077,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-05-04 19:04:35.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2019-05-05 18:40:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":2,"date":"2019-05-04 19:22:07.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-05-05 10:59:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":1,"date":"2019-05-05 10:59:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 19:22:07.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-05-05 10:59:53.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-05-05 18:40:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff2f214e8477108c29f7d9c5a6d433c9d3cdac33","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-03 22:29:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75b8613c37fb28df90c8d8dffef5db3d86a01645","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-03 23:19:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9991cfecf01b15daad73e559961d1ea93ac6c13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-03 23:20:13.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"84b04ae6af18526e27a173bc4b859472d546858b","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-03 23:20:42.000000000","message":"Patch Set 2:\n\nrecheck dependency changed","accounts_in_message":[],"_revision_number":2},{"id":"d7c8ef32b1ea8b4ed97a1afcb71f96b5a9f1eb7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 00:15:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/77/657077/2/check/grenade-py3/e1801cb/ : SUCCESS in 53m 34s\n- tempest-full-py3 http://logs.openstack.org/77/657077/2/check/tempest-full-py3/b7b173b/ : FAILURE in 24m 36s\n- openstack-tox-docs http://logs.openstack.org/77/657077/2/check/openstack-tox-docs/4c8c91e/html/ : SUCCESS in 4m 43s\n- openstack-tox-cover http://logs.openstack.org/77/657077/2/check/openstack-tox-cover/7650068/cover/ : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657077/2/check/openstack-tox-lower-constraints/fc86a9d/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/77/657077/2/check/openstack-tox-pep8/95c40f7/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/77/657077/2/check/openstack-tox-py27/70226e2/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/77/657077/2/check/openstack-tox-py36/7c5463a/ : SUCCESS in 4m 49s\n- openstack-tox-py37 http://logs.openstack.org/77/657077/2/check/openstack-tox-py37/cc02372/ : SUCCESS in 5m 03s\n- openstack-tox-functional http://logs.openstack.org/77/657077/2/check/openstack-tox-functional/4454cea/ : SUCCESS in 5m 01s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/657077/2/check/openstack-tox-functional-py36/b2f2940/ : SUCCESS in 5m 08s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/77/657077/2/check/placement-nova-tox-functional-py36/f7567a0/ : SUCCESS in 12m 57s\n- placement-perfload http://logs.openstack.org/77/657077/2/check/placement-perfload/1ee9df2/ : SUCCESS in 3m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"044f49d1088b5a9c62dae60cf9421d6183a3206e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-04 01:11:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4129affa2c6edc0e0c3f648c9d457c890cc6ca4b","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-04 01:55:38.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8eef38e4a5ba2801bfc8f9555c6ef678f38b851c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-04 02:23:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35d5d476f0e2801d072ed0eba67149a7b795e19f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-04 15:23:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fd708d0f80e97ea233672bac0727eff270d5aed2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 16:51:35.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f44b114a9bf6ecfab7513672d9689025cc5c8861","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-04 19:04:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2bbed829f990aed810f6db2d1897e669b584fbfd","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-04 19:22:07.000000000","message":"Patch Set 2: Code-Review+2\n\nNeat-o. Could definitely trim that test list down more but this is a reasonable first step.","accounts_in_message":[],"_revision_number":2},{"id":"299aa45b146236a0bd35902a7a648ab857c09c0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-04 20:33:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/77/657077/2/check/grenade-py3/f631019/ : SUCCESS in 54m 42s\n- tempest-full-py3 http://logs.openstack.org/77/657077/2/check/tempest-full-py3/61782fe/ : SUCCESS in 1h 27m 38s\n- openstack-tox-docs http://logs.openstack.org/77/657077/2/check/openstack-tox-docs/992da78/html/ : SUCCESS in 4m 19s\n- openstack-tox-cover http://logs.openstack.org/77/657077/2/check/openstack-tox-cover/a73ef57/cover/ : SUCCESS in 7m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657077/2/check/openstack-tox-lower-constraints/d7e07e6/ : SUCCESS in 7m 13s\n- openstack-tox-pep8 http://logs.openstack.org/77/657077/2/check/openstack-tox-pep8/daf6f63/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/77/657077/2/check/openstack-tox-py27/dfd3685/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/77/657077/2/check/openstack-tox-py36/50b5060/ : SUCCESS in 4m 56s\n- openstack-tox-py37 http://logs.openstack.org/77/657077/2/check/openstack-tox-py37/b769165/ : SUCCESS in 5m 26s\n- openstack-tox-functional http://logs.openstack.org/77/657077/2/check/openstack-tox-functional/3449b32/ : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/657077/2/check/openstack-tox-functional-py36/140aff6/ : SUCCESS in 4m 16s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/77/657077/2/check/placement-nova-tox-functional-py36/67763de/ : SUCCESS in 13m 34s\n- placement-perfload http://logs.openstack.org/77/657077/2/check/placement-perfload/48785fc/ : SUCCESS in 3m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2df15ae684bc33b78187941a2dfe32ec864b5ba4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-05 10:59:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n\u003e Neat-o. Could definitely trim that test list down more but this is\n \u003e a reasonable first step.\n\nWe agreed not to be too eager about restricting the test list but if we want I guess we can turn off the notification sample tests too.","accounts_in_message":[],"_revision_number":2},{"id":"cd621c4d6d581c2cdfc059bcf3af30073b40b54e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 11:00:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f06bfd1dca3812c3bdc1695cff88efb96c644fa8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 14:34:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/77/657077/2/gate/grenade-py3/7ec6db7/ : SUCCESS in 56m 52s\n- tempest-full-py3 http://logs.openstack.org/77/657077/2/gate/tempest-full-py3/1be1691/ : FAILURE in 27m 38s\n- openstack-tox-docs http://logs.openstack.org/77/657077/2/gate/openstack-tox-docs/7ac16bd/html/ : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657077/2/gate/openstack-tox-lower-constraints/b117844/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/77/657077/2/gate/openstack-tox-pep8/07f5157/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/77/657077/2/gate/openstack-tox-py27/429fe61/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/77/657077/2/gate/openstack-tox-py36/1652be0/ : SUCCESS in 4m 30s\n- openstack-tox-py37 http://logs.openstack.org/77/657077/2/gate/openstack-tox-py37/5cd4342/ : SUCCESS in 4m 53s\n- openstack-tox-functional http://logs.openstack.org/77/657077/2/gate/openstack-tox-functional/aa39d7d/ : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/657077/2/gate/openstack-tox-functional-py36/c627f03/ : SUCCESS in 6m 34s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/77/657077/2/gate/placement-nova-tox-functional-py36/4a573ef/ : SUCCESS in 11m 11s","accounts_in_message":[],"_revision_number":2},{"id":"d3f9d1c8cdd45ed82d695c63352eb3ed1ca3e481","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-05 15:44:22.000000000","message":"Patch Set 2:\n\nrecheck bug 1827083","accounts_in_message":[],"_revision_number":2},{"id":"19462a7689ce92e85a5733cc99ace9868938b5e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 17:05:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/77/657077/2/check/grenade-py3/b0fab4b/ : SUCCESS in 1h 00m 56s\n- tempest-full-py3 http://logs.openstack.org/77/657077/2/check/tempest-full-py3/2328560/ : SUCCESS in 1h 19m 42s\n- openstack-tox-docs http://logs.openstack.org/77/657077/2/check/openstack-tox-docs/270575e/html/ : SUCCESS in 4m 57s\n- openstack-tox-cover http://logs.openstack.org/77/657077/2/check/openstack-tox-cover/6082a58/cover/ : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657077/2/check/openstack-tox-lower-constraints/6a52605/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/77/657077/2/check/openstack-tox-pep8/8360cd8/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/77/657077/2/check/openstack-tox-py27/196ea94/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/77/657077/2/check/openstack-tox-py36/c26dd36/ : SUCCESS in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/77/657077/2/check/openstack-tox-py37/b4d91ab/ : SUCCESS in 9m 10s\n- openstack-tox-functional http://logs.openstack.org/77/657077/2/check/openstack-tox-functional/805a812/ : SUCCESS in 4m 53s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/657077/2/check/openstack-tox-functional-py36/3645c43/ : SUCCESS in 4m 21s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/77/657077/2/check/placement-nova-tox-functional-py36/8c0b40d/ : SUCCESS in 11m 52s\n- placement-perfload http://logs.openstack.org/77/657077/2/check/placement-perfload/735f042/ : SUCCESS in 3m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ac4e4d76d3a53b6951d40486c5efdd8f69a9626","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 17:05:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e48741a009ada80da880177bf0efb7cefb10b30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 18:40:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7369728c60fa8ed70c470d8ad5100b8588d66355","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 18:40:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/77/657077/2/gate/grenade-py3/7c60df5/ : SUCCESS in 57m 43s\n- tempest-full-py3 http://logs.openstack.org/77/657077/2/gate/tempest-full-py3/85a29e1/ : SUCCESS in 1h 29m 26s\n- openstack-tox-docs http://logs.openstack.org/77/657077/2/gate/openstack-tox-docs/6ca2ecf/html/ : SUCCESS in 4m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/77/657077/2/gate/openstack-tox-lower-constraints/28b1ec7/ : SUCCESS in 4m 21s\n- openstack-tox-pep8 http://logs.openstack.org/77/657077/2/gate/openstack-tox-pep8/25d02aa/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/77/657077/2/gate/openstack-tox-py27/c8ee22c/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/77/657077/2/gate/openstack-tox-py36/269b6cb/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/77/657077/2/gate/openstack-tox-py37/b1a7686/ : SUCCESS in 4m 38s\n- openstack-tox-functional http://logs.openstack.org/77/657077/2/gate/openstack-tox-functional/f7a5042/ : SUCCESS in 4m 36s\n- openstack-tox-functional-py36 http://logs.openstack.org/77/657077/2/gate/openstack-tox-functional-py36/43e2aa6/ : SUCCESS in 4m 34s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/77/657077/2/gate/placement-nova-tox-functional-py36/9ed58ab/ : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e621d30097e547cba716c0311d982b55debdf6c0","revisions":{"80ae067b781d9a78a841e4ecfa1a46ff8e02b8d6":{"kind":"REWORK","_number":1,"created":"2019-05-03 22:29:48.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/77/657077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/77/657077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/77/657077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/77/657077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/77/657077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/77/657077/1"}}},"commit":{"parents":[{"commit":"8f2a400208d6abb96e7c6feec68fae4dc31095ec","subject":"Merge \"Update worklist information for contributors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8f2a400208d6abb96e7c6feec68fae4dc31095ec"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-05-03 22:27:51.000000000","tz":-360},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-05-03 22:27:51.000000000","tz":-360},"subject":"WIP: Run nova-tox-functional-py36 in the placement gate","message":"WIP: Run nova-tox-functional-py36 in the placement gate\n\nThis requires some slight adjustsments to the job in nova,\nfound in the depends-on.\n\nThe result of this will be _this_ placement change running\nagainst nova master.\n\nIf it works I will tune it to blacklist some jobs.\n\nDepends-On: https://review.opendev.org/657074\nChange-Id: I72ae7f552879be13777fea90fa7055e056d309a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/80ae067b781d9a78a841e4ecfa1a46ff8e02b8d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/80ae067b781d9a78a841e4ecfa1a46ff8e02b8d6"}]},"branch":"refs/heads/master"},"e621d30097e547cba716c0311d982b55debdf6c0":{"kind":"REWORK","_number":2,"created":"2019-05-03 23:19:28.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/77/657077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/77/657077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/77/657077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/77/657077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/77/657077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/77/657077/2"}}},"commit":{"parents":[{"commit":"8f2a400208d6abb96e7c6feec68fae4dc31095ec","subject":"Merge \"Update worklist information for contributors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8f2a400208d6abb96e7c6feec68fae4dc31095ec"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-05-03 22:27:51.000000000","tz":-360},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-05-03 23:17:45.000000000","tz":-360},"subject":"Run nova-tox-functional-py36 in the placement gate","message":"Run nova-tox-functional-py36 in the placement gate\n\nThis requires some slight adjustsments to the job in nova,\nfound in the depends-on.\n\nThe result of this will be _this_ placement change running\nagainst nova master.\n\nIn order to blacklist the api_sample_tests (to save some time)\nthe job is inherited and some args are added to the tox command.\n\nStory: 2005562\nTask: 30746\nDepends-On: https://review.opendev.org/657074\nChange-Id: I72ae7f552879be13777fea90fa7055e056d309a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e621d30097e547cba716c0311d982b55debdf6c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e621d30097e547cba716c0311d982b55debdf6c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
