)]}'
{"id":"openstack%2Fplacement~660051","triplet_id":"openstack%2Fplacement~master~I2796c3e93317edd3ae8b02fb038cc39f23b41840","project":"openstack/placement","branch":"master","topic":"story/2005712","hashtags":[],"change_id":"I2796c3e93317edd3ae8b02fb038cc39f23b41840","subject":"Move seek providers with resource to context","status":"MERGED","created":"2019-05-20 08:33:30.000000000","updated":"2019-05-30 04:10:16.000000000","submitted":"2019-05-30 04:10:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660051-1559189416979-efcb9f90","meta_rev_id":"36042023a2f4ba238b909e071b07727c187de7e4","_number":660051,"virtual_id_number":660051,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-05-30 04:10:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-05-29 21:13:09.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-29 20:23:01.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-05-29 21:13:09.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 13:43:48.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-05-29 21:13:09.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-05-30 04:10:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38fb5f74b7c87d27dcb158f376aa51cc4e362f04","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-20 08:33:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a1fe4526375bda2589d15e7d28846bfe64661a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 10:10:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/51/660051/1/check/grenade-py3/5dd6aed/ : SUCCESS in 57m 34s\n- tempest-full-py3 http://logs.openstack.org/51/660051/1/check/tempest-full-py3/a6dfc74/ : SUCCESS in 1h 31m 14s\n- openstack-tox-docs http://logs.openstack.org/51/660051/1/check/openstack-tox-docs/4611937/html/ : SUCCESS in 5m 39s\n- openstack-tox-cover http://logs.openstack.org/51/660051/1/check/openstack-tox-cover/97a7611/cover/ : SUCCESS in 5m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660051/1/check/openstack-tox-lower-constraints/40366bb/ : SUCCESS in 6m 04s\n- openstack-tox-pep8 http://logs.openstack.org/51/660051/1/check/openstack-tox-pep8/1faa7e8/ : FAILURE in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/51/660051/1/check/openstack-tox-py27/8608c2c/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/51/660051/1/check/openstack-tox-py36/dfb069b/ : SUCCESS in 3m 49s\n- openstack-tox-py37 http://logs.openstack.org/51/660051/1/check/openstack-tox-py37/57d7b56/ : SUCCESS in 4m 43s\n- openstack-tox-functional http://logs.openstack.org/51/660051/1/check/openstack-tox-functional/748358f/ : SUCCESS in 4m 45s\n- openstack-tox-functional-py36 http://logs.openstack.org/51/660051/1/check/openstack-tox-functional-py36/bb98133/ : SUCCESS in 4m 25s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/51/660051/1/check/placement-nova-tox-functional-py36/94baf16/ : SUCCESS in 10m 55s\n- placement-perfload http://logs.openstack.org/51/660051/1/check/placement-perfload/9d85a27/ : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eda0e196536286e661f14611a401a50dfa8c8865","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-20 13:36:12.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nnice. Question within about ordering.","accounts_in_message":[],"_revision_number":1},{"id":"b0c2df55f49058ce38c079fa9ecc4112211a23d5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-20 20:32:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99007db356bf5f56e071bfc5f8725ff2685e69c1","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-21 11:17:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eccdd6f76c0dd45158e575ebecf508fd2c5a4563","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-21 13:55:01.000000000","message":"Patch Set 2: Code-Review+1\n\nCool. +1 pending unWIP.","accounts_in_message":[],"_revision_number":2},{"id":"f40bb974679b165adca98d17a25cf81345bd819d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 18:45:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/51/660051/2/check/grenade-py3/6e6dfd8/ : SUCCESS in 1h 05m 15s\n- tempest-full-py3 http://logs.openstack.org/51/660051/2/check/tempest-full-py3/7293afa/ : SUCCESS in 1h 37m 29s\n- openstack-tox-docs http://logs.openstack.org/51/660051/2/check/openstack-tox-docs/80e0140/html/ : SUCCESS in 4m 08s\n- openstack-tox-cover http://logs.openstack.org/51/660051/2/check/openstack-tox-cover/1ca4af0/cover/ : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660051/2/check/openstack-tox-lower-constraints/9b50591/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/51/660051/2/check/openstack-tox-pep8/621f2f9/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/51/660051/2/check/openstack-tox-py27/ba130cb/ : SUCCESS in 4m 10s\n- openstack-tox-py36 http://logs.openstack.org/51/660051/2/check/openstack-tox-py36/0a32608/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/51/660051/2/check/openstack-tox-py37/fbc3ea7/ : SUCCESS in 5m 00s\n- openstack-tox-functional http://logs.openstack.org/51/660051/2/check/openstack-tox-functional/5484bb0/ : SUCCESS in 4m 24s\n- openstack-tox-functional-py36 http://logs.openstack.org/51/660051/2/check/openstack-tox-functional-py36/1cb8f3b/ : SUCCESS in 5m 05s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/51/660051/2/check/placement-nova-tox-functional-py36/534c716/ : SUCCESS in 13m 10s\n- placement-perfload http://logs.openstack.org/51/660051/2/check/placement-perfload/df8c251/ : SUCCESS in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df674a7ded03eb56d34c3d22896aded0aea5b73a","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-22 08:04:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"eb1f7359b3ccfec0f0779d95addf3fa2c9afbf15","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 10:11:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/51/660051/3/check/grenade-py3/1493731/ : SUCCESS in 1h 01m 19s\n- tempest-full-py3 http://logs.openstack.org/51/660051/3/check/tempest-full-py3/9646140/ : SUCCESS in 1h 37m 57s\n- openstack-tox-docs http://logs.openstack.org/51/660051/3/check/openstack-tox-docs/92da935/html/ : SUCCESS in 4m 10s\n- openstack-tox-cover http://logs.openstack.org/51/660051/3/check/openstack-tox-cover/1d8f7ab/cover/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660051/3/check/openstack-tox-lower-constraints/72618d3/ : SUCCESS in 7m 48s\n- openstack-tox-pep8 http://logs.openstack.org/51/660051/3/check/openstack-tox-pep8/c0ca1ca/ : SUCCESS in 5m 09s\n- openstack-tox-py27 http://logs.openstack.org/51/660051/3/check/openstack-tox-py27/ba18913/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/51/660051/3/check/openstack-tox-py36/c8c48bd/ : SUCCESS in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/51/660051/3/check/openstack-tox-py37/055d9c5/ : SUCCESS in 4m 11s\n- openstack-tox-functional http://logs.openstack.org/51/660051/3/check/openstack-tox-functional/e8d5e37/ : SUCCESS in 4m 17s\n- openstack-tox-functional-py36 http://logs.openstack.org/51/660051/3/check/openstack-tox-functional-py36/d2cb0cd/ : SUCCESS in 4m 01s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/51/660051/3/check/placement-nova-tox-functional-py36/53d12f2/ : SUCCESS in 11m 42s\n- placement-perfload http://logs.openstack.org/51/660051/3/check/placement-perfload/655dd6f/ : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"160f27430a0da9496ce1c15faa63c37bd22f6330","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-22 13:43:48.000000000","message":"Patch Set 3: Code-Review+2\n\nnice","accounts_in_message":[],"_revision_number":3},{"id":"d039957988f2184b6e18184fe080b2c40b8d8a81","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-26 08:34:29.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"10000f1462fa2cdbd4af817f219e681a90acaba5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-26 10:10:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/51/660051/4/check/grenade-py3/70c5355/ : SUCCESS in 54m 30s\n- tempest-full-py3 http://logs.openstack.org/51/660051/4/check/tempest-full-py3/a2be794/ : SUCCESS in 1h 34m 42s\n- openstack-tox-docs http://logs.openstack.org/51/660051/4/check/openstack-tox-docs/1ba3eeb/html/ : SUCCESS in 3m 49s\n- openstack-tox-cover http://logs.openstack.org/51/660051/4/check/openstack-tox-cover/8e7c5d3/cover/ : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660051/4/check/openstack-tox-lower-constraints/cdecd3f/ : SUCCESS in 4m 42s\n- openstack-tox-pep8 http://logs.openstack.org/51/660051/4/check/openstack-tox-pep8/804c587/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/51/660051/4/check/openstack-tox-py27/e4ef456/ : SUCCESS in 5m 34s\n- openstack-tox-py36 http://logs.openstack.org/51/660051/4/check/openstack-tox-py36/9d7db0d/ : SUCCESS in 5m 38s\n- openstack-tox-py37 http://logs.openstack.org/51/660051/4/check/openstack-tox-py37/32ba1bb/ : SUCCESS in 4m 28s\n- openstack-tox-functional http://logs.openstack.org/51/660051/4/check/openstack-tox-functional/c0f9c7c/ : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/51/660051/4/check/openstack-tox-functional-py36/a9dfd1f/ : SUCCESS in 4m 06s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/51/660051/4/check/placement-nova-tox-functional-py36/43b525f/ : SUCCESS in 11m 39s\n- placement-perfload http://logs.openstack.org/51/660051/4/check/placement-perfload/dddd6a4/ : SUCCESS in 2m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b8b30d6690a8540032fd0c221da69a161a294411","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-29 20:23:01.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f7c0117e22359a70bbc8968c52ecdc9e14147cd4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-29 21:13:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e6ec741cc284f783affbbacf702abe622fb5b996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 02:10:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/51/660051/5/check/grenade-py3/138e875/ : SUCCESS in 1h 03m 05s\n- tempest-full-py3 http://logs.openstack.org/51/660051/5/check/tempest-full-py3/5302e36/ : SUCCESS in 1h 35m 43s\n- openstack-tox-docs http://logs.openstack.org/51/660051/5/check/openstack-tox-docs/ef55283/html/ : SUCCESS in 4m 12s\n- openstack-tox-cover http://logs.openstack.org/51/660051/5/check/openstack-tox-cover/37eda18/cover/ : SUCCESS in 7m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660051/5/check/openstack-tox-lower-constraints/ef28acb/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/51/660051/5/check/openstack-tox-pep8/641116a/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/51/660051/5/check/openstack-tox-py27/68eb181/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/51/660051/5/check/openstack-tox-py36/feaf797/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/51/660051/5/check/openstack-tox-py37/ae98997/ : SUCCESS in 5m 39s\n- openstack-tox-functional http://logs.openstack.org/51/660051/5/check/openstack-tox-functional/b139d0e/ : SUCCESS in 4m 37s\n- openstack-tox-functional-py36 http://logs.openstack.org/51/660051/5/check/openstack-tox-functional-py36/11875bf/ : SUCCESS in 5m 19s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/51/660051/5/check/placement-nova-tox-functional-py36/6d57c69/ : SUCCESS in 15m 43s\n- placement-perfload http://logs.openstack.org/51/660051/5/check/placement-perfload/626b405/ : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d93ec58d261343f54358ed10c947dd1d860ba9c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 02:23:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ede34459ef5131713379612dd083b22d528f520e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 04:10:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/51/660051/5/gate/grenade-py3/88f3037/ : SUCCESS in 1h 04m 48s\n- tempest-full-py3 http://logs.openstack.org/51/660051/5/gate/tempest-full-py3/1043524/ : SUCCESS in 1h 38m 43s\n- openstack-tox-docs http://logs.openstack.org/51/660051/5/gate/openstack-tox-docs/23ec855/html/ : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/51/660051/5/gate/openstack-tox-lower-constraints/691e1ef/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/51/660051/5/gate/openstack-tox-pep8/0e15114/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/51/660051/5/gate/openstack-tox-py27/d1028e7/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/51/660051/5/gate/openstack-tox-py36/f9e53bc/ : SUCCESS in 5m 55s\n- openstack-tox-py37 http://logs.openstack.org/51/660051/5/gate/openstack-tox-py37/e69f670/ : SUCCESS in 6m 34s\n- openstack-tox-functional http://logs.openstack.org/51/660051/5/gate/openstack-tox-functional/3bbd484/ : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 http://logs.openstack.org/51/660051/5/gate/openstack-tox-functional-py36/9a22aa1/ : SUCCESS in 5m 46s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/51/660051/5/gate/placement-nova-tox-functional-py36/1125b63/ : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":5},{"id":"df825ac2eb03291e04f6b5d03e33ffdca866cdbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 04:10:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f8bbda15e7d16cd27697fcf6a9d52d8233c77be3","revisions":{"5cd88064ff5887e5b84eefd44bebbd6d91ea3f49":{"kind":"REWORK","_number":1,"created":"2019-05-20 08:33:30.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/51/660051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/51/660051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/51/660051/1"}}},"commit":{"parents":[{"commit":"b08e27805ecc8584e42354e060c331aef0894cd3","subject":"WIP: Remove normalize trait map func","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b08e27805ecc8584e42354e060c331aef0894cd3"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-16 06:17:29.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-19 07:12:06.000000000","tz":0},"subject":"WIP: Move seek providers with resource to context","message":"WIP: Move seek providers with resource to context\n\nThis patch moves the resource look up to `RequestGroupSearchContext`\ninitialization. This enables us to reuse the results for\n_get_providers_with_shared_capacity(). This optimization is done in\na following patch.\n\nChange-Id: I2796c3e93317edd3ae8b02fb038cc39f23b41840\nStory: 2005712\nTask: 31038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5cd88064ff5887e5b84eefd44bebbd6d91ea3f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5cd88064ff5887e5b84eefd44bebbd6d91ea3f49"}]},"branch":"refs/heads/master"},"abe3ba8d520f206c92a5acf23bec18efe77d12ad":{"kind":"REWORK","_number":2,"created":"2019-05-21 11:17:58.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/51/660051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/51/660051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/51/660051/2"}}},"commit":{"parents":[{"commit":"722c3c983b0a382860f2670246e1b8709bafae93","subject":"WIP: Remove normalize trait map func","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/722c3c983b0a382860f2670246e1b8709bafae93"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-16 06:17:29.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-21 09:43:00.000000000","tz":0},"subject":"WIP: Move seek providers with resource to context","message":"WIP: Move seek providers with resource to context\n\nThis patch moves the resource look up to `RequestGroupSearchContext`\ninitialization. This enables us to reuse the results for\n_get_providers_with_shared_capacity(). This optimization is done in\na following patch.\n\nChange-Id: I2796c3e93317edd3ae8b02fb038cc39f23b41840\nStory: 2005712\nTask: 31038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/abe3ba8d520f206c92a5acf23bec18efe77d12ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/abe3ba8d520f206c92a5acf23bec18efe77d12ad"}]},"branch":"refs/heads/master"},"940289d35ca698558d3b6b2f7ce4e64997813f5e":{"kind":"REWORK","_number":3,"created":"2019-05-22 08:04:24.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/51/660051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/51/660051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/51/660051/3"}}},"commit":{"parents":[{"commit":"afb99f29ca332e7e9190c0f8dba5aacff5a5f774","subject":"Remove normalize trait map func","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/afb99f29ca332e7e9190c0f8dba5aacff5a5f774"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-16 06:17:29.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-21 18:08:17.000000000","tz":0},"subject":"Move seek providers with resource to context","message":"Move seek providers with resource to context\n\nThis patch moves the resource look up to `RequestGroupSearchContext`\ninitialization. This enables us to reuse the results for\n_get_providers_with_shared_capacity(). That optimization is coming\nin a following patch.\n\nChange-Id: I2796c3e93317edd3ae8b02fb038cc39f23b41840\nStory: 2005712\nTask: 31038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/940289d35ca698558d3b6b2f7ce4e64997813f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/940289d35ca698558d3b6b2f7ce4e64997813f5e"}]},"branch":"refs/heads/master"},"963a716f7cfab7523b453e568a38bc6ef08fe571":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-26 08:34:29.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/51/660051/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/51/660051/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/51/660051/4"}}},"commit":{"parents":[{"commit":"b548aa03f045759d5fb71508dcadd1d2771a018f","subject":"Remove normalize trait map func","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b548aa03f045759d5fb71508dcadd1d2771a018f"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-16 06:17:29.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-26 08:06:38.000000000","tz":0},"subject":"Move seek providers with resource to context","message":"Move seek providers with resource to context\n\nThis patch moves the resource look up to `RequestGroupSearchContext`\ninitialization. This enables us to reuse the results for\n_get_providers_with_shared_capacity(). That optimization is coming\nin a following patch.\n\nChange-Id: I2796c3e93317edd3ae8b02fb038cc39f23b41840\nStory: 2005712\nTask: 31038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/963a716f7cfab7523b453e568a38bc6ef08fe571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/963a716f7cfab7523b453e568a38bc6ef08fe571"}]},"branch":"refs/heads/master"},"f8bbda15e7d16cd27697fcf6a9d52d8233c77be3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-05-29 20:23:01.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/51/660051/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/51/660051/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/51/660051/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/51/660051/5"}}},"commit":{"parents":[{"commit":"d75bdbff41806a978cfb562b7c8e18a468adf4d3","subject":"Remove normalize trait map func","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d75bdbff41806a978cfb562b7c8e18a468adf4d3"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-16 06:17:29.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-05-29 20:22:38.000000000","tz":-300},"subject":"Move seek providers with resource to context","message":"Move seek providers with resource to context\n\nThis patch moves the resource look up to `RequestGroupSearchContext`\ninitialization. This enables us to reuse the results for\n_get_providers_with_shared_capacity(). That optimization is coming\nin a following patch.\n\nChange-Id: I2796c3e93317edd3ae8b02fb038cc39f23b41840\nStory: 2005712\nTask: 31038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f8bbda15e7d16cd27697fcf6a9d52d8233c77be3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f8bbda15e7d16cd27697fcf6a9d52d8233c77be3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
