)]}'
{"id":"openstack%2Fplacement~661132","triplet_id":"openstack%2Fplacement~master~I57e834311c73282a2e7a00b14823812cb6f3e1e6","project":"openstack/placement","branch":"master","topic":"os-traits-version-canary","hashtags":[],"change_id":"I57e834311c73282a2e7a00b14823812cb6f3e1e6","subject":"Canary test for os-traits version","status":"MERGED","created":"2019-05-23 23:03:09.000000000","updated":"2019-05-24 13:04:50.000000000","submitted":"2019-05-24 13:04:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661132-1558703090340-afcd82ef","meta_rev_id":"64d3885286806763736dac77566bb8d521b2709d","_number":661132,"virtual_id_number":661132,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2019-05-24 13:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":2,"date":"2019-05-24 08:22:48.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-05-24 10:50:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":1,"date":"2019-05-24 10:50:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 08:22:48.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-05-24 10:50:08.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-05-24 13:04:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4965478a43a03d78b9fa5ebe7074c346d877bf80","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-23 23:03:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4eb1c3c9ef5d0f4fc3450babecb64d70520019cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 23:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/661132/1/check/openstack-tox-docs/73cb9fc/html/ : SUCCESS in 4m 05s\n- openstack-tox-cover http://logs.openstack.org/32/661132/1/check/openstack-tox-cover/ca3aa9b/cover/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/661132/1/check/openstack-tox-lower-constraints/33c5ae7/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/32/661132/1/check/openstack-tox-pep8/75ec390/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/32/661132/1/check/openstack-tox-py27/357f88e/ : SUCCESS in 5m 21s\n- openstack-tox-py36 http://logs.openstack.org/32/661132/1/check/openstack-tox-py36/84d8c10/ : SUCCESS in 9m 24s\n- openstack-tox-py37 http://logs.openstack.org/32/661132/1/check/openstack-tox-py37/00f5515/ : SUCCESS in 5m 33s\n- openstack-tox-functional http://logs.openstack.org/32/661132/1/check/openstack-tox-functional/e65e51a/ : SUCCESS in 5m 43s\n- openstack-tox-functional-py36 http://logs.openstack.org/32/661132/1/check/openstack-tox-functional-py36/fa7e263/ : SUCCESS in 6m 28s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/32/661132/1/check/placement-nova-tox-functional-py36/fa75c91/ : SUCCESS in 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"91452355fe7c45ced9373c920a9a44fda093ca0b","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-05-24 08:22:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a06a9e936d37d81f03cc5f562342bb7543ed8ee5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-24 10:50:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"4c78d6284248b4edde6935669adbc9510a2cf931","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 10:50:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ead0040063180dc93d0b286c2fdaea8c41ccb69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 13:04:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d5fe28ccd79d16dd52c88ee69cf07a4e0e49bc10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 13:04:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/32/661132/1/gate/openstack-tox-docs/877d8db/html/ : SUCCESS in 3m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/32/661132/1/gate/openstack-tox-lower-constraints/a98d09f/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/32/661132/1/gate/openstack-tox-pep8/d90e283/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/32/661132/1/gate/openstack-tox-py27/6da7104/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/32/661132/1/gate/openstack-tox-py36/31e45cc/ : SUCCESS in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/32/661132/1/gate/openstack-tox-py37/69328de/ : SUCCESS in 4m 25s\n- openstack-tox-functional http://logs.openstack.org/32/661132/1/gate/openstack-tox-functional/af0f702/ : SUCCESS in 4m 40s\n- openstack-tox-functional-py36 http://logs.openstack.org/32/661132/1/gate/openstack-tox-functional-py36/d31de82/ : SUCCESS in 4m 51s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/32/661132/1/gate/placement-nova-tox-functional-py36/567c80b/ : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d39ee720c630e5e2d4c3b8ce4da4855022cc960","revisions":{"7d39ee720c630e5e2d4c3b8ce4da4855022cc960":{"kind":"REWORK","_number":1,"created":"2019-05-23 23:03:09.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/32/661132/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/32/661132/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/32/661132/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/32/661132/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/32/661132/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/32/661132/1"}}},"commit":{"parents":[{"commit":"4bfffd7f0b0737e49036c2560bb021d8cdc22523","subject":"Bump os-resource-classes requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4bfffd7f0b0737e49036c2560bb021d8cdc22523"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-05-23 23:00:44.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-05-23 23:00:44.000000000","tz":-300},"subject":"Canary test for os-traits version","message":"Canary test for os-traits version\n\nSince [1] we have decided to pin placement to the latest versions of\nos-resource-classes and os-traits. Turns out we had gabbi tests that\nwould fail as soon as the os-resource-classes upper constraint was\nbumped, alerting us that it was time to sync that requirement. This\ncommit adds a similar test for os-traits.\n\n[1] https://review.opendev.org/#/c/658419/\n\nChange-Id: I57e834311c73282a2e7a00b14823812cb6f3e1e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7d39ee720c630e5e2d4c3b8ce4da4855022cc960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7d39ee720c630e5e2d4c3b8ce4da4855022cc960"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
