)]}'
{"id":"openstack%2Fplacement~662785","triplet_id":"openstack%2Fplacement~master~I3fdd46a0a92bf9666696a1c5f98afc402cf43b33","project":"openstack/placement","branch":"master","topic":"story/2005575","hashtags":[],"change_id":"I3fdd46a0a92bf9666696a1c5f98afc402cf43b33","subject":"Prepare objects for allocation request mappings","status":"MERGED","created":"2019-06-03 15:08:42.000000000","updated":"2019-06-29 17:01:39.000000000","submitted":"2019-06-14 11:13:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662785-1560510798598-d35274ac","meta_rev_id":"c1b713c0bfb5c20b16a51d23be838ad781836799","_number":662785,"virtual_id_number":662785,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-06-13 08:02:45.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2019-06-14 11:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-06-12 21:19:04.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-14 09:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-13 01:11:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":1,"date":"2019-06-14 09:19:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 20:56:09.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-13 01:11:24.000000000","updated_by":{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"},"reviewer":{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"},"state":"REVIEWER"},{"updated":"2019-06-14 09:19:10.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-06-14 11:13:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4201bd98bba7b12c00aed6c4a876497ed4f063e1","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-03 15:08:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9aaff04409b40f81ee2d5fb7a971f5b92c430f30","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-03 15:11:09.000000000","message":"Patch Set 1: Workflow-1\n\nthis splits the object-side work out from the changes in the following patch, hopefully to help a bit with reviewing\n\nIt needs some tests","accounts_in_message":[],"_revision_number":1},{"id":"ca9bbac6eb51d85bd0c0eae7b61b402b7fffb1fd","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-03 16:14:26.000000000","message":"Patch Set 1:\n\nWe already have the suffix in _merge_candidates, as a key to the allocation requests we\u0027re merging, so we possibly don\u0027t need to add it to either RequestGroupSearchContext or AllocationRequestResource.\n\nI guess I was kind of thinking there would be a new object (SuffixToProviderMapping or similar) as a member of AllocationRequest itself, that we would calculate and fill in during _merge_candidates.\n\nBut I haven\u0027t reviewed the subsequent change yet. I\u0027ll be interested to see how this plays out.","accounts_in_message":[],"_revision_number":1},{"id":"381e5146e6aff1439cdfdb25d2715fc178f5c790","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-03 16:26:23.000000000","message":"Patch Set 1:\n\n\u003e We already have the suffix in _merge_candidates, as a key to the\n \u003e allocation requests we\u0027re merging, so we possibly don\u0027t need to add\n \u003e it to either RequestGroupSearchContext or AllocationRequestResource.\n \u003e \n \u003e I guess I was kind of thinking there would be a new object\n \u003e (SuffixToProviderMapping or similar) as a member of\n \u003e AllocationRequest itself, that we would calculate and fill in\n \u003e during _merge_candidates.\n \u003e \n \u003e But I haven\u0027t reviewed the subsequent change yet. I\u0027ll be\n \u003e interested to see how this plays out.\n\nManaging the suffix on AllocationRequest (instead of AllocationRequestResource) resulted in incorrect results (I TDD\u0027d and tried multiple solutions to the failing tests).\n\nNew version of this patch, with slightly more tests in this vesrion.","accounts_in_message":[],"_revision_number":1},{"id":"06d4a326410966e3e7b857e1fda1e36634af8520","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-03 16:27:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6c414457a9a3aacae3e212d3c777038f4e7061d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 20:40:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/85/662785/2/check/grenade-py3/f039cea/ : SUCCESS in 1h 01m 05s\n- tempest-full-py3 http://logs.openstack.org/85/662785/2/check/tempest-full-py3/60495cb/ : SUCCESS in 1h 43m 32s\n- openstack-tox-docs http://logs.openstack.org/85/662785/2/check/openstack-tox-docs/2935f50/html/ : SUCCESS in 4m 14s\n- openstack-tox-cover http://logs.openstack.org/85/662785/2/check/openstack-tox-cover/2bd0c75/cover/ : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/662785/2/check/openstack-tox-lower-constraints/35e179b/ : SUCCESS in 5m 34s\n- openstack-tox-pep8 http://logs.openstack.org/85/662785/2/check/openstack-tox-pep8/51f6a19/ : SUCCESS in 4m 10s\n- openstack-tox-py27 http://logs.openstack.org/85/662785/2/check/openstack-tox-py27/6fc22fa/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/85/662785/2/check/openstack-tox-py36/39966fb/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/85/662785/2/check/openstack-tox-py37/5b9d796/ : SUCCESS in 4m 17s\n- openstack-tox-functional http://logs.openstack.org/85/662785/2/check/openstack-tox-functional/b2d38be/ : FAILURE in 4m 31s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/662785/2/check/openstack-tox-functional-py36/c38274f/ : SUCCESS in 4m 11s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/85/662785/2/check/placement-nova-tox-functional-py36/9ea45eb/ : SUCCESS in 13m 18s\n- placement-perfload http://logs.openstack.org/85/662785/2/check/placement-perfload/a363f28/ : SUCCESS in 3m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf2afac68276f278fa5931481120e14cba470de6","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-04 08:29:51.000000000","message":"Patch Set 2: Workflow-1\n\nLooks like there may be a bit of unpredictability in one of the added tests.\n\nWhich since it didn\u0027t fail elsewhere suggests the algorithm for merging candidates may itself be unpredictable or python version dependent. Will dig around some.","accounts_in_message":[],"_revision_number":2},{"id":"95052ef82889101bb933d6f70b2e5b82e529556e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-04 11:31:42.000000000","message":"Patch Set 2:\n\nYeah, this is a python 2 v 3 issue, not chased it down yet but I don\u0027t think it has anything to do with the suffixes...","accounts_in_message":[],"_revision_number":2},{"id":"e13be023a4c330091968c3fafb49a912b80756b4","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-04 20:56:45.000000000","message":"Patch Set 2:\n\nA fix for the failures in this are in https://review.opendev.org/663137 but they point out that the hash handling will need to be suffix-aware.","accounts_in_message":[],"_revision_number":2},{"id":"d83e0d17d01f75acb6efcd0835a4e718c4138410","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-06 18:51:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d7a2fff1896d42800ef0b9d92bad0637cbf4e205","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 00:48:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/85/662785/3/check/grenade-py3/efd4559/ : SUCCESS in 1h 03m 59s\n- tempest-full-py3 http://logs.openstack.org/85/662785/3/check/tempest-full-py3/998a869/ : SUCCESS in 1h 24m 26s\n- openstack-tox-docs http://logs.openstack.org/85/662785/3/check/openstack-tox-docs/49e78ff/html/ : SUCCESS in 5m 11s\n- openstack-tox-cover http://logs.openstack.org/85/662785/3/check/openstack-tox-cover/8681ff7/cover/ : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/662785/3/check/openstack-tox-lower-constraints/35213ef/ : SUCCESS in 5m 03s\n- openstack-tox-pep8 http://logs.openstack.org/85/662785/3/check/openstack-tox-pep8/765c792/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/85/662785/3/check/openstack-tox-py27/df53c31/ : SUCCESS in 6m 38s\n- openstack-tox-py36 http://logs.openstack.org/85/662785/3/check/openstack-tox-py36/21cb3dc/ : SUCCESS in 4m 39s\n- openstack-tox-py37 http://logs.openstack.org/85/662785/3/check/openstack-tox-py37/3728eed/ : SUCCESS in 7m 14s\n- openstack-tox-functional http://logs.openstack.org/85/662785/3/check/openstack-tox-functional/18b0334/ : FAILURE in 5m 00s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/662785/3/check/openstack-tox-functional-py36/c497157/ : SUCCESS in 5m 27s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/85/662785/3/check/placement-nova-tox-functional-py36/2884b79/ : SUCCESS in 11m 33s\n- placement-perfload http://logs.openstack.org/85/662785/3/check/placement-perfload/5538f52/ : SUCCESS in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92bcdbfcf470f496cd88ae0be178317a2cb0a4ce","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-07 13:32:44.000000000","message":"Patch Set 3:\n\nThe test failure goes away once https://review.opendev.org/#/c/663137/ is in?\n\nAnd this patch is or is not the right place to add the suffix to the hash/sort?","accounts_in_message":[],"_revision_number":3},{"id":"dab1a24fc8566c6e6bb26370ab4e677538562047","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-07 14:09:46.000000000","message":"Patch Set 3:\n\n\u003e The test failure goes away once https://review.opendev.org/#/c/663137/\n \u003e is in?\n\nConceptually, yes, but there will be merge conflicts.\n\n \u003e And this patch is or is not the right place to add the suffix to\n \u003e the hash/sort?\n\nIt may or may not need to be in the AR hash\u0027s sort (need to experiment), but it will definitely need to go in the ARR hash. In this patch.","accounts_in_message":[],"_revision_number":3},{"id":"c7ac60d0dc011225d87e64d2d7107042023fb5ba","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-07 15:00:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"849cfaa3ac87cecba0af18fff1276aabac529f7f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-07 15:02:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b6513808df39ad8d59830a242d88a5c0c02d747e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-07 16:39:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ef8ac8196e9e9bb99186617a05f20be21d2d004f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-07 20:56:09.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"536961961a2b0633b7c0a724545772d5d3ee84f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 03:36:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/85/662785/6/check/grenade-py3/087f709/ : SUCCESS in 1h 04m 32s\n- tempest-full-py3 http://logs.openstack.org/85/662785/6/check/tempest-full-py3/126f724/ : SUCCESS in 1h 40m 09s\n- openstack-tox-docs http://logs.openstack.org/85/662785/6/check/openstack-tox-docs/7041f77/html/ : SUCCESS in 4m 00s\n- openstack-tox-cover http://logs.openstack.org/85/662785/6/check/openstack-tox-cover/6c7864f/cover/ : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/662785/6/check/openstack-tox-lower-constraints/2711314/ : SUCCESS in 8m 54s\n- openstack-tox-pep8 http://logs.openstack.org/85/662785/6/check/openstack-tox-pep8/237e789/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/85/662785/6/check/openstack-tox-py27/7d05361/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/85/662785/6/check/openstack-tox-py36/e99cb0c/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/85/662785/6/check/openstack-tox-py37/4852b8d/ : SUCCESS in 5m 33s\n- openstack-tox-functional http://logs.openstack.org/85/662785/6/check/openstack-tox-functional/4e323d5/ : SUCCESS in 4m 29s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/662785/6/check/openstack-tox-functional-py36/8e3fc09/ : SUCCESS in 4m 30s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/85/662785/6/check/placement-nova-tox-functional-py36/30fc74c/ : SUCCESS in 12m 24s\n- placement-perfload http://logs.openstack.org/85/662785/6/check/placement-perfload/dfb75e1/ : SUCCESS in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"81a4259a5cc7f0a448789a5f589ae33f2500341f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-10 09:28:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"db986864427b18e840137bfdc3e445153a215b33","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-12 21:19:04.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"609ff25b5b0fa4f3746579af1ab7592ee4605cdd","author":{"_account_id":30526,"name":"AWCHENjun","email":"chenjun@awcloud.com"},"date":"2019-06-13 01:11:24.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"44e0f4db7191d0e4cd7356416548d9947c06428c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 05:11:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/85/662785/7/check/grenade-py3/435a941/ : SUCCESS in 57m 57s\n- tempest-full-py3 http://logs.openstack.org/85/662785/7/check/tempest-full-py3/307e5de/ : SUCCESS in 1h 33m 43s\n- openstack-tox-docs http://logs.openstack.org/85/662785/7/check/openstack-tox-docs/d24a289/html/ : SUCCESS in 4m 38s\n- openstack-tox-cover http://logs.openstack.org/85/662785/7/check/openstack-tox-cover/69866db/cover/ : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/662785/7/check/openstack-tox-lower-constraints/e6e0a4e/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/85/662785/7/check/openstack-tox-pep8/9516300/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/85/662785/7/check/openstack-tox-py27/304fb11/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/85/662785/7/check/openstack-tox-py36/a689ea3/ : SUCCESS in 5m 55s\n- openstack-tox-py37 http://logs.openstack.org/85/662785/7/check/openstack-tox-py37/453398e/ : SUCCESS in 4m 53s\n- openstack-tox-functional http://logs.openstack.org/85/662785/7/check/openstack-tox-functional/65192ea/ : SUCCESS in 5m 30s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/662785/7/check/openstack-tox-functional-py36/24a9679/ : SUCCESS in 7m 17s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/85/662785/7/check/placement-nova-tox-functional-py36/43aed3d/ : SUCCESS in 11m 05s\n- placement-perfload http://logs.openstack.org/85/662785/7/check/placement-perfload/cbad533/ : SUCCESS in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dd6a83faddaa2a2067137962d8f7802ebc07cb24","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-13 08:02:45.000000000","message":"Patch Set 7:\n\nNot surprisingly, this is a small performance degradation. That\u0027s expected, but it would be good to do some profiling to find somewhere to get back some of the time lost.","accounts_in_message":[],"_revision_number":7},{"id":"73b221ea1eae9ef558308e10a9f90381963cefc5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-14 09:08:59.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)\n\nLooks good. The missing param docs can be fixed in a follow up","accounts_in_message":[],"_revision_number":7},{"id":"84163387ce1058654b13829faa81b6653789200a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-14 09:19:10.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4ea4b246a4f060a94bf9b21254fb9d00b58b8843","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 09:19:23.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"4bb8ab2b50422e4b29d11d3f05b4c8e44fdb64ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 11:13:18.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/85/662785/7/gate/grenade-py3/cd3fec2/ : SUCCESS in 53m 50s\n- tempest-full-py3 http://logs.openstack.org/85/662785/7/gate/tempest-full-py3/7e5e902/ : SUCCESS in 1h 50m 11s\n- openstack-tox-docs http://logs.openstack.org/85/662785/7/gate/openstack-tox-docs/b866b84/html/ : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/662785/7/gate/openstack-tox-lower-constraints/fc6e801/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/85/662785/7/gate/openstack-tox-pep8/6bb3f5f/ : SUCCESS in 8m 30s\n- openstack-tox-py27 http://logs.openstack.org/85/662785/7/gate/openstack-tox-py27/ccede4b/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/85/662785/7/gate/openstack-tox-py36/4b95e4e/ : SUCCESS in 4m 52s\n- openstack-tox-py37 http://logs.openstack.org/85/662785/7/gate/openstack-tox-py37/2bcc3c4/ : SUCCESS in 4m 44s\n- openstack-tox-functional http://logs.openstack.org/85/662785/7/gate/openstack-tox-functional/93a6c17/ : SUCCESS in 4m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/85/662785/7/gate/openstack-tox-functional-py36/861b225/ : SUCCESS in 4m 51s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/85/662785/7/gate/placement-nova-tox-functional-py36/df3880d/ : SUCCESS in 11m 42s","accounts_in_message":[],"_revision_number":7},{"id":"b30be73babf44dd43b7f5be4bfb71e8911ac4e0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 11:13:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"49ae4ac2afca7276d4e13661f485cd34f53b9db6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-29 17:01:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"eb07913442974102b36b8127db8491a742f5eee1","revisions":{"73bc8234c822e5fce9ff8fe893d3b97beff8a0c7":{"kind":"REWORK","_number":1,"created":"2019-06-03 15:08:42.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/85/662785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/662785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/662785/1"}}},"commit":{"parents":[{"commit":"58a256bd0e337fa495d77153047b4f19db525cbf","subject":"Add NUMANetworkFixture for gabbits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/58a256bd0e337fa495d77153047b4f19db525cbf"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-03 15:03:32.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-03 15:03:32.000000000","tz":60},"subject":"Prepare objects for allocation request mappings","message":"Prepare objects for allocation request mappings\n\nWe need to carry a \u0027suffix\u0027 through the process of generating\nallocation requests in order to be able to present those suffixes\nas a \u0027mappings\u0027 key in a forthcoming microversion.\n\nIn this patch, the suffix is tracked, but the data is not used\nwhen presenting results.\n\nChange-Id: I3fdd46a0a92bf9666696a1c5f98afc402cf43b33\nStory: 2005575\nTask: 33536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/73bc8234c822e5fce9ff8fe893d3b97beff8a0c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/73bc8234c822e5fce9ff8fe893d3b97beff8a0c7"}]},"branch":"refs/heads/master"},"3e2d5858c82f88ab3258e0f51ebe91d58473356f":{"kind":"REWORK","_number":2,"created":"2019-06-03 16:27:19.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/85/662785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/662785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/662785/2"}}},"commit":{"parents":[{"commit":"58a256bd0e337fa495d77153047b4f19db525cbf","subject":"Add NUMANetworkFixture for gabbits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/58a256bd0e337fa495d77153047b4f19db525cbf"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-03 15:03:32.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-03 16:26:33.000000000","tz":60},"subject":"Prepare objects for allocation request mappings","message":"Prepare objects for allocation request mappings\n\nWe need to carry a \u0027suffix\u0027 through the process of generating\nallocation requests in order to be able to present those suffixes\nas a \u0027mappings\u0027 key in a forthcoming microversion.\n\nIn this patch, the suffix is tracked, but the data is not used\nwhen presenting results.\n\nA small number of confirming tests are added to\ndb.test_allocation_candidates.\n\nChange-Id: I3fdd46a0a92bf9666696a1c5f98afc402cf43b33\nStory: 2005575\nTask: 33536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3e2d5858c82f88ab3258e0f51ebe91d58473356f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3e2d5858c82f88ab3258e0f51ebe91d58473356f"}]},"branch":"refs/heads/master"},"f886b19f7145924016203d4bf294d6518856e07d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-06 18:51:59.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/85/662785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/662785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/662785/3"}}},"commit":{"parents":[{"commit":"bb8a3f89e47fc743d1c7852f9652b4ead5f37c00","subject":"Add NUMANetworkFixture for gabbits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bb8a3f89e47fc743d1c7852f9652b4ead5f37c00"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-03 15:03:32.000000000","tz":60},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-06 18:51:40.000000000","tz":-300},"subject":"Prepare objects for allocation request mappings","message":"Prepare objects for allocation request mappings\n\nWe need to carry a \u0027suffix\u0027 through the process of generating\nallocation requests in order to be able to present those suffixes\nas a \u0027mappings\u0027 key in a forthcoming microversion.\n\nIn this patch, the suffix is tracked, but the data is not used\nwhen presenting results.\n\nA small number of confirming tests are added to\ndb.test_allocation_candidates.\n\nChange-Id: I3fdd46a0a92bf9666696a1c5f98afc402cf43b33\nStory: 2005575\nTask: 33536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f886b19f7145924016203d4bf294d6518856e07d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f886b19f7145924016203d4bf294d6518856e07d"}]},"branch":"refs/heads/master"},"db7f6aacfee460eb595ac15395cb3e59758b6071":{"kind":"REWORK","_number":4,"created":"2019-06-07 15:00:38.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/85/662785/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/662785/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/662785/4"}}},"commit":{"parents":[{"commit":"d63b16828a1924063a62022ce9ca73391864cd9b","subject":"Add NUMANetworkFixture for gabbits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d63b16828a1924063a62022ce9ca73391864cd9b"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-03 15:03:32.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-07 14:58:23.000000000","tz":60},"subject":"Prepare objects for allocation request mappings","message":"Prepare objects for allocation request mappings\n\nWe need to carry a \u0027suffix\u0027 through the process of generating\nallocation requests in order to be able to present those suffixes\nas a \u0027mappings\u0027 key in a forthcoming microversion.\n\nIn this patch, the suffix is tracked, but the data is not used\nwhen presenting results.\n\nA test is added to validate results. Existing tests which\ncheck the count of results are adjusted to reflect new results:\nThe possible combinations is now increased because we are\naccounting for the suffix as a differentiator.\n\nChange-Id: I3fdd46a0a92bf9666696a1c5f98afc402cf43b33\nStory: 2005575\nTask: 33536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/db7f6aacfee460eb595ac15395cb3e59758b6071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/db7f6aacfee460eb595ac15395cb3e59758b6071"}]},"branch":"refs/heads/master"},"66ff8ef0a23c55d56f5124e04933eeba28aa5230":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-07 15:02:13.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/85/662785/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/662785/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/662785/5"}}},"commit":{"parents":[{"commit":"3578c8c54f5b938ec472586cdb298451b83a538d","subject":"Add NUMANetworkFixture for gabbits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3578c8c54f5b938ec472586cdb298451b83a538d"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-03 15:03:32.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-07 15:02:13.000000000","tz":0},"subject":"Prepare objects for allocation request mappings","message":"Prepare objects for allocation request mappings\n\nWe need to carry a \u0027suffix\u0027 through the process of generating\nallocation requests in order to be able to present those suffixes\nas a \u0027mappings\u0027 key in a forthcoming microversion.\n\nIn this patch, the suffix is tracked, but the data is not used\nwhen presenting results.\n\nA test is added to validate results. Existing tests which\ncheck the count of results are adjusted to reflect new results:\nThe possible combinations is now increased because we are\naccounting for the suffix as a differentiator.\n\nChange-Id: I3fdd46a0a92bf9666696a1c5f98afc402cf43b33\nStory: 2005575\nTask: 33536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/66ff8ef0a23c55d56f5124e04933eeba28aa5230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/66ff8ef0a23c55d56f5124e04933eeba28aa5230"}]},"branch":"refs/heads/master"},"e61e88c153692032a0757d2465faf07b5322f424":{"kind":"REWORK","_number":6,"created":"2019-06-07 16:39:19.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/85/662785/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/662785/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/662785/6"}}},"commit":{"parents":[{"commit":"fab707399ec9e05353ace9c9d920ceb00c27b52f","subject":"Add NUMANetworkFixture for gabbits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/fab707399ec9e05353ace9c9d920ceb00c27b52f"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-03 15:03:32.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-07 16:15:14.000000000","tz":60},"subject":"Prepare objects for allocation request mappings","message":"Prepare objects for allocation request mappings\n\nWe need to carry a \u0027suffix\u0027 through the process of generating\nallocation requests in order to be able to present those suffixes\nas a \u0027mappings\u0027 key in a forthcoming microversion.\n\nIn this patch, the suffix is tracked, but the data is not used\nwhen presenting results.\n\nA test is added to validate results. Existing tests which\ncheck the count of results are adjusted to reflect new results:\nThe possible combinations is now increased because we are\naccounting for the suffix as a differentiator.\n\nChange-Id: I3fdd46a0a92bf9666696a1c5f98afc402cf43b33\nStory: 2005575\nTask: 33536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e61e88c153692032a0757d2465faf07b5322f424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e61e88c153692032a0757d2465faf07b5322f424"}]},"branch":"refs/heads/master"},"eb07913442974102b36b8127db8491a742f5eee1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-06-12 21:19:04.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/85/662785/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/85/662785/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/85/662785/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/85/662785/7"}}},"commit":{"parents":[{"commit":"b04a15cac7b67dda05759531f4e1409b8e27a8cb","subject":"Merge \"Add NUMANetworkFixture for gabbits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b04a15cac7b67dda05759531f4e1409b8e27a8cb"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-03 15:03:32.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-12 21:19:04.000000000","tz":0},"subject":"Prepare objects for allocation request mappings","message":"Prepare objects for allocation request mappings\n\nWe need to carry a \u0027suffix\u0027 through the process of generating\nallocation requests in order to be able to present those suffixes\nas a \u0027mappings\u0027 key in a forthcoming microversion.\n\nIn this patch, the suffix is tracked, but the data is not used\nwhen presenting results.\n\nA test is added to validate results. Existing tests which\ncheck the count of results are adjusted to reflect new results:\nThe possible combinations is now increased because we are\naccounting for the suffix as a differentiator.\n\nChange-Id: I3fdd46a0a92bf9666696a1c5f98afc402cf43b33\nStory: 2005575\nTask: 33536\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/eb07913442974102b36b8127db8491a742f5eee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/eb07913442974102b36b8127db8491a742f5eee1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
