)]}'
{"id":"openstack%2Fplacement~665691","triplet_id":"openstack%2Fplacement~master~Id433b0540e2051ab27b2c22cc14c62105555ea8f","project":"openstack/placement","branch":"master","topic":"story/2005575-33753","hashtags":[],"change_id":"Id433b0540e2051ab27b2c22cc14c62105555ea8f","subject":"Miscellaneous doc/comment/log cleanups","status":"MERGED","created":"2019-06-17 14:32:18.000000000","updated":"2019-06-27 05:47:50.000000000","submitted":"2019-06-27 05:47:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665691-1561614470922-e8d46088","meta_rev_id":"db3598a2d289e8a3771e113755c946d513900dff","_number":665691,"virtual_id_number":665691,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2019-06-26 05:03:32.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":2,"date":"2019-06-27 05:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-06-27 04:12:34.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-26 15:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-06-27 04:12:34.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 08:29:25.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-19 07:26:55.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-06-26 15:33:40.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-06-27 04:12:34.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-06-27 05:47:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b35a7e9be5225ef6a724f196cc07da1dc08b2a57","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 14:32:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3509fe1abe6ef24c2dc2bd519febc55b290e766d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 15:26:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4e2903f37ef711b7888ba92a5a8d5e465998b55d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 15:28:03.000000000","message":"Topic set to story/2005575-33753","accounts_in_message":[],"_revision_number":2},{"id":"10c4b692bada2352dbb0fdac5be124f04774c82e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 17:46:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f010e9a01d676d652a116328a9b6d4ff26ad60f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-17 20:56:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nnitish thing within, but since there\u0027s gonna be a rebase anyway, and the fix to the nit will be above this, may as well\n\nOtherwise changes are legit","accounts_in_message":[],"_revision_number":3},{"id":"d37a158dcc9fbcf2f35d56afb37d8f3d30429f1c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 22:28:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"84a47827854915bad5e31ab31713586c84ef1717","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 22:42:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/91/665691/3/check/grenade-py3/2fbc447/ : SUCCESS in 58m 42s\n- tempest-full-py3 http://logs.openstack.org/91/665691/3/check/tempest-full-py3/1169a0b/ : SUCCESS in 1h 38m 57s\n- openstack-tox-docs http://logs.openstack.org/91/665691/3/check/openstack-tox-docs/64f5a99/html/ : SUCCESS in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/91/665691/3/check/openstack-tox-cover/b824363/ : FAILURE in 5m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/3/check/openstack-tox-lower-constraints/cebe739/ : FAILURE in 5m 56s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/3/check/openstack-tox-pep8/f81af71/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/3/check/openstack-tox-py27/bd0c119/ : FAILURE in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/3/check/openstack-tox-py36/35da5ce/ : FAILURE in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/3/check/openstack-tox-py37/de94b6a/ : FAILURE in 4m 52s\n- openstack-tox-functional http://logs.openstack.org/91/665691/3/check/openstack-tox-functional/807a403/ : SUCCESS in 5m 58s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/3/check/openstack-tox-functional-py36/185e2a0/ : SUCCESS in 4m 31s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/3/check/placement-nova-tox-functional-py36/c00c3c8/ : SUCCESS in 11m 27s\n- placement-perfload http://logs.openstack.org/91/665691/3/check/placement-perfload/988da40/ : SUCCESS in 7m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"41a30259f147cc8a9ba6dcfc17d0620daa749e5b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 22:43:55.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"0b48543b9ccaf0d4138e1bace95292835ba6c549","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 22:45:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9832ca849a06b74368d94011ec9ef052f3e7cd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 02:34:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/665691/4/check/grenade-py3/d21ffc7/ : SUCCESS in 57m 07s\n- tempest-full-py3 http://logs.openstack.org/91/665691/4/check/tempest-full-py3/6552e7b/ : SUCCESS in 1h 33m 10s\n- openstack-tox-docs http://logs.openstack.org/91/665691/4/check/openstack-tox-docs/f7da7b2/html/ : SUCCESS in 4m 20s\n- openstack-tox-cover http://logs.openstack.org/91/665691/4/check/openstack-tox-cover/34abac9/cover/ : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/4/check/openstack-tox-lower-constraints/59241c3/ : SUCCESS in 5m 36s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/4/check/openstack-tox-pep8/a5e4414/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/4/check/openstack-tox-py27/9b30d42/ : SUCCESS in 7m 20s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/4/check/openstack-tox-py36/9cbd2fa/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/4/check/openstack-tox-py37/d6f04b8/ : SUCCESS in 4m 57s\n- openstack-tox-functional http://logs.openstack.org/91/665691/4/check/openstack-tox-functional/c5c0a7b/ : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/4/check/openstack-tox-functional-py36/957a14f/ : SUCCESS in 5m 07s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/4/check/placement-nova-tox-functional-py36/450134e/ : SUCCESS in 11m 34s\n- placement-perfload http://logs.openstack.org/91/665691/4/check/placement-perfload/39bcfd4/ : SUCCESS in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"59eb26f20e57a8d4bed55ed0386d8d9602a36d42","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-18 08:29:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f98b004797a770b3eecfac67520a24310ac2134a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-18 14:34:04.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"a8b69a4fee29b14bd203484df2524e7bd086b3c9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-18 20:34:39.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"277b5c32a629da6263fc4a687a53aee7121b8211","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 05:50:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/665691/6/check/grenade-py3/5647322/ : SUCCESS in 1h 02m 53s\n- tempest-full-py3 http://logs.openstack.org/91/665691/6/check/tempest-full-py3/02d471c/ : SUCCESS in 1h 32m 35s\n- openstack-tox-docs http://logs.openstack.org/91/665691/6/check/openstack-tox-docs/ca72686/html/ : SUCCESS in 4m 14s\n- openstack-tox-cover http://logs.openstack.org/91/665691/6/check/openstack-tox-cover/3cd4357/cover/ : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/6/check/openstack-tox-lower-constraints/6159685/ : SUCCESS in 5m 01s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/6/check/openstack-tox-pep8/e89a412/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/6/check/openstack-tox-py27/577de3a/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/6/check/openstack-tox-py36/1106047/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/6/check/openstack-tox-py37/7950f04/ : SUCCESS in 5m 27s\n- openstack-tox-functional http://logs.openstack.org/91/665691/6/check/openstack-tox-functional/30e9587/ : SUCCESS in 4m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/6/check/openstack-tox-functional-py36/3ce928c/ : SUCCESS in 5m 18s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/6/check/placement-nova-tox-functional-py36/0344083/ : SUCCESS in 11m 20s\n- placement-perfload http://logs.openstack.org/91/665691/6/check/placement-perfload/0baa767/ : SUCCESS in 7m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4418cd55feff450f76508bb66835ca5db1363f7a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-06-19 07:26:55.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"fb98c1494695083a8873cebc3ba23409ba650d98","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-19 10:17:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":6},{"id":"91e18332dea709a0d207c6fe5c5891feb370381c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-19 23:27:07.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"a34fc3b0dc13a566b95ad6f6d2a2c87b3c93666c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 03:26:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/91/665691/7/check/grenade-py3/9e37e24/ : SUCCESS in 54m 04s\n- tempest-full-py3 http://logs.openstack.org/91/665691/7/check/tempest-full-py3/ddcc9ea/ : SUCCESS in 1h 41m 46s\n- openstack-tox-docs http://logs.openstack.org/91/665691/7/check/openstack-tox-docs/ffe24ed/html/ : SUCCESS in 5m 42s\n- openstack-tox-cover http://logs.openstack.org/91/665691/7/check/openstack-tox-cover/fc9439b/cover/ : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/7/check/openstack-tox-lower-constraints/492dca6/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/7/check/openstack-tox-pep8/a29933a/ : SUCCESS in 4m 41s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/7/check/openstack-tox-py27/103317f/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/7/check/openstack-tox-py36/78e97f6/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/7/check/openstack-tox-py37/7dfc176/ : SUCCESS in 31m 20s\n- openstack-tox-functional http://logs.openstack.org/91/665691/7/check/openstack-tox-functional/8de1924/ : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/7/check/openstack-tox-functional-py36/1027989/ : TIMED_OUT in 30m 57s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/7/check/placement-nova-tox-functional-py36/47552ff/ : SUCCESS in 42m 03s\n- placement-perfload http://logs.openstack.org/91/665691/7/check/placement-perfload/1a0e197/ : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4310c7ceca61f0454d6aeef4d8575fb1bf3e2695","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 03:45:39.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"3570cbb0a756d882e7692633e523499f41e631c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 05:16:27.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/665691/7/check/grenade-py3/a9f685f/ : SUCCESS in 56m 54s\n- tempest-full-py3 http://logs.openstack.org/91/665691/7/check/tempest-full-py3/8914c2f/ : SUCCESS in 1h 28m 08s\n- openstack-tox-docs http://logs.openstack.org/91/665691/7/check/openstack-tox-docs/64a842f/html/ : SUCCESS in 10m 17s\n- openstack-tox-cover http://logs.openstack.org/91/665691/7/check/openstack-tox-cover/41afa16/cover/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/7/check/openstack-tox-lower-constraints/00ca579/ : SUCCESS in 11m 52s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/7/check/openstack-tox-pep8/7f8c7c9/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/7/check/openstack-tox-py27/fb58523/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/7/check/openstack-tox-py36/0e63554/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/7/check/openstack-tox-py37/602c040/ : SUCCESS in 4m 26s\n- openstack-tox-functional http://logs.openstack.org/91/665691/7/check/openstack-tox-functional/8430ace/ : SUCCESS in 4m 41s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/7/check/openstack-tox-functional-py36/0f13d4e/ : SUCCESS in 4m 16s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/7/check/placement-nova-tox-functional-py36/2c6bd06/ : SUCCESS in 10m 46s\n- placement-perfload http://logs.openstack.org/91/665691/7/check/placement-perfload/d2f83c7/ : SUCCESS in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"aba450d49449af93878b79a9cc1b24b65389500e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 13:50:37.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"0b8516875d5d49bb6e31e7809216774be9fb24ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 19:45:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/665691/8/check/grenade-py3/b8e3bbb/ : SUCCESS in 1h 09m 07s\n- tempest-full-py3 http://logs.openstack.org/91/665691/8/check/tempest-full-py3/58d9d3e/ : SUCCESS in 1h 27m 42s\n- openstack-tox-docs http://logs.openstack.org/91/665691/8/check/openstack-tox-docs/6260177/html/ : SUCCESS in 4m 15s\n- openstack-tox-cover http://logs.openstack.org/91/665691/8/check/openstack-tox-cover/403988a/cover/ : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/8/check/openstack-tox-lower-constraints/612efc4/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/8/check/openstack-tox-pep8/4bd3a15/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/8/check/openstack-tox-py27/786848b/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/8/check/openstack-tox-py36/c069abf/ : SUCCESS in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/8/check/openstack-tox-py37/f5f395d/ : SUCCESS in 5m 22s\n- openstack-tox-functional http://logs.openstack.org/91/665691/8/check/openstack-tox-functional/7932fc9/ : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/8/check/openstack-tox-functional-py36/28ee0bd/ : SUCCESS in 5m 24s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/8/check/placement-nova-tox-functional-py36/da985f2/ : SUCCESS in 12m 20s\n- placement-perfload http://logs.openstack.org/91/665691/8/check/placement-perfload/326db06/ : SUCCESS in 7m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"98b4e4dcb8654e5d4d2fa95a6fbd9e41174f67c8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 21:58:35.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"f99f4ac41f02ab6a37b2f22d217bf9811a157538","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 22:27:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8513dcdbb6fe4600692f6d4c4ffdf921d696492c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 22:30:44.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"486a932a1c7782fc30c8be4164c5c2b0960899c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 03:55:51.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/665691/11/check/grenade-py3/1990212/ : SUCCESS in 1h 02m 09s\n- tempest-full-py3 http://logs.openstack.org/91/665691/11/check/tempest-full-py3/3914086/ : SUCCESS in 1h 51m 59s\n- openstack-tox-docs http://logs.openstack.org/91/665691/11/check/openstack-tox-docs/963e8d3/html/ : SUCCESS in 4m 09s\n- openstack-tox-cover http://logs.openstack.org/91/665691/11/check/openstack-tox-cover/a644683/cover/ : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/11/check/openstack-tox-lower-constraints/b1b30c0/ : SUCCESS in 4m 45s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/11/check/openstack-tox-pep8/719cb61/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/11/check/openstack-tox-py27/f8ffef4/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/11/check/openstack-tox-py36/0d65ee8/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/11/check/openstack-tox-py37/7ff0feb/ : SUCCESS in 4m 42s\n- openstack-tox-functional http://logs.openstack.org/91/665691/11/check/openstack-tox-functional/af7c332/ : SUCCESS in 4m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/11/check/openstack-tox-functional-py36/1b1d9a8/ : SUCCESS in 4m 37s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/11/check/placement-nova-tox-functional-py36/6c9e470/ : SUCCESS in 14m 20s\n- placement-perfload http://logs.openstack.org/91/665691/11/check/placement-perfload/0308452/ : SUCCESS in 6m 43s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"cec5a0776cff526ae0e2ddca7def146dbccef8b4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-25 13:34:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9343b13b97e49fc525b2bc3b12a002b55863f765","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 03:09:08.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/91/665691/12/check/grenade-py3/7ad337a/ : SUCCESS in 59m 07s\n- tempest-full-py3 http://logs.openstack.org/91/665691/12/check/tempest-full-py3/8e27d47/ : SUCCESS in 1h 42m 18s\n- openstack-tox-docs http://logs.openstack.org/91/665691/12/check/openstack-tox-docs/5c67f27/html/ : SUCCESS in 3m 57s\n- openstack-tox-cover http://logs.openstack.org/91/665691/12/check/openstack-tox-cover/e9beb5b/cover/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/12/check/openstack-tox-lower-constraints/abe2746/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/12/check/openstack-tox-pep8/799f0b1/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/12/check/openstack-tox-py27/75d1b4a/ : SUCCESS in 4m 08s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/12/check/openstack-tox-py36/421632a/ : SUCCESS in 4m 10s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/12/check/openstack-tox-py37/80c213b/ : SUCCESS in 4m 50s\n- openstack-tox-functional http://logs.openstack.org/91/665691/12/check/openstack-tox-functional/3052924/ : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/12/check/openstack-tox-functional-py36/1c1867f/ : SUCCESS in 4m 18s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/12/check/placement-nova-tox-functional-py36/4ea130b/ : FAILURE in 11m 18s\n- placement-perfload http://logs.openstack.org/91/665691/12/check/placement-perfload/30a4640/ : SUCCESS in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"df47d7fa36a2103341a7a14c51c6d6221a907b6f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-26 05:03:32.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"0b1bdb6f5c789a551506cd953a1397048ffd6c04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 14:11:24.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/91/665691/12/check/grenade-py3/8ec2967/ : SUCCESS in 1h 03m 09s\n- tempest-full-py3 http://logs.openstack.org/91/665691/12/check/tempest-full-py3/5fc8181/ : SUCCESS in 1h 28m 20s\n- openstack-tox-docs http://logs.openstack.org/91/665691/12/check/openstack-tox-docs/c04909f/html/ : SUCCESS in 4m 06s\n- openstack-tox-cover http://logs.openstack.org/91/665691/12/check/openstack-tox-cover/82ada1d/cover/ : SUCCESS in 8m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/12/check/openstack-tox-lower-constraints/c12d305/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/12/check/openstack-tox-pep8/b1e926e/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/12/check/openstack-tox-py27/ee37809/ : SUCCESS in 4m 15s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/12/check/openstack-tox-py36/201b78d/ : SUCCESS in 4m 16s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/12/check/openstack-tox-py37/d00759a/ : SUCCESS in 5m 18s\n- openstack-tox-functional http://logs.openstack.org/91/665691/12/check/openstack-tox-functional/e87437b/ : SUCCESS in 4m 34s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/12/check/openstack-tox-functional-py36/ab516da/ : SUCCESS in 8m 35s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/12/check/placement-nova-tox-functional-py36/1d2a86c/ : FAILURE in 15m 28s\n- placement-perfload http://logs.openstack.org/91/665691/12/check/placement-perfload/776664d/ : SUCCESS in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4e68019aaa4233fedffab5d3fc471bead5843805","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-26 15:32:14.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"d23ccde7ff61765ce7d2320a4813e86327acc143","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-26 15:33:40.000000000","message":"Patch Set 12: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":12},{"id":"70ffec852bcd74f62c32c1360db653b0fd406b14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 03:16:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/665691/12/check/grenade-py3/a54091d/ : SUCCESS in 1h 14m 45s\n- tempest-full-py3 http://logs.openstack.org/91/665691/12/check/tempest-full-py3/156084c/ : SUCCESS in 1h 39m 25s\n- openstack-tox-docs http://logs.openstack.org/91/665691/12/check/openstack-tox-docs/1869597/html/ : SUCCESS in 4m 08s\n- openstack-tox-cover http://logs.openstack.org/91/665691/12/check/openstack-tox-cover/3cbae17/cover/ : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/12/check/openstack-tox-lower-constraints/813aad2/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/12/check/openstack-tox-pep8/8170fc9/ : SUCCESS in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/12/check/openstack-tox-py27/d9139b2/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/12/check/openstack-tox-py36/67b20a9/ : SUCCESS in 6m 25s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/12/check/openstack-tox-py37/dd81c87/ : SUCCESS in 4m 51s\n- openstack-tox-functional http://logs.openstack.org/91/665691/12/check/openstack-tox-functional/eb402ca/ : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/12/check/openstack-tox-functional-py36/3f07f76/ : SUCCESS in 5m 02s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/12/check/placement-nova-tox-functional-py36/2d2f216/ : SUCCESS in 10m 52s\n- placement-perfload http://logs.openstack.org/91/665691/12/check/placement-perfload/ac8b3bc/ : SUCCESS in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0ac93d3fd3507ec9290271809959332d3666e3fe","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-06-27 04:12:34.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"737051985377d0cc1f176e9ec67a5b5886f90ae2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 04:12:46.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"ec6e8853cf73cb5f8090f55184bbc78739a42f6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 05:47:50.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/91/665691/12/gate/grenade-py3/06b490d/ : SUCCESS in 1h 00m 37s\n- tempest-full-py3 http://logs.openstack.org/91/665691/12/gate/tempest-full-py3/f0bc27e/ : SUCCESS in 1h 28m 22s\n- openstack-tox-docs http://logs.openstack.org/91/665691/12/gate/openstack-tox-docs/e616a75/html/ : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/665691/12/gate/openstack-tox-lower-constraints/a7ffcb1/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/91/665691/12/gate/openstack-tox-pep8/df63f52/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/91/665691/12/gate/openstack-tox-py27/89c37be/ : SUCCESS in 4m 33s\n- openstack-tox-py36 http://logs.openstack.org/91/665691/12/gate/openstack-tox-py36/bce76a6/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/91/665691/12/gate/openstack-tox-py37/146e467/ : SUCCESS in 4m 35s\n- openstack-tox-functional http://logs.openstack.org/91/665691/12/gate/openstack-tox-functional/f8c3014/ : SUCCESS in 4m 38s\n- openstack-tox-functional-py36 http://logs.openstack.org/91/665691/12/gate/openstack-tox-functional-py36/3d8ebd6/ : SUCCESS in 6m 08s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/91/665691/12/gate/placement-nova-tox-functional-py36/ebcad76/ : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":12},{"id":"db3598a2d289e8a3771e113755c946d513900dff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 05:47:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"698645ca5ceaf273b93fc7129553e9e22cec85d1","revisions":{"5aa59158d802e05077a8971a0703a27b4ed336c2":{"kind":"REWORK","_number":1,"created":"2019-06-17 14:32:18.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/1"}}},"commit":{"parents":[{"commit":"b694c6077a7ad9dfefc9c52f1e86aa3d073cd1b0","subject":"WIP: Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b694c6077a7ad9dfefc9c52f1e86aa3d073cd1b0"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 14:24:20.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 14:24:20.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5aa59158d802e05077a8971a0703a27b4ed336c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5aa59158d802e05077a8971a0703a27b4ed336c2"}]},"branch":"refs/heads/master"},"00d696153810d110571d807cd9a864480f33c184":{"kind":"NO_CHANGE","_number":2,"created":"2019-06-17 15:26:49.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/2"}}},"commit":{"parents":[{"commit":"17eabb9968449c9a9ac322de44b9e7d37747cae8","subject":"WIP: Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/17eabb9968449c9a9ac322de44b9e7d37747cae8"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/00d696153810d110571d807cd9a864480f33c184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/00d696153810d110571d807cd9a864480f33c184"}]},"branch":"refs/heads/master"},"e39efb971a6f666db55d5e1b341ccab9e18cc924":{"kind":"REWORK","_number":3,"created":"2019-06-17 17:46:20.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/3"}}},"commit":{"parents":[{"commit":"b4066967a4e6efcc5e52e975a0aa121dcf83b867","subject":"WIP: Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b4066967a4e6efcc5e52e975a0aa121dcf83b867"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 17:45:53.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e39efb971a6f666db55d5e1b341ccab9e18cc924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e39efb971a6f666db55d5e1b341ccab9e18cc924"}]},"branch":"refs/heads/master"},"e800c5fc3e2b3b7b78700a3e9da07e30ca1348b3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-17 22:43:55.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/4"}}},"commit":{"parents":[{"commit":"d422a27a958d833b965e3dc262a58a95f91833e7","subject":"WIP: Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d422a27a958d833b965e3dc262a58a95f91833e7"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 22:43:14.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e800c5fc3e2b3b7b78700a3e9da07e30ca1348b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e800c5fc3e2b3b7b78700a3e9da07e30ca1348b3"}]},"branch":"refs/heads/master"},"b34b46ff78740ab67df95ac9fe34846111385fe0":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-18 14:34:04.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/5"}}},"commit":{"parents":[{"commit":"864648ab2e7f918747d44c4b608cbd16980d8abc","subject":"WIP: Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/864648ab2e7f918747d44c4b608cbd16980d8abc"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-18 14:34:00.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b34b46ff78740ab67df95ac9fe34846111385fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b34b46ff78740ab67df95ac9fe34846111385fe0"}]},"branch":"refs/heads/master"},"e11d13a0d963bdd7f59db17f887e795fba8832c9":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-06-18 20:34:39.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/6"}}},"commit":{"parents":[{"commit":"924e7fcd0e4898d48b19f25990c67c323a28202f","subject":"Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/924e7fcd0e4898d48b19f25990c67c323a28202f"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-18 20:34:23.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e11d13a0d963bdd7f59db17f887e795fba8832c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e11d13a0d963bdd7f59db17f887e795fba8832c9"}]},"branch":"refs/heads/master"},"38f61dd625731804513fb0f3150522ff876a3875":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-06-19 23:27:07.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/7"}}},"commit":{"parents":[{"commit":"bdf68ca2e571f8381b5021c11cae5d6e0382176f","subject":"Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bdf68ca2e571f8381b5021c11cae5d6e0382176f"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-19 23:27:03.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/38f61dd625731804513fb0f3150522ff876a3875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/38f61dd625731804513fb0f3150522ff876a3875"}]},"branch":"refs/heads/master"},"72f0f673117cba4dc1a7c86f7b5b769928fe7158":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-06-20 13:50:37.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/8"}}},"commit":{"parents":[{"commit":"777e25b11f6ea2ffde2da7d52ee085c4ee5e5156","subject":"Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/777e25b11f6ea2ffde2da7d52ee085c4ee5e5156"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-20 13:50:31.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/72f0f673117cba4dc1a7c86f7b5b769928fe7158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/72f0f673117cba4dc1a7c86f7b5b769928fe7158"}]},"branch":"refs/heads/master"},"bcb4f51b400819856b2978604c0f5dc92912823c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-06-20 21:58:35.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/9"}}},"commit":{"parents":[{"commit":"cbecaeb40190539405695288faed743fcd37480c","subject":"Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cbecaeb40190539405695288faed743fcd37480c"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-20 21:54:24.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bcb4f51b400819856b2978604c0f5dc92912823c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bcb4f51b400819856b2978604c0f5dc92912823c"}]},"branch":"refs/heads/master"},"e87b4df1eda8f53092c38f25d1bab65870e822ff":{"kind":"REWORK","_number":10,"created":"2019-06-20 22:27:33.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/10"}}},"commit":{"parents":[{"commit":"cbecaeb40190539405695288faed743fcd37480c","subject":"Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cbecaeb40190539405695288faed743fcd37480c"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-20 22:27:09.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e87b4df1eda8f53092c38f25d1bab65870e822ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e87b4df1eda8f53092c38f25d1bab65870e822ff"}]},"branch":"refs/heads/master"},"8108d65ad00e8c7f4c0eb2feaec9e579c2b1947c":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-06-20 22:30:44.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/11"}}},"commit":{"parents":[{"commit":"b733786a0a5bca5b56b9ace5b75e23365f10d9ce","subject":"Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b733786a0a5bca5b56b9ace5b75e23365f10d9ce"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-20 22:30:37.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8108d65ad00e8c7f4c0eb2feaec9e579c2b1947c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8108d65ad00e8c7f4c0eb2feaec9e579c2b1947c"}]},"branch":"refs/heads/master"},"698645ca5ceaf273b93fc7129553e9e22cec85d1":{"kind":"REWORK","_number":12,"created":"2019-06-25 13:34:21.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/91/665691/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/91/665691/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/91/665691/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/91/665691/12"}}},"commit":{"parents":[{"commit":"b733786a0a5bca5b56b9ace5b75e23365f10d9ce","subject":"Microversion 1.35: root_required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b733786a0a5bca5b56b9ace5b75e23365f10d9ce"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:26:02.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-25 13:33:28.000000000","tz":-300},"subject":"Miscellaneous doc/comment/log cleanups","message":"Miscellaneous doc/comment/log cleanups\n\nThis commit is a grab-bag of tweaks and cleanups noticed while working\nin the neighborhood on something unrelated.\n\n- Add a descriptive comment to the ResourceProviderNotFound exception.\n  This is a marker exception used for inter-method communication.\n- Add and enhance debug logs in the in_tree$S filter code path.\n- Correct a SQL comment in get_providers_with_resource (missed when the\n  code was tweaked).\n- Removes a TODO about an optimization that became obsolete when\n  RequestGroupSearchContext was integrated into\n  get_provider_ids_matching.\n- Corrects the get_sharing_providers docstring s/list/set/ to reflect\n  reality.\n- Changes a link in the provider-tree doc from the spec-in-progress to\n  the merged and published version.\n\nChange-Id: Id433b0540e2051ab27b2c22cc14c62105555ea8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/698645ca5ceaf273b93fc7129553e9e22cec85d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/698645ca5ceaf273b93fc7129553e9e22cec85d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
