)]}'
{"id":"openstack%2Fplacement~665695","triplet_id":"openstack%2Fplacement~master~I617161fde5b844d7f52dc766f85c1b9f1b139e4a","project":"openstack/placement","branch":"master","topic":"story/2005443","hashtags":[],"change_id":"I617161fde5b844d7f52dc766f85c1b9f1b139e4a","subject":"Nested provider performance testing","status":"MERGED","created":"2019-06-17 14:56:00.000000000","updated":"2019-07-01 11:55:48.000000000","submitted":"2019-06-28 09:05:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":54,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665695-1561712717523-e6a449f5","meta_rev_id":"ebb438a1f078f00bfe7b2eb2f7f1f69e5775e873","_number":665695,"virtual_id_number":665695,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"date":"2019-06-20 21:33:03.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-28 09:05:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-06-20 12:11:48.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-06-28 07:28:29.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-06-28 07:28:29.000000000","_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 12:11:48.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-28 07:28:29.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-06-28 09:05:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e4d3bdfd905790a1821a8424f8cf8877b7f8b80","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-17 14:56:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"709f290b2f77c4f9281bb48a102ad1bf3f6946e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:56:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/placement on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027templates\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n      check:\n        jobs:\n          - placement-nested-perfload:\n              voting: false\n          - placement-perfload:\n              voting: false\n      gate:\n        jobs:\n          - openstack-tox-functional\n          - openstack-tox-functional-py36\n\n  in \"openstack/placement/.zuul.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":1},{"id":"368887acf58e4f4bdfaeb41f0e9a723a78b90258","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-17 14:56:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1efd8a3332b9bc897ddd12e4fc39bcdaf6419343","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:56:54.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/placement on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027templates\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n      check:\n        jobs:\n          - placement-nested-perfload:\n              voting: false\n          - placement-perfload:\n              voting: false\n      gate:\n        jobs:\n          - openstack-tox-functional\n          - openstack-tox-functional-py36\n\n  in \"openstack/placement/.zuul.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":2},{"id":"0ae6ed36a9b628f1e1e91f79331f99570216cde5","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-17 14:57:10.000000000","message":"Patch Set 2: Workflow-1\n\nthis is going to take several rounds\n\nonce it is working I\u0027ll dedupe with the other perfload","accounts_in_message":[],"_revision_number":2},{"id":"974a0095b495b07fba2e48a9219263f01bfea71f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-17 14:57:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ce6b55fa0100fbaeffa68e40c892133440458e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:57:54.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/placement on branch master.  The error was:\n\n  Job placement-nested-perfload not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - placement-nested-perfload:\n              voting: false\n          - placement-perfload:\n              voting: false\n      gate:\n        jobs:\n          - openstack-tox-functional\n          - openstack-tox-functional-py36\n\n  in \"openstack/placement/.zuul.yaml@master\", line 3, column 3","accounts_in_message":[],"_revision_number":3},{"id":"69b411c97ca32efaf950fde06b7ebb1dccabd2a8","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-17 14:59:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f11e8719b7d9c3561a7795d84c7d8b9e080ae58a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:51:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- placement-nested-perfload http://logs.openstack.org/95/665695/4/check/placement-nested-perfload/ea0499d/ : FAILURE in 2m 37s (non-voting)\n- placement-perfload http://logs.openstack.org/95/665695/4/check/placement-perfload/feb57e9/ : SUCCESS in 7m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9850b9e9705d820a6b908c0da90de2535565a03d","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-17 20:10:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b2fabb788972301af4cb87c720d8361f19a3da6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:27:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- placement-nested-perfload http://logs.openstack.org/95/665695/5/check/placement-nested-perfload/95ec953/ : SUCCESS in 3m 25s (non-voting)\n- placement-perfload http://logs.openstack.org/95/665695/5/check/placement-perfload/0cb7233/ : SUCCESS in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7b3cc4203bb0a07a947294b63e9c6b6e8397bb91","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-18 08:12:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"73a45fb24708fffb7b8b44f9193b265dd8e8ee1f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-18 10:59:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"007b2f318d19a06779a5ae15cdba25900b018c80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:10:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- placement-nested-perfload http://logs.openstack.org/95/665695/7/check/placement-nested-perfload/e89133d/ : SUCCESS in 4m 38s (non-voting)\n- placement-perfload http://logs.openstack.org/95/665695/7/check/placement-perfload/a589d36/ : SUCCESS in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b8894273f8b44b7580b4d31edfb6802e7115b8ea","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-19 09:46:04.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f5ca023377373c3b4c28a210c7f4df904ae7fe2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 16:23:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- placement-nested-perfload http://logs.openstack.org/95/665695/8/check/placement-nested-perfload/6617312/ : SUCCESS in 19m 36s (non-voting)\n- placement-perfload http://logs.openstack.org/95/665695/8/check/placement-perfload/037d1d1/ : SUCCESS in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"30138b34e006984e9220797442dcca4c7466a083","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-19 16:51:46.000000000","message":"Patch Set 8:\n\n(5 comments)\n\nI\u0027ve left some comments within to help guide people\u0027s thoughts while reviewing this. It\u0027s something we\u0027ll want to iterate to perfection.\n\nIn addition to what\u0027s within:\n\n* don\u0027t merge this until .zuul.yaml is up to date. I don\u0027t want to take that back to normal until we know this is right, to not waste CI (it took more than 6 hours to get a node for this), but plenty of people won\u0027t review something that is -W or marked as a WIP...\n\n* Getting nested candidates, even from a simple topology, is considerably slower than non-nested. This isn\u0027t a surprise but it also isn\u0027t great. Each child boost the cost. Ripe for profiling, which I haven\u0027t had an opportunity to do, yet.","accounts_in_message":[],"_revision_number":8},{"id":"7151c01355c1671626fd8b39860016bba2a204f0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-19 19:07:10.000000000","message":"Patch Set 8: Code-Review+2 Workflow-1\n\n(13 comments)\n\nThis seems entirely sane to me.\n\nIf we wanted to move this to the experimental queue, we could merge it with much less concern for gate hogging, if that makes iterating easier.\n\nSome questions and suggestions inline, but nothing that couldn\u0027t wait for a fup.\n\nI think this is good enough to go (+2) other than .zuul.yaml (-W).","accounts_in_message":[],"_revision_number":8},{"id":"201bf3cbb3d873955e6b5302c283748b61e9e572","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-20 08:40:03.000000000","message":"Patch Set 8:\n\n(10 comments)\n\nsome responses within, will come back with more and a revision later in the day","accounts_in_message":[],"_revision_number":8},{"id":"430868d55b4c152358fe5da245610d91796ed627","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 09:03:42.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d324f554320a3ad6fb90da89f0cab9034a6afb56","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-20 11:39:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"42570b7b0b4f6315231a8feb945204966e7ecc31","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 12:11:48.000000000","message":"Patch Set 9: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"93c3880a8a7ba5720017dfd136546b0138384e07","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-20 12:23:10.000000000","message":"Patch Set 9:\n\n(4 comments)\n\nresponses within, which mostly boils down to \"yeah, will come in future tidies\"","accounts_in_message":[],"_revision_number":9},{"id":"c8a2f070bdcb35badd30d1c7b81cc9da877c5cae","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 12:34:43.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"98ba1a9d18e13ce729ed74f272057408cb99de26","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-20 14:37:53.000000000","message":"Patch Set 9:\n\n3 * cpu looks like it may be less than ideal: takes about 12 minutes: http://logs.openstack.org/95/665695/9/check/placement-nested-perfload/819907b/job-output.txt.gz#_2019-06-20_14_15_08_403733\n\nThe machine has nproc of 8: http://logs.openstack.org/95/665695/9/check/placement-nested-perfload/819907b/zuul-info/host-info.ubuntu-bionic.yaml\n\nsince mysql and uwsgi have to fit in there too, making it 1 might make sense. I\u0027ll add fup where we can see","accounts_in_message":[],"_revision_number":9},{"id":"09a3b5370df84a0a6b67511d20ebad0fb7bf1071","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 15:43:27.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/95/665695/9/check/grenade-py3/a8f3f93/ : SUCCESS in 1h 11m 16s\n- tempest-full-py3 http://logs.openstack.org/95/665695/9/check/tempest-full-py3/c50db3a/ : SUCCESS in 1h 31m 30s\n- openstack-tox-docs http://logs.openstack.org/95/665695/9/check/openstack-tox-docs/588ce41/html/ : SUCCESS in 5m 19s\n- openstack-tox-cover http://logs.openstack.org/95/665695/9/check/openstack-tox-cover/068dc61/cover/ : SUCCESS in 7m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/665695/9/check/openstack-tox-lower-constraints/b8b7f8a/ : SUCCESS in 6m 59s\n- openstack-tox-pep8 http://logs.openstack.org/95/665695/9/check/openstack-tox-pep8/96b17bc/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/95/665695/9/check/openstack-tox-py27/fcdf6ae/ : SUCCESS in 7m 10s\n- openstack-tox-py36 http://logs.openstack.org/95/665695/9/check/openstack-tox-py36/174ddd6/ : SUCCESS in 11m 16s\n- openstack-tox-py37 http://logs.openstack.org/95/665695/9/check/openstack-tox-py37/53c3c9c/ : SUCCESS in 4m 50s\n- openstack-tox-functional http://logs.openstack.org/95/665695/9/check/openstack-tox-functional/2a862d1/ : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 http://logs.openstack.org/95/665695/9/check/openstack-tox-functional-py36/5b3187d/ : SUCCESS in 5m 08s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/95/665695/9/check/placement-nova-tox-functional-py36/fe0b655/ : FAILURE in 12m 55s\n- placement-nested-perfload http://logs.openstack.org/95/665695/9/check/placement-nested-perfload/819907b/ : SUCCESS in 16m 06s (non-voting)\n- placement-perfload http://logs.openstack.org/95/665695/9/check/placement-perfload/ea44dd7/ : SUCCESS in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2baba94419004c26a426104d87d1f3b38c60be32","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-20 21:33:03.000000000","message":"Patch Set 9:\n\nrecheck nova weird","accounts_in_message":[],"_revision_number":9},{"id":"7a50a1fafd1e28482430a07feb86d93c63f3ecd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 02:55:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/95/665695/9/check/grenade-py3/67dfa2a/ : SUCCESS in 1h 03m 52s\n- tempest-full-py3 http://logs.openstack.org/95/665695/9/check/tempest-full-py3/8b9492a/ : SUCCESS in 1h 33m 22s\n- openstack-tox-docs http://logs.openstack.org/95/665695/9/check/openstack-tox-docs/0b493de/html/ : SUCCESS in 3m 47s\n- openstack-tox-cover http://logs.openstack.org/95/665695/9/check/openstack-tox-cover/c4622ee/cover/ : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/665695/9/check/openstack-tox-lower-constraints/9438170/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/95/665695/9/check/openstack-tox-pep8/467d1dd/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/95/665695/9/check/openstack-tox-py27/ee18898/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/95/665695/9/check/openstack-tox-py36/a20126c/ : SUCCESS in 5m 36s\n- openstack-tox-py37 http://logs.openstack.org/95/665695/9/check/openstack-tox-py37/64e9d58/ : SUCCESS in 4m 31s\n- openstack-tox-functional http://logs.openstack.org/95/665695/9/check/openstack-tox-functional/a41b681/ : SUCCESS in 11m 52s\n- openstack-tox-functional-py36 http://logs.openstack.org/95/665695/9/check/openstack-tox-functional-py36/c6a278e/ : SUCCESS in 4m 54s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/95/665695/9/check/placement-nova-tox-functional-py36/6659932/ : SUCCESS in 12m 48s\n- placement-nested-perfload http://logs.openstack.org/95/665695/9/check/placement-nested-perfload/a5a6a6a/ : SUCCESS in 18m 37s (non-voting)\n- placement-perfload http://logs.openstack.org/95/665695/9/check/placement-perfload/6a729a8/ : SUCCESS in 7m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"76b7abca7665b63d816cd01bc0b08c311142d9f1","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-06-28 07:28:29.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(5 comments)\n\nThank you very much for doing this! Works for me.\nI left some corrections in code comments but it looks like one of them is also related to the following patch so another follow up is a good way to go.\nAggregate thing maybe a separate independent thing though.","accounts_in_message":[],"_revision_number":9},{"id":"e3b7b66ae0a14a1baef610ff2c37078cbec8c506","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 07:28:48.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"3d02ce41526b24aad0d0aa6e0330fb1bdede3e72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 09:05:17.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/95/665695/9/gate/grenade-py3/7be14bb/ : SUCCESS in 1h 01m 08s\n- tempest-full-py3 http://logs.openstack.org/95/665695/9/gate/tempest-full-py3/f0f7442/ : SUCCESS in 1h 26m 08s\n- openstack-tox-docs http://logs.openstack.org/95/665695/9/gate/openstack-tox-docs/c772b60/html/ : SUCCESS in 3m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/665695/9/gate/openstack-tox-lower-constraints/ac110a1/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/95/665695/9/gate/openstack-tox-pep8/5a7d142/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/95/665695/9/gate/openstack-tox-py27/9705912/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/95/665695/9/gate/openstack-tox-py36/7bd3ead/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/95/665695/9/gate/openstack-tox-py37/731ad34/ : SUCCESS in 5m 05s\n- openstack-tox-functional http://logs.openstack.org/95/665695/9/gate/openstack-tox-functional/1603146/ : SUCCESS in 4m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/95/665695/9/gate/openstack-tox-functional-py36/1b737bf/ : SUCCESS in 4m 18s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/95/665695/9/gate/placement-nova-tox-functional-py36/2654daf/ : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":9},{"id":"cd418979307d1f9bd70eca6367a7a549268b5a21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 09:05:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"c6daa8cd27998c7ad56a0857579251baaa5c29c4","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-01 11:55:48.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nI\u0027ve done some of the minor cleanups in https://review.opendev.org/668449","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"8723bd7772f2b90deceaf0887951f90d0dc0346d","revisions":{"7837952d402310dc4ac0fcc5a59c913919005a8f":{"kind":"REWORK","_number":1,"created":"2019-06-17 14:56:00.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/95/665695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/95/665695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/95/665695/1"}}},"commit":{"parents":[{"commit":"c2df483bec79cbec7575064e298939abf9407c31","subject":"Merge \"Remove a redundant test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c2df483bec79cbec7575064e298939abf9407c31"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:47:36.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:55:52.000000000","tz":60},"subject":"WIP: Nested performance testing","message":"WIP: Nested performance testing\n\nThis change duplicates the ideas started in with the placement-perfload\njob and builds on it to create a set of nested trees that can be\nexercised.\n\nIn placement-perfload placeload is used to create the provides. This\nproves to be cumbersome for nested topologies so this change starts\na new model: Using parallel [1] plus instrumented gabbi to create\nnested topologies in a declarative fashion.\n\ngate/perfload-server.sh sets up placement db and starts a uwsgi server.\n\ngate/perfload-nested-loader.sh is called in the playbook to get gabbi\nto create the nested topology described in\ngate/gabbits/nested-perfload.yaml. That topology is intentionally very\nnaive right now as there\u0027s no point getting too complicated until it\nactually works, which will surely take a few iterations.\n\nZuul is limited to just the perfload jobs while iterating.\n\nChange-Id: I617161fde5b844d7f52dc766f85c1b9f1b139e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7837952d402310dc4ac0fcc5a59c913919005a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7837952d402310dc4ac0fcc5a59c913919005a8f"}]},"branch":"refs/heads/master"},"eb3f821bc699628006d742ff6573fe2703b99a86":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-17 14:56:39.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/95/665695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/95/665695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/95/665695/2"}}},"commit":{"parents":[{"commit":"c2df483bec79cbec7575064e298939abf9407c31","subject":"Merge \"Remove a redundant test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c2df483bec79cbec7575064e298939abf9407c31"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:47:36.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:56:19.000000000","tz":60},"subject":"WIP: Nested performance testing","message":"WIP: Nested performance testing\n\nThis change duplicates the ideas started in with the placement-perfload\njob and builds on it to create a set of nested trees that can be\nexercised.\n\nIn placement-perfload placeload is used to create the provides. This\nproves to be cumbersome for nested topologies so this change starts\na new model: Using parallel [1] plus instrumented gabbi to create\nnested topologies in a declarative fashion.\n\ngate/perfload-server.sh sets up placement db and starts a uwsgi server.\n\ngate/perfload-nested-loader.sh is called in the playbook to get gabbi\nto create the nested topology described in\ngate/gabbits/nested-perfload.yaml. That topology is intentionally very\nnaive right now as there\u0027s no point getting too complicated until it\nactually works, which will surely take a few iterations.\n\nZuul is limited to just the perfload jobs while iterating.\n\nStory: 2005443\nTask: 30487\nChange-Id: I617161fde5b844d7f52dc766f85c1b9f1b139e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/eb3f821bc699628006d742ff6573fe2703b99a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/eb3f821bc699628006d742ff6573fe2703b99a86"}]},"branch":"refs/heads/master"},"3cc6fd7e3d90c87654b7975b9ca7d9111b50c9d1":{"kind":"REWORK","_number":3,"created":"2019-06-17 14:57:39.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/95/665695/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/95/665695/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/95/665695/3"}}},"commit":{"parents":[{"commit":"c2df483bec79cbec7575064e298939abf9407c31","subject":"Merge \"Remove a redundant test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c2df483bec79cbec7575064e298939abf9407c31"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:47:36.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:57:30.000000000","tz":60},"subject":"WIP: Nested performance testing","message":"WIP: Nested performance testing\n\nThis change duplicates the ideas started in with the placement-perfload\njob and builds on it to create a set of nested trees that can be\nexercised.\n\nIn placement-perfload placeload is used to create the provides. This\nproves to be cumbersome for nested topologies so this change starts\na new model: Using parallel [1] plus instrumented gabbi to create\nnested topologies in a declarative fashion.\n\ngate/perfload-server.sh sets up placement db and starts a uwsgi server.\n\ngate/perfload-nested-loader.sh is called in the playbook to get gabbi\nto create the nested topology described in\ngate/gabbits/nested-perfload.yaml. That topology is intentionally very\nnaive right now as there\u0027s no point getting too complicated until it\nactually works, which will surely take a few iterations.\n\nZuul is limited to just the perfload jobs while iterating.\n\nStory: 2005443\nTask: 30487\nChange-Id: I617161fde5b844d7f52dc766f85c1b9f1b139e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3cc6fd7e3d90c87654b7975b9ca7d9111b50c9d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/3cc6fd7e3d90c87654b7975b9ca7d9111b50c9d1"}]},"branch":"refs/heads/master"},"cccf8f9175a7df67aed674ce301280edce1ff292":{"kind":"REWORK","_number":4,"created":"2019-06-17 14:59:19.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/95/665695/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/95/665695/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/95/665695/4"}}},"commit":{"parents":[{"commit":"c2df483bec79cbec7575064e298939abf9407c31","subject":"Merge \"Remove a redundant test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c2df483bec79cbec7575064e298939abf9407c31"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:47:36.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:58:58.000000000","tz":60},"subject":"WIP: Nested performance testing","message":"WIP: Nested performance testing\n\nThis change duplicates the ideas started in with the placement-perfload\njob and builds on it to create a set of nested trees that can be\nexercised.\n\nIn placement-perfload placeload is used to create the provides. This\nproves to be cumbersome for nested topologies so this change starts\na new model: Using parallel [1] plus instrumented gabbi to create\nnested topologies in a declarative fashion.\n\ngate/perfload-server.sh sets up placement db and starts a uwsgi server.\n\ngate/perfload-nested-loader.sh is called in the playbook to get gabbi\nto create the nested topology described in\ngate/gabbits/nested-perfload.yaml. That topology is intentionally very\nnaive right now as there\u0027s no point getting too complicated until it\nactually works, which will surely take a few iterations.\n\nZuul is limited to just the perfload jobs while iterating.\n\nStory: 2005443\nTask: 30487\nChange-Id: I617161fde5b844d7f52dc766f85c1b9f1b139e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cccf8f9175a7df67aed674ce301280edce1ff292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/cccf8f9175a7df67aed674ce301280edce1ff292"}]},"branch":"refs/heads/master"},"77db29df0d4272e14d4d10c532dcc0c8ebeedde3":{"kind":"REWORK","_number":5,"created":"2019-06-17 20:10:32.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/95/665695/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/95/665695/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/95/665695/5"}}},"commit":{"parents":[{"commit":"c2df483bec79cbec7575064e298939abf9407c31","subject":"Merge \"Remove a redundant test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c2df483bec79cbec7575064e298939abf9407c31"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:47:36.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 20:10:23.000000000","tz":60},"subject":"WIP: Nested performance testing","message":"WIP: Nested performance testing\n\nThis change duplicates the ideas started in with the placement-perfload\njob and builds on it to create a set of nested trees that can be\nexercised.\n\nIn placement-perfload placeload is used to create the provides. This\nproves to be cumbersome for nested topologies so this change starts\na new model: Using parallel [1] plus instrumented gabbi to create\nnested topologies in a declarative fashion.\n\ngate/perfload-server.sh sets up placement db and starts a uwsgi server.\n\ngate/perfload-nested-loader.sh is called in the playbook to get gabbi\nto create the nested topology described in\ngate/gabbits/nested-perfload.yaml. That topology is intentionally very\nnaive right now as there\u0027s no point getting too complicated until it\nactually works, which will surely take a few iterations.\n\nZuul is limited to just the perfload jobs while iterating.\n\nStory: 2005443\nTask: 30487\nChange-Id: I617161fde5b844d7f52dc766f85c1b9f1b139e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/77db29df0d4272e14d4d10c532dcc0c8ebeedde3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/77db29df0d4272e14d4d10c532dcc0c8ebeedde3"}]},"branch":"refs/heads/master"},"e049236fb5bc3227b79df6eb5a90d90e7bca2bb5":{"kind":"REWORK","_number":6,"created":"2019-06-18 08:12:05.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/95/665695/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/95/665695/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/95/665695/6"}}},"commit":{"parents":[{"commit":"c2df483bec79cbec7575064e298939abf9407c31","subject":"Merge \"Remove a redundant test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c2df483bec79cbec7575064e298939abf9407c31"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:47:36.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-18 08:11:50.000000000","tz":60},"subject":"WIP: Nested performance testing","message":"WIP: Nested performance testing\n\nThis change duplicates the ideas started in with the placement-perfload\njob and builds on it to create a set of nested trees that can be\nexercised.\n\nIn placement-perfload placeload is used to create the provides. This\nproves to be cumbersome for nested topologies so this change starts\na new model: Using parallel [1] plus instrumented gabbi to create\nnested topologies in a declarative fashion.\n\ngate/perfload-server.sh sets up placement db and starts a uwsgi server.\n\ngate/perfload-nested-loader.sh is called in the playbook to get gabbi\nto create the nested topology described in\ngate/gabbits/nested-perfload.yaml. That topology is intentionally very\nnaive right now as there\u0027s no point getting too complicated until it\nactually works, which will surely take a few iterations.\n\nZuul is limited to just the perfload jobs while iterating.\n\nStory: 2005443\nTask: 30487\nChange-Id: I617161fde5b844d7f52dc766f85c1b9f1b139e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e049236fb5bc3227b79df6eb5a90d90e7bca2bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e049236fb5bc3227b79df6eb5a90d90e7bca2bb5"}]},"branch":"refs/heads/master"},"0c560288cc2c12fd83634bd46d8a4f115d947b73":{"kind":"REWORK","_number":7,"created":"2019-06-18 10:59:24.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/95/665695/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/95/665695/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/95/665695/7"}}},"commit":{"parents":[{"commit":"c2df483bec79cbec7575064e298939abf9407c31","subject":"Merge \"Remove a redundant test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c2df483bec79cbec7575064e298939abf9407c31"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:47:36.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-18 10:59:14.000000000","tz":60},"subject":"WIP: Nested performance testing","message":"WIP: Nested performance testing\n\nThis change duplicates the ideas started in with the placement-perfload\njob and builds on it to create a set of nested trees that can be\nexercised.\n\nIn placement-perfload placeload is used to create the provides. This\nproves to be cumbersome for nested topologies so this change starts\na new model: Using parallel [1] plus instrumented gabbi to create\nnested topologies in a declarative fashion.\n\ngate/perfload-server.sh sets up placement db and starts a uwsgi server.\n\ngate/perfload-nested-loader.sh is called in the playbook to get gabbi\nto create the nested topology described in\ngate/gabbits/nested-perfload.yaml. That topology is intentionally very\nnaive right now as there\u0027s no point getting too complicated until it\nactually works, which will surely take a few iterations.\n\nZuul is limited to just the perfload jobs while iterating.\n\nStory: 2005443\nTask: 30487\nChange-Id: I617161fde5b844d7f52dc766f85c1b9f1b139e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0c560288cc2c12fd83634bd46d8a4f115d947b73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/0c560288cc2c12fd83634bd46d8a4f115d947b73"}]},"branch":"refs/heads/master"},"327419362cf570e3887e67c724a6830b89f389c9":{"kind":"REWORK","_number":8,"created":"2019-06-19 09:46:04.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/95/665695/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/95/665695/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/95/665695/8"}}},"commit":{"parents":[{"commit":"c2df483bec79cbec7575064e298939abf9407c31","subject":"Merge \"Remove a redundant test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c2df483bec79cbec7575064e298939abf9407c31"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:47:36.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-19 09:39:36.000000000","tz":60},"subject":"Nested provider performance testing","message":"Nested provider performance testing\n\nThis change duplicates the ideas started in with the placement-perfload\njob and builds on it to create a set of nested trees that can be\nexercised.\n\nIn placement-perfload placeload is used to create the provides. This\nproves to be cumbersome for nested topologies so this change starts\na new model: Using parallel [1] plus instrumented gabbi to create\nnested topologies in a declarative fashion.\n\ngate/perfload-server.sh sets up placement db and starts a uwsgi server.\n\ngate/perfload-nested-loader.sh is called in the playbook to cause gabbi\nto create the nested topology described in\ngate/gabbits/nested-perfload.yaml. That topology is intentionally very\nnaive right now as there\u0027s no point getting too complicated until it\nactually works, which will surely take a few iterations.\n\nZuul is limited to just the perfload jobs while iterating.\n\nThere\u0027s some duplication between perfload.yaml and\nnested-perfload.yaml that will be cleared up in a followup once we\ndecided that this is working as we like.\n\n[1] https://www.gnu.org/software/parallel/ (although the version on\nubuntu is a non-GPL clone)\n\nStory: 2005443\nTask: 30487\nChange-Id: I617161fde5b844d7f52dc766f85c1b9f1b139e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/327419362cf570e3887e67c724a6830b89f389c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/327419362cf570e3887e67c724a6830b89f389c9"}]},"branch":"refs/heads/master"},"8723bd7772f2b90deceaf0887951f90d0dc0346d":{"kind":"REWORK","_number":9,"created":"2019-06-20 11:39:10.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/95/665695/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/95/665695/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/95/665695/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/95/665695/9"}}},"commit":{"parents":[{"commit":"9de03e1b17ddc202532757bb221ee81f02f89bbe","subject":"Uniquify allocation mappings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9de03e1b17ddc202532757bb221ee81f02f89bbe"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-17 14:47:36.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-06-20 11:37:28.000000000","tz":60},"subject":"Nested provider performance testing","message":"Nested provider performance testing\n\nThis change duplicates the ideas started in with the placement-perfload\njob and builds on it to create a set of nested trees that can be\nexercised.\n\nIn placement-perfload, placeload is used to create the providers. This\nproves to be cumbersome for nested topologies so this change starts\na new model: Using parallel [1] plus instrumented gabbi to create\nnested topologies in a declarative fashion.\n\ngate/perfload-server.sh sets up placement db and starts a uwsgi server.\n\ngate/perfload-nested-loader.sh is called in the playbook to cause gabbi\nto create the nested topology described in\ngate/gabbits/nested-perfload.yaml. That topology is intentionally very\nnaive right now but should be made more realisitc as we continue to\ndevelop nested features.\n\nThere\u0027s some duplication between perfload.yaml and\nnested-perfload.yaml that will be cleared up in a followup.\n\n[1] https://www.gnu.org/software/parallel/ (although the version on\nubuntu is a non-GPL clone)\n\nStory: 2005443\nTask: 30487\nChange-Id: I617161fde5b844d7f52dc766f85c1b9f1b139e4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8723bd7772f2b90deceaf0887951f90d0dc0346d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8723bd7772f2b90deceaf0887951f90d0dc0346d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
