)]}'
{"id":"openstack%2Fplacement~665712","triplet_id":"openstack%2Fplacement~master~I572ec8e6c0f9d81ead49964f41b46ab68a3de9a6","project":"openstack/placement","branch":"master","topic":"story/2005575-33753","hashtags":[],"change_id":"I572ec8e6c0f9d81ead49964f41b46ab68a3de9a6","subject":"RequestWideParams and RequestWideSearchContext","status":"MERGED","created":"2019-06-17 15:26:49.000000000","updated":"2019-06-26 18:43:14.000000000","submitted":"2019-06-26 18:43:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665712-1561574594908-90acf6b2","meta_rev_id":"17c8ec77679d7d9cb930ab9a512fedc5dedf4907","_number":665712,"virtual_id_number":665712,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-21 03:46:53.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-26 18:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-06-20 21:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2019-06-20 21:58:35.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2019-06-20 22:20:06.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 08:59:07.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-06-20 22:20:06.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-26 18:43:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee8b8bd6c15e32cf09454f6763ba4abd2d5bc687","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 15:26:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b8980265dc891a73e3b865330c27374c3ebdd9b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 15:27:48.000000000","message":"Topic set to story/2005575-33753","accounts_in_message":[],"_revision_number":1},{"id":"6d784339f1e82eceeef3848aa11edaa8fd69e2c5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 17:46:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"540cd7d9430733f330600b5664e9824dfcfd09f7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 17:56:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"223fbcc83bba70c1ac5a8a49eb15b97caa96948e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-17 20:34:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nyeah, may as well","accounts_in_message":[],"_revision_number":2},{"id":"22a13fedeabc0b079fc3e062692ced4a70efc41e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 22:24:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/12/665712/2/check/grenade-py3/ed1b768/ : SUCCESS in 1h 03m 49s\n- tempest-full-py3 http://logs.openstack.org/12/665712/2/check/tempest-full-py3/c99b51d/ : SUCCESS in 1h 32m 56s\n- openstack-tox-docs http://logs.openstack.org/12/665712/2/check/openstack-tox-docs/4da5d2e/html/ : SUCCESS in 4m 11s\n- openstack-tox-cover http://logs.openstack.org/12/665712/2/check/openstack-tox-cover/b34f11b/ : FAILURE in 5m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/665712/2/check/openstack-tox-lower-constraints/2c994e3/ : FAILURE in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/12/665712/2/check/openstack-tox-pep8/b8f192b/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/12/665712/2/check/openstack-tox-py27/816f424/ : FAILURE in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/12/665712/2/check/openstack-tox-py36/ce340a1/ : FAILURE in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/12/665712/2/check/openstack-tox-py37/d8d8baf/ : FAILURE in 4m 42s\n- openstack-tox-functional http://logs.openstack.org/12/665712/2/check/openstack-tox-functional/f13f89f/ : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/12/665712/2/check/openstack-tox-functional-py36/be43c49/ : SUCCESS in 6m 33s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/12/665712/2/check/placement-nova-tox-functional-py36/ee1fd85/ : SUCCESS in 10m 44s\n- placement-perfload http://logs.openstack.org/12/665712/2/check/placement-perfload/a574050/ : SUCCESS in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42a5b815a61ea7cbc4a87f3c1bfb043daa201b7b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 22:43:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7be35e57d51514a18d05a2ba67361b7946885163","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-17 22:43:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cb5775f4f41093c9a2be97972a44cdf77aacb4b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 03:01:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/12/665712/3/check/grenade-py3/ebb2bf8/ : SUCCESS in 1h 09m 37s\n- tempest-full-py3 http://logs.openstack.org/12/665712/3/check/tempest-full-py3/9e41061/ : SUCCESS in 1h 53m 12s\n- openstack-tox-docs http://logs.openstack.org/12/665712/3/check/openstack-tox-docs/c6f1615/html/ : SUCCESS in 4m 42s\n- openstack-tox-cover http://logs.openstack.org/12/665712/3/check/openstack-tox-cover/9748fd7/cover/ : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/665712/3/check/openstack-tox-lower-constraints/306751e/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/12/665712/3/check/openstack-tox-pep8/866aa44/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/12/665712/3/check/openstack-tox-py27/7dc3496/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/12/665712/3/check/openstack-tox-py36/04387fc/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/12/665712/3/check/openstack-tox-py37/0364401/ : SUCCESS in 5m 02s\n- openstack-tox-functional http://logs.openstack.org/12/665712/3/check/openstack-tox-functional/412f4c8/ : SUCCESS in 5m 34s\n- openstack-tox-functional-py36 http://logs.openstack.org/12/665712/3/check/openstack-tox-functional-py36/4913e03/ : SUCCESS in 4m 47s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/12/665712/3/check/placement-nova-tox-functional-py36/cdb0f4a/ : SUCCESS in 11m 26s\n- placement-perfload http://logs.openstack.org/12/665712/3/check/placement-perfload/1908a67/ : SUCCESS in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5259331e36565f5f40b4740a31c5cb5b706718c2","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-18 08:16:17.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6ada823c88eec8d83c560a8a3e6d7a979c9aede9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-19 08:59:07.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"02de426bca705d39d81a89c374e8c5e2ba3c8772","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 13:50:37.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"be5581fe8e20ebfb1280f01648a810df28480b3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 19:25:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/12/665712/4/check/grenade-py3/abf8247/ : SUCCESS in 1h 04m 06s\n- tempest-full-py3 http://logs.openstack.org/12/665712/4/check/tempest-full-py3/116e9cf/ : SUCCESS in 1h 19m 55s\n- openstack-tox-docs http://logs.openstack.org/12/665712/4/check/openstack-tox-docs/1611ff3/html/ : SUCCESS in 4m 46s\n- openstack-tox-cover http://logs.openstack.org/12/665712/4/check/openstack-tox-cover/edaf23c/cover/ : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/665712/4/check/openstack-tox-lower-constraints/b10d32f/ : SUCCESS in 5m 09s\n- openstack-tox-pep8 http://logs.openstack.org/12/665712/4/check/openstack-tox-pep8/72d9da7/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/12/665712/4/check/openstack-tox-py27/8e4490d/ : SUCCESS in 11m 54s\n- openstack-tox-py36 http://logs.openstack.org/12/665712/4/check/openstack-tox-py36/bbc4d5b/ : SUCCESS in 4m 06s\n- openstack-tox-py37 http://logs.openstack.org/12/665712/4/check/openstack-tox-py37/a841c31/ : SUCCESS in 12m 23s\n- openstack-tox-functional http://logs.openstack.org/12/665712/4/check/openstack-tox-functional/e06db1b/ : SUCCESS in 4m 58s\n- openstack-tox-functional-py36 http://logs.openstack.org/12/665712/4/check/openstack-tox-functional-py36/d3a0058/ : SUCCESS in 4m 53s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/12/665712/4/check/placement-nova-tox-functional-py36/51ecfa5/ : SUCCESS in 12m 00s\n- placement-perfload http://logs.openstack.org/12/665712/4/check/placement-perfload/f1d2e48/ : SUCCESS in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"40033199975ae4693a67efbb2fb753d6c7d5af3f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-20 21:58:35.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"7b417163869fa7e70ff34cd6653eda3bf9963307","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-20 22:20:06.000000000","message":"Patch Set 5: Workflow+1\n\ngood rebase","accounts_in_message":[],"_revision_number":5},{"id":"d1d16c7c5cf96fd3c0ccc7bf00740aad1ef1e1e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 03:14:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/12/665712/5/check/grenade-py3/499c515/ : FAILURE in 18m 07s\n- tempest-full-py3 http://logs.openstack.org/12/665712/5/check/tempest-full-py3/de6500c/ : SUCCESS in 1h 33m 52s\n- openstack-tox-docs http://logs.openstack.org/12/665712/5/check/openstack-tox-docs/522e373/html/ : SUCCESS in 3m 57s\n- openstack-tox-cover http://logs.openstack.org/12/665712/5/check/openstack-tox-cover/79e40b7/cover/ : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/665712/5/check/openstack-tox-lower-constraints/feb883c/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/12/665712/5/check/openstack-tox-pep8/b3af081/ : SUCCESS in 11m 14s\n- openstack-tox-py27 http://logs.openstack.org/12/665712/5/check/openstack-tox-py27/6f80ba3/ : SUCCESS in 7m 17s\n- openstack-tox-py36 http://logs.openstack.org/12/665712/5/check/openstack-tox-py36/a646de9/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/12/665712/5/check/openstack-tox-py37/7b26d11/ : SUCCESS in 4m 56s\n- openstack-tox-functional http://logs.openstack.org/12/665712/5/check/openstack-tox-functional/16b1a84/ : SUCCESS in 4m 34s\n- openstack-tox-functional-py36 http://logs.openstack.org/12/665712/5/check/openstack-tox-functional-py36/5ee8d59/ : SUCCESS in 6m 31s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/12/665712/5/check/placement-nova-tox-functional-py36/41e6c67/ : SUCCESS in 10m 42s\n- placement-perfload http://logs.openstack.org/12/665712/5/check/placement-perfload/ace4f88/ : SUCCESS in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"66333b12390224dd5d1d7b2ae07e9ad7fc91e245","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-21 03:46:53.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e35b68b377994820feac78e1147dee1a5017b545","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 05:28:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/12/665712/5/check/grenade-py3/11891f2/ : SUCCESS in 57m 31s\n- tempest-full-py3 http://logs.openstack.org/12/665712/5/check/tempest-full-py3/58860b7/ : SUCCESS in 1h 36m 15s\n- openstack-tox-docs http://logs.openstack.org/12/665712/5/check/openstack-tox-docs/88b3090/html/ : SUCCESS in 3m 46s\n- openstack-tox-cover http://logs.openstack.org/12/665712/5/check/openstack-tox-cover/6f21e31/cover/ : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/665712/5/check/openstack-tox-lower-constraints/46c26bd/ : SUCCESS in 11m 28s\n- openstack-tox-pep8 http://logs.openstack.org/12/665712/5/check/openstack-tox-pep8/834b279/ : SUCCESS in 9m 12s\n- openstack-tox-py27 http://logs.openstack.org/12/665712/5/check/openstack-tox-py27/319120f/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/12/665712/5/check/openstack-tox-py36/382e58e/ : SUCCESS in 4m 25s\n- openstack-tox-py37 http://logs.openstack.org/12/665712/5/check/openstack-tox-py37/f477312/ : SUCCESS in 12m 09s\n- openstack-tox-functional http://logs.openstack.org/12/665712/5/check/openstack-tox-functional/ab9918d/ : SUCCESS in 10m 51s\n- openstack-tox-functional-py36 http://logs.openstack.org/12/665712/5/check/openstack-tox-functional-py36/e3a83e3/ : SUCCESS in 4m 08s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/12/665712/5/check/placement-nova-tox-functional-py36/5b9411d/ : SUCCESS in 14m 44s\n- placement-perfload http://logs.openstack.org/12/665712/5/check/placement-perfload/f6781b7/ : SUCCESS in 16m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a21343c07ff606004628ef8da04aba331ce40b04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 15:14:48.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3f5c620623db19c227212c87d7cd0c3bac8469f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 18:43:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"17c8ec77679d7d9cb930ab9a512fedc5dedf4907","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 18:43:14.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/12/665712/5/gate/grenade-py3/8c705c4/ : SUCCESS in 1h 01m 49s\n- tempest-full-py3 http://logs.openstack.org/12/665712/5/gate/tempest-full-py3/dcdf97b/ : SUCCESS in 1h 46m 14s\n- openstack-tox-docs http://logs.openstack.org/12/665712/5/gate/openstack-tox-docs/71c8cf7/html/ : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/12/665712/5/gate/openstack-tox-lower-constraints/0fdc3e6/ : SUCCESS in 7m 19s\n- openstack-tox-pep8 http://logs.openstack.org/12/665712/5/gate/openstack-tox-pep8/ffc31b7/ : SUCCESS in 8m 17s\n- openstack-tox-py27 http://logs.openstack.org/12/665712/5/gate/openstack-tox-py27/e4a8b3c/ : SUCCESS in 7m 46s\n- openstack-tox-py36 http://logs.openstack.org/12/665712/5/gate/openstack-tox-py36/a56d269/ : SUCCESS in 7m 32s\n- openstack-tox-py37 http://logs.openstack.org/12/665712/5/gate/openstack-tox-py37/7989110/ : SUCCESS in 5m 06s\n- openstack-tox-functional http://logs.openstack.org/12/665712/5/gate/openstack-tox-functional/b7aaad0/ : SUCCESS in 8m 28s\n- openstack-tox-functional-py36 http://logs.openstack.org/12/665712/5/gate/openstack-tox-functional-py36/100ffaa/ : SUCCESS in 8m 06s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/12/665712/5/gate/placement-nova-tox-functional-py36/3287ad6/ : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5a6884ae12cab3e9627699b7a8045a0386343e84","revisions":{"9f9b6271ca9ff1e8fce04bbdc7a84b87d4233b2f":{"kind":"REWORK","_number":1,"created":"2019-06-17 15:26:49.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/12/665712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/12/665712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/12/665712/1"}}},"commit":{"parents":[{"commit":"8e4a00d08cb64db5c98db56a817a1772eb54df9c","subject":"Refactor anchors_for_sharing_providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8e4a00d08cb64db5c98db56a817a1772eb54df9c"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:13:07.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:13:07.000000000","tz":-300},"subject":"Introduce RequestWideParams","message":"Introduce RequestWideParams\n\nRecent design discussions have reified the concept that a request to GET\n/allocation_candidates has two kinds of query parameters:\n\n- Those that apply to request groups: resources[$S], required[$S], etc.\n  Essentially all the ones that can be (optionally) suffixed.\n- Those that apply to the request as a whole: limit, group_policy.\n\nGroup-specific parameters are already encapsulated in the\nplacement.lib.RequestGroup class, one instance of which is used to\nrepresent each request group in the query.\n\nThis commit pulls the existing request-wide parameters, limit and\ngroup_policy, into a new class, placement.lib.RequestWideParams.\n\nIn so doing, it adjusts the signatures of methods which used to use\nthose individual parameters to use an instance of RequestWideParams as\nwell.\n\nAt the same time, parameters and local variables referring to the list\nof RequestGroup are renamed from `requests` to `groups` to\ndisambiguate/clarify what they are.\n\nChange-Id: I572ec8e6c0f9d81ead49964f41b46ab68a3de9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9f9b6271ca9ff1e8fce04bbdc7a84b87d4233b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9f9b6271ca9ff1e8fce04bbdc7a84b87d4233b2f"}]},"branch":"refs/heads/master"},"4c24a8ce78fc4c89a19fa681cedba225365d5e99":{"kind":"REWORK","_number":2,"created":"2019-06-17 17:46:20.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/12/665712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/12/665712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/12/665712/2"}}},"commit":{"parents":[{"commit":"2c0407297fc31cfabd1cc6ecb84b16fb4bdbd285","subject":"Refactor anchors_for_sharing_providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2c0407297fc31cfabd1cc6ecb84b16fb4bdbd285"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:13:07.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 17:20:41.000000000","tz":-300},"subject":"RequestWideParams and RequestWideSearchContext","message":"RequestWideParams and RequestWideSearchContext\n\nRecent design discussions have reified the concept that a request to GET\n/allocation_candidates has two kinds of query parameters:\n\n- Those that apply to request groups: resources[$S], required[$S], etc.\n  Essentially all the ones that can be (optionally) suffixed.\n- Those that apply to the request as a whole: limit, group_policy.\n\nGroup-specific parameters are already encapsulated in the\nplacement.lib.RequestGroup class, one instance of which is used to\nrepresent each request group in the query. And then database and other\nfiltering logic is encapsulated in\nplacement.objects.research_context.RequestGroupSearchContext.\n\nThis commit pulls the existing request-wide parameters, limit and\ngroup_policy, into a new class, placement.lib.RequestWideParams; and\nintroduces placement.objects.research_context.RequestWideSearchContext\nto encompass the filtering logic for the same.\n\nIn so doing, we adjust the signatures of methods which used to use those\nindividual parameters.\n\nAt the same time, parameters and local variables referring to the list\nof RequestGroup are renamed from `requests` to `groups` to\ndisambiguate/clarify what they are.\n\nIn the end, this is just a refactor. No logic is changed.\n\nChange-Id: I572ec8e6c0f9d81ead49964f41b46ab68a3de9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4c24a8ce78fc4c89a19fa681cedba225365d5e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/4c24a8ce78fc4c89a19fa681cedba225365d5e99"}]},"branch":"refs/heads/master"},"90ec6ee124fe819ee1c325e8b455febd44e952ac":{"kind":"REWORK","_number":3,"created":"2019-06-17 22:43:55.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/12/665712/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/12/665712/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/12/665712/3"}}},"commit":{"parents":[{"commit":"2c0407297fc31cfabd1cc6ecb84b16fb4bdbd285","subject":"Refactor anchors_for_sharing_providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2c0407297fc31cfabd1cc6ecb84b16fb4bdbd285"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:13:07.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 22:43:12.000000000","tz":-300},"subject":"RequestWideParams and RequestWideSearchContext","message":"RequestWideParams and RequestWideSearchContext\n\nRecent design discussions have reified the concept that a request to GET\n/allocation_candidates has two kinds of query parameters:\n\n- Those that apply to request groups: resources[$S], required[$S], etc.\n  Essentially all the ones that can be (optionally) suffixed.\n- Those that apply to the request as a whole: limit, group_policy.\n\nGroup-specific parameters are already encapsulated in the\nplacement.lib.RequestGroup class, one instance of which is used to\nrepresent each request group in the query. And then database and other\nfiltering logic is encapsulated in\nplacement.objects.research_context.RequestGroupSearchContext.\n\nThis commit pulls the existing request-wide parameters, limit and\ngroup_policy, into a new class, placement.lib.RequestWideParams; and\nintroduces placement.objects.research_context.RequestWideSearchContext\nto encompass the filtering logic for the same.\n\nIn so doing, we adjust the signatures of methods which used to use those\nindividual parameters.\n\nAt the same time, parameters and local variables referring to the list\nof RequestGroup are renamed from `requests` to `groups` to\ndisambiguate/clarify what they are.\n\nIn the end, this is just a refactor. No logic is changed.\n\nChange-Id: I572ec8e6c0f9d81ead49964f41b46ab68a3de9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/90ec6ee124fe819ee1c325e8b455febd44e952ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/90ec6ee124fe819ee1c325e8b455febd44e952ac"}]},"branch":"refs/heads/master"},"9ca32d1194ce4c8aa5ddbcf671e2ba3d75ba32b3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-20 13:50:37.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/12/665712/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/12/665712/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/12/665712/4"}}},"commit":{"parents":[{"commit":"a54e457cf7feb02d17ef7772ff13757c2a7d7905","subject":"Refactor anchors_for_sharing_providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a54e457cf7feb02d17ef7772ff13757c2a7d7905"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:13:07.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-20 13:43:48.000000000","tz":-300},"subject":"RequestWideParams and RequestWideSearchContext","message":"RequestWideParams and RequestWideSearchContext\n\nRecent design discussions have reified the concept that a request to GET\n/allocation_candidates has two kinds of query parameters:\n\n- Those that apply to request groups: resources[$S], required[$S], etc.\n  Essentially all the ones that can be (optionally) suffixed.\n- Those that apply to the request as a whole: limit, group_policy.\n\nGroup-specific parameters are already encapsulated in the\nplacement.lib.RequestGroup class, one instance of which is used to\nrepresent each request group in the query. And then database and other\nfiltering logic is encapsulated in\nplacement.objects.research_context.RequestGroupSearchContext.\n\nThis commit pulls the existing request-wide parameters, limit and\ngroup_policy, into a new class, placement.lib.RequestWideParams; and\nintroduces placement.objects.research_context.RequestWideSearchContext\nto encompass the filtering logic for the same.\n\nIn so doing, we adjust the signatures of methods which used to use those\nindividual parameters.\n\nAt the same time, parameters and local variables referring to the list\nof RequestGroup are renamed from `requests` to `groups` to\ndisambiguate/clarify what they are.\n\nIn the end, this is just a refactor. No logic is changed.\n\nChange-Id: I572ec8e6c0f9d81ead49964f41b46ab68a3de9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9ca32d1194ce4c8aa5ddbcf671e2ba3d75ba32b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9ca32d1194ce4c8aa5ddbcf671e2ba3d75ba32b3"}]},"branch":"refs/heads/master"},"5a6884ae12cab3e9627699b7a8045a0386343e84":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-20 21:58:35.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/12/665712/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/12/665712/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/12/665712/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/12/665712/5"}}},"commit":{"parents":[{"commit":"c774416dd124ab95eeca942a3a8992bcdee3fadf","subject":"Refactor anchors_for_sharing_providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c774416dd124ab95eeca942a3a8992bcdee3fadf"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-17 15:13:07.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-20 19:58:29.000000000","tz":-300},"subject":"RequestWideParams and RequestWideSearchContext","message":"RequestWideParams and RequestWideSearchContext\n\nRecent design discussions have reified the concept that a request to GET\n/allocation_candidates has two kinds of query parameters:\n\n- Those that apply to request groups: resources[$S], required[$S], etc.\n  Essentially all the ones that can be (optionally) suffixed.\n- Those that apply to the request as a whole: limit, group_policy.\n\nGroup-specific parameters are already encapsulated in the\nplacement.lib.RequestGroup class, one instance of which is used to\nrepresent each request group in the query. And then database and other\nfiltering logic is encapsulated in\nplacement.objects.research_context.RequestGroupSearchContext.\n\nThis commit pulls the existing request-wide parameters, limit and\ngroup_policy, into a new class, placement.lib.RequestWideParams; and\nintroduces placement.objects.research_context.RequestWideSearchContext\nto encompass the filtering logic for the same.\n\nIn so doing, we adjust the signatures of methods which used to use those\nindividual parameters.\n\nAt the same time, parameters and local variables referring to the list\nof RequestGroup are renamed from `requests` to `groups` to\ndisambiguate/clarify what they are.\n\nIn the end, this is just a refactor. No logic is changed.\n\nChange-Id: I572ec8e6c0f9d81ead49964f41b46ab68a3de9a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5a6884ae12cab3e9627699b7a8045a0386343e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5a6884ae12cab3e9627699b7a8045a0386343e84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
