)]}'
{"id":"openstack%2Fplacement~666488","triplet_id":"openstack%2Fplacement~master~I0f241b2ca576ca3a29d3b5c74c6502908db8f681","project":"openstack/placement","branch":"master","topic":"review-comment","hashtags":[],"change_id":"I0f241b2ca576ca3a29d3b5c74c6502908db8f681","subject":"DNM: review comment related to agg flow down","status":"ABANDONED","created":"2019-06-20 04:58:55.000000000","updated":"2019-06-27 10:24:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c62cb41e50ee86edf7ef0a357170f170c472dcea","_number":666488,"virtual_id_number":666488,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-20 05:15:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 05:15:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3186ebde541bc4616c3bcca1d89cf14d8058cbf","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-06-20 04:58:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e748e6808480979d9e5aa0fb0093850d13db3861","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 05:15:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/666488/1/check/openstack-tox-docs/6c6494c/html/ : SUCCESS in 5m 05s\n- openstack-tox-cover http://logs.openstack.org/88/666488/1/check/openstack-tox-cover/0577ff1/cover/ : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/666488/1/check/openstack-tox-lower-constraints/4bbcafc/ : SUCCESS in 6m 59s\n- openstack-tox-pep8 http://logs.openstack.org/88/666488/1/check/openstack-tox-pep8/5b620f9/ : SUCCESS in 5m 57s\n- openstack-tox-py27 http://logs.openstack.org/88/666488/1/check/openstack-tox-py27/5177b02/ : SUCCESS in 6m 22s\n- openstack-tox-py36 http://logs.openstack.org/88/666488/1/check/openstack-tox-py36/24c02a4/ : SUCCESS in 6m 50s\n- openstack-tox-py37 http://logs.openstack.org/88/666488/1/check/openstack-tox-py37/5b84239/ : SUCCESS in 6m 55s\n- openstack-tox-functional http://logs.openstack.org/88/666488/1/check/openstack-tox-functional/3466af3/ : SUCCESS in 7m 06s\n- openstack-tox-functional-py36 http://logs.openstack.org/88/666488/1/check/openstack-tox-functional-py36/42dec1f/ : SUCCESS in 6m 57s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/88/666488/1/check/placement-nova-tox-functional-py36/72d3b8d/ : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ae8f2a150eda65eafd1df445e461b41dfc3abdfa","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-06-27 10:24:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6e4c0f6f78c7b43b39db30d38fcf653964a134b","revisions":{"b6e4c0f6f78c7b43b39db30d38fcf653964a134b":{"kind":"REWORK","_number":1,"created":"2019-06-20 04:58:55.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/88/666488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/88/666488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/88/666488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/88/666488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/88/666488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/88/666488/1"}}},"commit":{"parents":[{"commit":"9de03e1b17ddc202532757bb221ee81f02f89bbe","subject":"Uniquify allocation mappings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9de03e1b17ddc202532757bb221ee81f02f89bbe"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-06-19 05:21:57.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-06-19 05:21:57.000000000","tz":0},"subject":"DNM: review comment related to agg flow down","message":"DNM: review comment related to agg flow down\n\nThis patch is for a comment in\nhttps://review.opendev.org/#/c/662191/6/doc/source/specs/train/approved/2005575-nested-magic-1.rst\n\nThis describes the current behavior of aggregate flow down feature\nand disuss the need of root_member_of parameter.\n\nChange-Id: I0f241b2ca576ca3a29d3b5c74c6502908db8f681\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b6e4c0f6f78c7b43b39db30d38fcf653964a134b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b6e4c0f6f78c7b43b39db30d38fcf653964a134b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
