)]}'
{"id":"openstack%2Fplacement~667383","triplet_id":"openstack%2Fplacement~master~Ie94ab144005a5846fc951f4f387703bd241181c8","project":"openstack/placement","branch":"master","topic":"story/2005950","hashtags":[],"change_id":"Ie94ab144005a5846fc951f4f387703bd241181c8","subject":"Centralize and clarify pip in the docs","status":"MERGED","created":"2019-06-25 15:08:50.000000000","updated":"2019-07-01 13:48:57.000000000","submitted":"2019-07-01 13:48:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667383-1561988937588-e6a449f5","meta_rev_id":"02e5a3b009b096e905347d0170eabff261a20c7f","_number":667383,"virtual_id_number":667383,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-07-01 13:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2019-07-01 00:50:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-06-27 16:53:09.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-07-01 11:37:15.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-07-01 11:37:15.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 16:53:09.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-07-01 00:50:51.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-07-01 11:37:15.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-07-01 13:48:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9514e92a7332d888d67b82e308cf82adeba46ea9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-25 15:08:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f80a7c944f04ff8caf60f19ccd20c0b2b584e188","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-25 15:09:39.000000000","message":"Patch Set 1:\n\nReviewers: There may be a nicer way to organize this...","accounts_in_message":[],"_revision_number":1},{"id":"9b1a94c1451cd70d146dce04d3af69f951786468","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-27 16:41:16.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nrecheck zuul restart","accounts_in_message":[],"_revision_number":1},{"id":"fb3ac0bd453b5d83935e6812945199e730cf47c1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-27 16:45:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d0d3a005d565315f774908d3c61b0f5fa46daa7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 16:50:26.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fea319f813cd391b2ac0ecd5115daded2b2d9a1d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 16:52:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4cd7010323237b2ef99e81a2be0c9d1067b6a45f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-27 16:53:09.000000000","message":"Patch Set 2: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"872b687b757ba20a28f11a0177007baf0ba19822","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 00:27:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/83/667383/2/check/openstack-tox-docs/e4d5002/html/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/83/667383/2/check/openstack-tox-pep8/925d99b/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"2d1a439c18925168232f2818772ba328c6441726","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-07-01 00:50:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"88980d0981bd3565726cc67d8ab2da9dbddcaca0","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-01 11:37:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nYeah, good idea, thanks.","accounts_in_message":[],"_revision_number":2},{"id":"d50f421145cb0b19b072387f2c22adb39c8b86b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:37:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19cd22691f54f101f9b260bb54c0a911aabaa0cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 13:48:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/83/667383/2/gate/openstack-tox-docs/3f2971d/html/ : SUCCESS in 7m 08s\n- openstack-tox-pep8 http://logs.openstack.org/83/667383/2/gate/openstack-tox-pep8/9a101d4/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c40daaee691d8d9f1795f385f756484892908529","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 13:48:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"180c83b52bbf290d88f088752b0a2c29a31da730","revisions":{"ce97d26219cbb7ffa050a9f9f73b900149f38d3a":{"kind":"REWORK","_number":1,"created":"2019-06-25 15:08:50.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/83/667383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/83/667383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/667383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/667383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/667383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/83/667383/1"}}},"commit":{"parents":[{"commit":"93a7470d8606bc72353ed738cba38e14bece6c2d","subject":"Merge \"Spec for nested magic 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/93a7470d8606bc72353ed738cba38e14bece6c2d"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-25 15:06:49.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-25 15:06:49.000000000","tz":-300},"subject":"Centralize and clarify pip in the docs","message":"Centralize and clarify pip in the docs\n\nAdd a clarifying note about running pip (e.g. you might have to spell it\npip3) in a centralized place and link it from elsewhere. We should link\nto this whenever we mention pip so we don\u0027t duplicate information.\n\nChange-Id: Ie94ab144005a5846fc951f4f387703bd241181c8\nStory: #2005950\nTask: #34315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ce97d26219cbb7ffa050a9f9f73b900149f38d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ce97d26219cbb7ffa050a9f9f73b900149f38d3a"}]},"branch":"refs/heads/master"},"180c83b52bbf290d88f088752b0a2c29a31da730":{"kind":"REWORK","_number":2,"created":"2019-06-27 16:52:17.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/83/667383/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/83/667383/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/667383/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/667383/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/83/667383/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/83/667383/2"}}},"commit":{"parents":[{"commit":"93a7470d8606bc72353ed738cba38e14bece6c2d","subject":"Merge \"Spec for nested magic 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/93a7470d8606bc72353ed738cba38e14bece6c2d"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-25 15:06:49.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-06-27 16:52:15.000000000","tz":-300},"subject":"Centralize and clarify pip in the docs","message":"Centralize and clarify pip in the docs\n\nAdd a clarifying note about running pip (e.g. you might have to spell it\npip3) in a centralized place and link it from elsewhere. We should link\nto this whenever we mention pip so we don\u0027t duplicate information.\n\nChange-Id: Ie94ab144005a5846fc951f4f387703bd241181c8\nStory: #2005950\nTask: #34315\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/180c83b52bbf290d88f088752b0a2c29a31da730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/180c83b52bbf290d88f088752b0a2c29a31da730"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
