)]}'
{"id":"openstack%2Fplacement~667800","triplet_id":"openstack%2Fplacement~master~I9a379e0e60ae8eb53280b8296229d2f0412eae4a","project":"openstack/placement","branch":"master","topic":"story/2005842","hashtags":[],"change_id":"I9a379e0e60ae8eb53280b8296229d2f0412eae4a","subject":"Add OsProfiler config options to generated reference","status":"MERGED","created":"2019-06-27 07:39:08.000000000","updated":"2019-07-01 13:44:21.000000000","submitted":"2019-06-27 16:14:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667800-1561652084380-c2951fcc","meta_rev_id":"892d449fda59781095eb916ddd5d10882c396377","_number":667800,"virtual_id_number":667800,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-06-27 16:14:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-06-27 14:34:58.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-06-27 12:40:25.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-06-27 14:34:58.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 12:40:25.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-06-27 14:34:58.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-27 16:14:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7eb7ae620f560d4e0d83d99618d841097ff4e17","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-06-27 07:39:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99cc19f7c555daf4af3d0fa901cf404574a3de40","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-06-27 12:40:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nLooks OK to me, this is a copy of how it\u0027s done in nova.\n\nhttp://logs.openstack.org/00/667800/1/check/openstack-tox-docs/56a947e/html/configuration/config.html#profiler","accounts_in_message":[],"_revision_number":1},{"id":"4d343daf79df1ffb47ea58b3931f5a6b71259463","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:36:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/00/667800/1/check/build-openstack-api-ref/6ddc543/html/ : SUCCESS in 6m 43s\n- grenade-py3 http://logs.openstack.org/00/667800/1/check/grenade-py3/d51031a/ : SUCCESS in 59m 24s\n- tempest-full-py3 http://logs.openstack.org/00/667800/1/check/tempest-full-py3/85b5ff9/ : SUCCESS in 1h 34m 50s\n- openstack-tox-docs http://logs.openstack.org/00/667800/1/check/openstack-tox-docs/56a947e/html/ : SUCCESS in 4m 21s\n- openstack-tox-cover http://logs.openstack.org/00/667800/1/check/openstack-tox-cover/44ebfbd/cover/ : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/667800/1/check/openstack-tox-lower-constraints/5c2be53/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/00/667800/1/check/openstack-tox-pep8/264e80b/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/00/667800/1/check/openstack-tox-py27/4c1ec4e/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/00/667800/1/check/openstack-tox-py36/280d56c/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/00/667800/1/check/openstack-tox-py37/adaf261/ : SUCCESS in 4m 45s\n- build-openstack-releasenotes http://logs.openstack.org/00/667800/1/check/build-openstack-releasenotes/d74f797/html/ : SUCCESS in 4m 09s\n- openstack-tox-functional http://logs.openstack.org/00/667800/1/check/openstack-tox-functional/8be6b4c/ : SUCCESS in 4m 39s\n- openstack-tox-functional-py36 http://logs.openstack.org/00/667800/1/check/openstack-tox-functional-py36/2a818a8/ : SUCCESS in 4m 11s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/00/667800/1/check/placement-nova-tox-functional-py36/4b93a12/ : SUCCESS in 11m 10s\n- placement-perfload http://logs.openstack.org/00/667800/1/check/placement-perfload/a7c0cd2/ : SUCCESS in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8e3c65a148909c54aca52d1932f62a9296270b1","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-27 14:34:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b7f05942203941a5629a8018ab8feb471849180","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:35:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bfcaad1cd775a3ce83ea3c1ffd30e3b53cd4ef77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:14:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"369dc0b8df1ef0c944b6736adbce75a74a5ed73f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 16:14:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/00/667800/1/gate/build-openstack-api-ref/c6bf015/html/ : SUCCESS in 4m 10s\n- grenade-py3 http://logs.openstack.org/00/667800/1/gate/grenade-py3/ddf83eb/ : SUCCESS in 1h 02m 59s\n- tempest-full-py3 http://logs.openstack.org/00/667800/1/gate/tempest-full-py3/48041af/ : SUCCESS in 1h 31m 54s\n- openstack-tox-docs http://logs.openstack.org/00/667800/1/gate/openstack-tox-docs/68a4521/html/ : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/667800/1/gate/openstack-tox-lower-constraints/15646f3/ : SUCCESS in 7m 13s\n- openstack-tox-pep8 http://logs.openstack.org/00/667800/1/gate/openstack-tox-pep8/18b4618/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/00/667800/1/gate/openstack-tox-py27/0555d7d/ : SUCCESS in 6m 30s\n- openstack-tox-py36 http://logs.openstack.org/00/667800/1/gate/openstack-tox-py36/df47991/ : SUCCESS in 7m 23s\n- openstack-tox-py37 http://logs.openstack.org/00/667800/1/gate/openstack-tox-py37/b7a9e1a/ : SUCCESS in 5m 16s\n- build-openstack-releasenotes http://logs.openstack.org/00/667800/1/gate/build-openstack-releasenotes/6f45741/html/ : SUCCESS in 4m 25s\n- openstack-tox-functional http://logs.openstack.org/00/667800/1/gate/openstack-tox-functional/19d1597/ : SUCCESS in 4m 31s\n- openstack-tox-functional-py36 http://logs.openstack.org/00/667800/1/gate/openstack-tox-functional-py36/a66db82/ : SUCCESS in 9m 13s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/00/667800/1/gate/placement-nova-tox-functional-py36/c82e860/ : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e1a06af40d9c9273cae1dc281f16d0949e668e3e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-01 13:44:21.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nnote within for future reference","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"663780060461177a3c88c2c2d252224511287fa8","revisions":{"663780060461177a3c88c2c2d252224511287fa8":{"kind":"REWORK","_number":1,"created":"2019-06-27 07:39:08.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/00/667800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/00/667800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/00/667800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/00/667800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/00/667800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/00/667800/1"}}},"commit":{"parents":[{"commit":"93a7470d8606bc72353ed738cba38e14bece6c2d","subject":"Merge \"Spec for nested magic 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/93a7470d8606bc72353ed738cba38e14bece6c2d"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2019-06-26 12:46:54.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2019-06-27 07:37:56.000000000","tz":120},"subject":"Add OsProfiler config options to generated reference","message":"Add OsProfiler config options to generated reference\n\nOsProfiler is optional, but if it\u0027s installed we\u0027ll load up the configuration\noptions from the library, but they weren\u0027t in the generated config sample so\npeople would have to find the osprofiler docs, or worse the code, to figure\nout how to configure it.\n\nThis simply adds the OsProfiler config options to the config sample, which\nwill also show up in the config reference docs.\n\nChange-Id: I9a379e0e60ae8eb53280b8296229d2f0412eae4a\nStory: 2005842\nTask: 34191\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/663780060461177a3c88c2c2d252224511287fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/663780060461177a3c88c2c2d252224511287fa8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
