)]}'
{"id":"openstack%2Fplacement~669170","triplet_id":"openstack%2Fplacement~master~Ic74c3d1b8aa272e9c7d1070fe7658b928e693889","project":"openstack/placement","branch":"master","topic":"bp/support-consumer-types","hashtags":[],"change_id":"Ic74c3d1b8aa272e9c7d1070fe7658b928e693889","subject":"Add consumer_types migration, database and object changes","status":"MERGED","created":"2019-07-04 12:45:30.000000000","updated":"2021-08-25 22:19:45.000000000","submitted":"2021-08-25 22:18:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":118,"unresolved_comment_count":33,"has_review_started":true,"submission_id":"669170-bp/support-consumer-types","meta_rev_id":"7f670bf639d7ab456f7d8567a537f34851855dea","_number":669170,"virtual_id_number":669170,"owner":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-25 22:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2021-08-19 17:13:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2021-08-24 10:11:33.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2021-08-06 08:07:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2021-08-24 10:11:33.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"value":0,"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-30 18:33:32.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-13 14:22:53.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2020-01-17 09:57:48.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2020-03-11 02:14:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-12 12:20:21.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2020-03-18 19:46:25.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2020-03-19 03:31:04.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2021-04-27 14:47:12.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-07-30 14:59:38.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-08-19 17:13:12.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-08-24 10:11:33.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"}],"messages":[{"id":"2452702f6bd4eac0ef95d531a09dbbee7110b71e","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-07-04 12:45:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d51bbcd41f84d28bff544f2820d814032841fca","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-04 13:22:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\n1 cleanup (downgrade) and 1 suggestion to make potentially less code, within\n\nbut overall looks really good","accounts_in_message":[],"_revision_number":1},{"id":"bf2954af9d9d78d80fd800978d9fd76c71ee32f2","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-07-04 13:38:08.000000000","message":"Patch Set 1:\n\n(2 comments)\n\n\u003e (2 comments)\n \u003e \n \u003e 1 cleanup (downgrade) and 1 suggestion to make potentially less\n \u003e code, within\n \u003e \n \u003e but overall looks really good\n\nthanks for the quick review, I\u0027ll address them in the next iteration.","accounts_in_message":[],"_revision_number":1},{"id":"1f001d6768da7a0589149baca0b4e1cdf21b67db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 20:09:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/70/669170/1/check/grenade-py3/e183880/ : FAILURE in 51m 56s\n- tempest-full-py3 http://logs.openstack.org/70/669170/1/check/tempest-full-py3/7bc7cee/ : SUCCESS in 1h 47m 30s\n- openstack-tox-docs http://logs.openstack.org/70/669170/1/check/openstack-tox-docs/99261a8/html/ : SUCCESS in 3m 33s\n- openstack-tox-cover http://logs.openstack.org/70/669170/1/check/openstack-tox-cover/2b2b322/cover/ : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/669170/1/check/openstack-tox-lower-constraints/da8c76d/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/70/669170/1/check/openstack-tox-pep8/cfe3af8/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/70/669170/1/check/openstack-tox-py27/12c62ed/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/70/669170/1/check/openstack-tox-py36/727ad18/ : SUCCESS in 5m 51s\n- openstack-tox-py37 http://logs.openstack.org/70/669170/1/check/openstack-tox-py37/6ace6ac/ : SUCCESS in 4m 29s\n- openstack-tox-functional http://logs.openstack.org/70/669170/1/check/openstack-tox-functional/dcb0031/ : SUCCESS in 4m 25s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/669170/1/check/openstack-tox-functional-py36/2f147a6/ : SUCCESS in 4m 46s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/70/669170/1/check/placement-nova-tox-functional-py36/4696cf2/ : SUCCESS in 10m 57s\n- placement-nested-perfload http://logs.openstack.org/70/669170/1/check/placement-nested-perfload/497481b/ : SUCCESS in 14m 28s (non-voting)\n- placement-perfload http://logs.openstack.org/70/669170/1/check/placement-perfload/921eaff/ : SUCCESS in 7m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20b69d70ec297c82801ca3e668c81a1dc9d98e51","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-07-24 09:15:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"aff9eae40dff55ebf78e496f44c75655ce289f73","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-07-24 09:15:44.000000000","message":"Patch Set 2: Workflow-1\n\n\u003e Uploaded patch set 2: Patch Set 1 was rebased.\n\nstill need to address the comments.","accounts_in_message":[],"_revision_number":2},{"id":"a8cabea2f46ca7b345b3206b5be43d85cbb555e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:53:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/70/669170/2/check/grenade-py3/fc19b95/ : FAILURE in 58m 28s\n- tempest-full-py3 http://logs.openstack.org/70/669170/2/check/tempest-full-py3/8ad4983/ : SUCCESS in 1h 38m 01s\n- openstack-tox-docs http://logs.openstack.org/70/669170/2/check/openstack-tox-docs/f7d92bd/html/ : SUCCESS in 5m 25s\n- openstack-tox-cover http://logs.openstack.org/70/669170/2/check/openstack-tox-cover/481a6e1/cover/ : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/669170/2/check/openstack-tox-lower-constraints/3cb07d5/ : SUCCESS in 4m 39s\n- openstack-tox-pep8 http://logs.openstack.org/70/669170/2/check/openstack-tox-pep8/3e076d7/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/70/669170/2/check/openstack-tox-py27/73b6859/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/70/669170/2/check/openstack-tox-py36/7d7f323/ : SUCCESS in 5m 16s\n- openstack-tox-py37 http://logs.openstack.org/70/669170/2/check/openstack-tox-py37/131f374/ : SUCCESS in 4m 34s\n- openstack-tox-functional http://logs.openstack.org/70/669170/2/check/openstack-tox-functional/6dedbf8/ : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/669170/2/check/openstack-tox-functional-py36/3d327cb/ : SUCCESS in 6m 21s\n- openstack-tox-functional-py37 http://logs.openstack.org/70/669170/2/check/openstack-tox-functional-py37/8d110c9/ : SUCCESS in 6m 51s\n- placement-nova-tox-functional-py37 http://logs.openstack.org/70/669170/2/check/placement-nova-tox-functional-py37/194fa9c/ : SUCCESS in 12m 14s\n- placement-nested-perfload http://logs.openstack.org/70/669170/2/check/placement-nested-perfload/f080361/ : SUCCESS in 19m 19s (non-voting)\n- placement-perfload http://logs.openstack.org/70/669170/2/check/placement-perfload/dc667a3/ : SUCCESS in 9m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0b67d1f87eb38e9311e4b09064cee95fecef7fc","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-29 13:31:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5d3a6d49652274010a1acf0eb08db6b9e378bdc4","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-29 14:46:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aee12681a0cbf3ca9a54ca4d6e9b30f770f275e6","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-29 17:00:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e967dd1a6e21258a54038cc131e4bf261163d1ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 22:19:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e35687bab6484c828573e653cf481591 : FAILURE in 1h 21m 16s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/eae8d706c2314711b075dad5d9dbe5f2 : SUCCESS in 1h 55m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3450dbc16afa4181bd10df2fc550f8ad : SUCCESS in 5m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edff969db47e424c9a4158a89966f69a : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/014ef85bcf794ed398039ec2d3022b64 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19019d62b81b4e29b8934f97ae508416 : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7549aeb3c8854da092d233d16eeffa04 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f49d87bd455048f7a80dc977c5a41d60 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/906b81abbc9f43229a6df4325a67bc4c : SUCCESS in 7m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/43fca5a494054e4c9f20d8390358a2f3 : SUCCESS in 5m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3bb091ff403a4541bdc3a32e98e2d6db : SUCCESS in 6m 21s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c105021b0e194e899e4bb1820fba29de : SUCCESS in 5m 58s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/039d0f82a75b4bdc8d8dcad4e5906c56 : SUCCESS in 12m 45s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/c1756e81c89b4bd7aee5c72a6d0031b0 : SUCCESS in 10m 45s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/01caa7e63df14e789697acb071b305bc : SUCCESS in 13m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84036d9c099c487fb82d4219cb569a46 : SUCCESS in 1h 19m 09s","accounts_in_message":[],"_revision_number":3},{"id":"473bed743de53a796cfc227cfe8205445f835211","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-30 13:09:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"25118afea88b0589ef7001716f464454f3a64021","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 17:00:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5f0f6c7689bf4712861e925a27315b4b : FAILURE in 54m 10s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/31d24d0e0e09476cafaf082581f840a6 : FAILURE in 46m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74bfd7f4a9124e0db240c9ab36ebd46f : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bf47a537ad347f0a931ec1e93198326 : SUCCESS in 6m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07d9a7f7cef9422fb2f052fb5772b2ce : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdffcc66bf14864a54d6118a6468903 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d1c848b14854c6bacfbe6b1793d27de : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20764ffd6eb14e2b8090993a371f34d9 : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff77dc88d1904d9c8eefea79e510e487 : SUCCESS in 5m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fef8e55b27164ee2ac221e0ff60d34b5 : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/38303988f8534830bd0fec53053968f8 : SUCCESS in 7m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5d5c9e2a471544f3be1eb2826b171215 : SUCCESS in 6m 05s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e5011bc70ca940a2a105c00b017c0ab2 : SUCCESS in 12m 05s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/067d5dd874474ac990dd99ffac23bddf : SUCCESS in 40m 22s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/67f4539dbd5748749f4fafed14264d06 : SUCCESS in 9m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4807d0f85b64469e8d8ca32ad2611df3 : FAILURE in 49m 23s","accounts_in_message":[],"_revision_number":4},{"id":"08f24ef7c4ae46a35f44a3e41c5e8322b45ba694","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-30 18:33:32.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4b3ca9d223451dc9a979cd9aeea104376b0964f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 20:57:37.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ebe15007164941b396a6d2d54da3dc6c : FAILURE in 52m 33s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f01afd75991b44fd8d5f3e4e9c95d217 : FAILURE in 51m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ba57e0b70943d1be3d5f9477ebdf90 : SUCCESS in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/698959275d3a48f0b83bc4ef92578af3 : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d76b2d1ae19a4415960c0e7fa271ca3a : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac99fe82a114f5cb1afa43704686e9e : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/423171df2fc84e438771a2203c836f04 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d507179fd1da4b3c8a1edde7f9e78b65 : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c356515a34104f1d8b0cbd6513b228fb : SUCCESS in 4m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2439eb019fde43e0a5f6c43c3fc6924e : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/50d845d454204281adb0acf7abb1eb25 : SUCCESS in 5m 14s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ffc88778e48549c5aa2bf69caeedef64 : SUCCESS in 5m 41s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/156b5977103c412586fdbe8d60a96df9 : SUCCESS in 12m 55s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/035e649d3703468ca3794073f071a411 : SUCCESS in 33m 53s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/4c3e77900e6945bd8da74ce2344fffaa : SUCCESS in 12m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66f65f0519254182a1dcb0a734642ec4 : FAILURE in 43m 13s","accounts_in_message":[],"_revision_number":4},{"id":"cfd604e7ba0a431ab7886c3aa58c30478ff5f61b","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-31 11:17:25.000000000","message":"Patch Set 4: Workflow-1\n\ngrenade and tempest failures are real. some of the code in the next path needs to be in this one.","accounts_in_message":[],"_revision_number":4},{"id":"17afde6b34d4a4930352b9efce88feff2af61f0e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-02 09:35:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2a1ef735f473cf80c2dfc8f538d2899b8cbb23ba","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-02 09:36:11.000000000","message":"Patch Set 5:\n\nI collapsed the two database oriented changes into one.","accounts_in_message":[],"_revision_number":5},{"id":"ad485e965d14979c51dfefdee22b5eb56e62ff22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 11:00:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f7568053396941e5b5e2f8ef228e8ee4 : FAILURE in 54m 03s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/20d0f2fd1cf14a77b432badc22da4da6 : SUCCESS in 1h 19m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a8eb63e2a8f4768bcf248aa19221c2c : SUCCESS in 5m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4f033742c8d46f381f4c1a30c813864 : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a4d98af80b44545a396de6e015648db : FAILURE in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b1b95267214b6394aba9c56b745f58 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb115a4e3d1240ad875ae3e9d6bc3b57 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2b30ebeb70444fcbe2fb457aa26b68f : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86ea792766014f85b35196e9a6beecfb : SUCCESS in 5m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a7588d51f3324af9a2e1d8350d1ddb6c : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0cc798796b54a129c6644047e1c231d : SUCCESS in 5m 16s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/6f5c419c329849618615b1894d9e7407 : SUCCESS in 5m 46s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ddbf463f4f2f459aae34552378678421 : SUCCESS in 12m 56s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/c44a26521c1e40c4bfcaca6a53eb2028 : SUCCESS in 10m 01s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/2acd99363acc4a39ad4351288ba1b22e : SUCCESS in 10m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5dac817461c441d2a58ce0bad61156ed : SUCCESS in 1h 16m 32s","accounts_in_message":[],"_revision_number":5},{"id":"256d1fbd8f707a4899f709653f36e5755bcaf009","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-02 11:12:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ed365fe3278d9988f11944e1659159ea4e4f659e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 12:37:50.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/47f201b1d9434de5b656fde4a6ff7398 : FAILURE in 49m 34s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/8811f323f0a34fbbae1ac3ce3b42b12d : SUCCESS in 1h 19m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d681dde19c948b38b58169847e63273 : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1fc6995242346fb8262ff96dae9ac2d : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f332b067db9481985df443cf9bc9a15 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de8b909eac174d9ea776ff61b0c41b77 : SUCCESS in 7m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/306b87ca06e54db88059403545b187e0 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3aaf0e70f869486d9497d37142b4041c : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3787d0598633400e8fc1612cbf99d9a7 : SUCCESS in 6m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/498e61cc61454c00ac5be7d84cca0f84 : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/18f453a5c6be4c899bfcacf596730677 : SUCCESS in 4m 46s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/26c2a20440f24368b5b7b0198637119a : SUCCESS in 7m 17s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/dfb82aa237a44d6eaff899d50a4ec593 : SUCCESS in 13m 04s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/deebf3c4381941e2a4e6acfd42cabe1a : SUCCESS in 10m 26s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/eaa858f192ae43b29bff5d537f8ed84c : SUCCESS in 12m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2148192d5ca405a82282e7d82a4255d : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":6},{"id":"7f4f128dade6ca3da3ad099e61331d9ceaf1c54c","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-02 14:16:40.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"7db868b4cfbd323dcf3a4ac2ef81968375bcf973","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 15:34:00.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"05fbe317789c6421b0e78da7923b526e8033906f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-02 15:42:08.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"6f8a8b4f4517d7017aa6e76f70729e2adf604406","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 16:42:26.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":7},{"id":"2c06767f0feed1a2406ef20b8e181671efd718da","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-02 20:42:57.000000000","message":"Patch Set 7:\n\nrecheck grenade fix","accounts_in_message":[],"_revision_number":7},{"id":"b61c0623d158d7a6aecc00238609ee7c4f9f7c29","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-02 21:41:46.000000000","message":"Patch Set 7: Workflow-1\n\nthis is going to fail again. need to try the migration with local mysql and postgresql","accounts_in_message":[],"_revision_number":7},{"id":"3d443fb817499154cc235757aeb74cea1a0a6f27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 21:53:57.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9cafff58fa8d48428873a0a562980a6b : FAILURE in 43m 06s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/ce4264b45e2a44b9947a4450bea3cac8 : SUCCESS in 1h 09m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7258206fc6b4f0899efb42d93486d3f : SUCCESS in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/934592a7668a478381ce074640f84f4a : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eac3576e563042afa89b93c416940685 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59f8a548690f45388698470c00c8ad89 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6729786b9d024f84911b0bc43228d7fb : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c787ab57760b455990468585b10fe647 : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/628c8dcfc3c142f48b19f754e1b98b3c : SUCCESS in 7m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/02d3c6b7140947fd8a8ab201b29b0742 : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/02ae37e397e049b6bae8ded566506457 : SUCCESS in 5m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/84367b4f1bc2416e837c7df5d154d87b : SUCCESS in 6m 16s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/530b5a81e1ac44a1a9fe84c93b33c246 : SUCCESS in 14m 24s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/ff53a78d783740ff99c6b3162d2fad6a : SUCCESS in 23m 19s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/da8049d0848b4d209c33c7c3c5244a34 : SUCCESS in 10m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62a962a0ac5946bb83589ece199060ff : SUCCESS in 59m 28s","accounts_in_message":[],"_revision_number":7},{"id":"912f3091eb2ed21cbd630a4e34f511b24812499e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-03 09:43:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2b46969d21f0aaef3aec0a0a9b983f375d591507","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 11:14:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a3d5907f94084bccb4233f6c742b9b76 : SUCCESS in 1h 03m 04s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/52ff2c9aa2474fed9ec4c52878ec95f8 : SUCCESS in 1h 24m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab74ef7eeff0404aac94fadbe59d0736 : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54816c7eb1784509b2acad02b7e7b777 : SUCCESS in 7m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9349a1c433664c3e8a69fd2b16c46369 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bf0cf5e73d4ea090fafa82a6a3c109 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ab0b978c5614e308377429b6c392df5 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/149f8276dc254db8b1c6dbd62fe747c3 : SUCCESS in 7m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5aa137d8e9374dabac3d54d773d08ad7 : SUCCESS in 7m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/893acef885b94d7b8a9cbd1bb859c1ea : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/028b3a83a75641efb8408e205c8af56c : SUCCESS in 7m 13s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/284e58fddd6645ab8cd8f1596a9aca4c : SUCCESS in 8m 55s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/504dc45a1b324866967bcf19b3f5bf49 : SUCCESS in 15m 13s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/91f02eacb29e4ab3b00cfaae2bf46740 : SUCCESS in 29m 09s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/60356ea1e8a84cc29915292a40e1cbd1 : SUCCESS in 12m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e95f28e12a2c4a6595c27fb5335a4950 : SUCCESS in 1h 15m 15s","accounts_in_message":[],"_revision_number":8},{"id":"d1a686ec8cdce18563035e8aadc56bc1f5021dd6","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-09-13 00:30:43.000000000","message":"Patch Set 8:\n\n(8 comments)\n\nImplementation looks good from what I can tell, but have some comments on the tests.","accounts_in_message":[],"_revision_number":8},{"id":"0bcfd6d7dc826791ac37f313118c872aab251bef","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-13 09:35:19.000000000","message":"Patch Set 8:\n\n(2 comments)\n\nI can try to get to mel\u0027s comments today but I may not make it and I\u0027m off next week. If Surya, or Mel, or anyone else would like to do them (if I don\u0027t) that would be ideal.\n\nAs discussed in IRC there\u0027s no need to too fussy about Feature Freeze on this unless we start getting pushed for time or find other issues elsewhere in placement.","accounts_in_message":[],"_revision_number":8},{"id":"d398fd8b32a1bffe05bb920d0754a9f47ffceafa","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-09-13 19:41:08.000000000","message":"Patch Set 8:\n\n(1 comment)\n\n\u003e (2 comments)\n \u003e \n \u003e I can try to get to mel\u0027s comments today but I may not make it and\n \u003e I\u0027m off next week. If Surya, or Mel, or anyone else would like to\n \u003e do them (if I don\u0027t) that would be ideal.\n \u003e \n \u003e As discussed in IRC there\u0027s no need to too fussy about Feature\n \u003e Freeze on this unless we start getting pushed for time or find\n \u003e other issues elsewhere in placement.\n\nI will probably be able to do the updates next week.","accounts_in_message":[],"_revision_number":8},{"id":"7eaf84edf74b713c35ed665af6e24ac084a22108","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-23 10:32:49.000000000","message":"Patch Set 8: Code-Review-2\n\nLet\u0027s continue working on this stack but not merged it for Train.","accounts_in_message":[],"_revision_number":8},{"id":"5d79ddc6e9654f50a005126b9f63638c611cfd73","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-10-07 13:13:04.000000000","message":"Patch Set 8:\n\n\u003e Let\u0027s continue working on this stack but not merged it for Train.\n\nSorry about this, was in a training school the past month, couldn\u0027t get to this on time.","accounts_in_message":[],"_revision_number":8},{"id":"dd9f8c3ecce5cfc2d56a3d73e6b015478505e1c4","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-10-16 20:42:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"834003b393df4a68ab28ea22522253cafa8eaf67","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-10-16 20:44:06.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"00ce55a9bd992df31149f71977ba76c80bca7966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 23:35:10.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ecb61fac96ce4386a71a2455eb82138f : SUCCESS in 1h 01m 28s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/471d42a4d37a442489311c7d607bf7ec : SUCCESS in 1h 08m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66cb0767043c40b3b6847871edfcce77 : SUCCESS in 8m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fe767915005496fa177eea47916b64c : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85db0614144443fca7dcd7883b47ddf4 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/121ef88aca3d45cb9844bcd696d7ba56 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3f1b192fae342b69933d6f672b47bf1 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25de5aab48254c27bc90c7312fe74c39 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc7f9bc18c584e59b267e054fb37d506 : SUCCESS in 5m 34s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/466634bd2e414cd28751c6a8743d0487 : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a2b30dcc671247c29869592aebd46ee0 : SUCCESS in 5m 51s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/5f9d3c08b3284d66a463a1c25aeac93b : SUCCESS in 6m 07s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d2a0a68f4a5e45508c6c6bb1a5550be3 : SUCCESS in 13m 18s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/a06eaa5aca9645169abed23339941e15 : SUCCESS in 18m 53s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/2fff73b7ccf84dc99d7c7c6a77358f8d : SUCCESS in 9m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c868d16c56da421a86c9ac0746e6da92 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":9},{"id":"3fc781b3c3f2af6a5299641ecd73cee753313c60","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-11-04 11:43:32.000000000","message":"Patch Set 9: -Code-Review\n\nTaking off my -2 because Train has left the station","accounts_in_message":[],"_revision_number":9},{"id":"11a7cddbf9c47741532faf79ab561dc1fc3156f7","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-11-04 11:51:31.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"fdf87d52eb65a09013214049177faa3c195547ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 13:10:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/29fdace3ac464d03902d7895bb6c1f34 : SUCCESS in 54m 03s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/8d1c489004494eb49391bbd89a31d4c8 : SUCCESS in 1h 12m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a6f0b1fe6e14977820dfefd00cb8677 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35f44235c2f8496e9d57e9f97d436357 : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f636387bb3e54b3eb8681cdec2dc6ed7 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/685ed8d299624c4d8c42f3728f781bca : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc7f58fd53c64df7b5439236f829c4fa : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51365b6d3d8c4d4696c926485fae1881 : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8b95c5971c04d69b98fc42ceb949afc : SUCCESS in 5m 47s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/bcc4f83c5fcc4be19d7f23f810ba0d7e : SUCCESS in 5m 22s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ddc0ae1c09f7416d98a2484dfd2880ab : SUCCESS in 12m 23s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/792f1fa160be4a0b84b667d950df185d : SUCCESS in 31m 47s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/0189658f275740dd9d5d2d85c712e54a : SUCCESS in 10m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9778d398c57a4f22baf8706c219de4e1 : SUCCESS in 1h 05m 53s","accounts_in_message":[],"_revision_number":10},{"id":"6a95929fd660a49f4ba99ec742e8dab79fd4e144","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-11-04 22:28:37.000000000","message":"Patch Set 10: Code-Review+1\n\nLGTM, my addressed my own testing comments in PS9.","accounts_in_message":[],"_revision_number":10},{"id":"2039c47651c90062d4557ef218a47de92817b888","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-11-13 14:22:53.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"80a6f6f8fefce556c2c2b3316d13ef9a168c4f19","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-16 12:36:35.000000000","message":"Patch Set 10:\n\n(3 comments)\n\nThe code looks good to me.\nJust one question on design level why not have the same change on GET /usages/\u003crp_uud\u003e API?","accounts_in_message":[],"_revision_number":10},{"id":"da88b468c6171b7f4d27906df2ffbf7ceddf3c5b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-01-17 00:20:28.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nThanks so much for reviewing tetsuro! I need to investigate a bit on your design question.","accounts_in_message":[],"_revision_number":10},{"id":"84ac641ff6e90942c8d73d55d648db07abb36cb5","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-17 02:10:50.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9b5c05e9aaa81001230b96df62ef51fa5193f5a3","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-01-17 06:03:24.000000000","message":"Patch Set 10: -Code-Review\n\n(1 comment)\n\nI\u0027m going to update this series to address Tetsuro\u0027s comments, including the design issue. I think both /usages APIs should get the same format update in microversion 1.37.","accounts_in_message":[],"_revision_number":10},{"id":"faa0bb98008076927c9a94cfd78c642af927ae65","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2020-01-17 09:57:48.000000000","message":"Patch Set 10:\n\nI was under the impression that the reason for not seeing consumer type in `/resource_provider/{uuid}/usages` is because we don\u0027t see consumers there and the reason we don\u0027t see consumers there is because we don\u0027t anyone to be misleaded into thinking that the usages information at resource_provider is a complete representation of what that consumer is consuming (since that consumer is notionally able to consumer from multiple providers for one instance of allocation).\n\nSo I would argue that rp usages and global usages are much different things for which it is okay to have different representations.","accounts_in_message":[],"_revision_number":10},{"id":"c67f3ee40932b1e0a0d163ecbc6fb716a20a6361","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-01-17 16:02:04.000000000","message":"Patch Set 10:\n\nThanks Chris for giving the context. Let\u0027s see what Tetsuro thinks about it.","accounts_in_message":[],"_revision_number":10},{"id":"63e1bec9d4c8e2ce580cbe6ad14b70229a3c3fb2","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-21 10:45:28.000000000","message":"Patch Set 10:\n\nAs long as we see usages there, consumers are there whether we see part of it or whole of it. For example, when I want to know whether, for example, rp1 is used by nova or by ironic standalone, I would call ‘/rp/{uuid}/usages’ rather than whole usage API, and expect the API describes how it is used.\n\n \u003e we don\u0027t anyone to be misleaded into thinking that the usages information at resource_provider\n \u003e is a complete representation of what that consumer is consuming\n\nThis is a good point but IIUC, we already have exposing-only-part-of-a-consumer API,  GET `/resource_providers/{uuid}/allocations`[1]\n\n[1] https://docs.openstack.org/api-ref/placement/?expanded\u003dmanage-allocations-detail,list-resource-provider-allocations-detail#list-resource-provider-allocations\n\nPlus,  the new API’s response wouldn’t be like this,\n\n{\n   \"usages\" : {\n      “consumer_uuid1” : {\n         \"consumer_type” : “INSTNACE”,\n         \"MEMORY_MB\" : 1024,\n         \"VCPU\" : 2,\n         \"DISK_GB\" : 5\n      },\n      “consumer_uuid2” : {\n         \"consumer_type” : “MIGRATION”,\n         \"MEMORY_MB\" : 1024,\n         \"VCPU\" : 2,\n         \"DISK_GB\" : 5\n      },\n   },\n}\n\nbut like this:\n\n{\n   \"usages\" : {\n      \"INSTANCE\" : {\n         \"consumer_count\" : 5,\n         \"MEMORY_MB\" : 512,\n         \"VCPU\" : 2,\n         \"DISK_GB\" : 5\n      },\n      \"MIGRATION\" : {\n         \"DISK_GB\" : 5,\n         \"VCPU\" : 2,\n         \"consumer_count\" : 2,\n         \"MEMORY_MB\" : 512\n      },\n   }\n}\n\nwhich I think is less misleading into “this is a complete representation of what that consumer is consuming”, that said, there is no consumer uuid representation in the response. (i.e.  which consumer did you mean by “that consumer”?)","accounts_in_message":[],"_revision_number":10},{"id":"11e80a08e037fcc6679478075e46ce036899f7f2","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-01-24 07:43:34.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)\n\n...and found one bug in the code. Please also see my comment in the next patch[1].\n\n[1] https://review.opendev.org/#/c/679441/9/placement/tests/functional/gabbits/consumer-types-1.37.yaml@144","accounts_in_message":[],"_revision_number":10},{"id":"b5dcca62be6ce7be30a396a1b4eaf6eb5bf2a764","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-01-30 00:10:00.000000000","message":"Patch Set 10:\n\nThanks Tetsuro for your latest review, sorry I have not gotten time to update the patches yet. I plan to update them this week.","accounts_in_message":[],"_revision_number":10},{"id":"4887476163ea5efe37e61e939491a60ae2835114","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-01-31 01:17:19.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"367156cb7d6c7c0838c7764ea22f5d62e136de8d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-06 20:06:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3260c35f314180f22836a7cbe8fb1d5c3709a6e8","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-07 00:05:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"dcc652cdfc434a63e0906787ada648ab276eac7d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-07 01:18:37.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"d164ff3ded142312e464981e8c91e6f7fb61fbb0","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-02-07 01:53:13.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"121a1368da07fc0f75fed02b6ddf93d8258cf281","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-07 02:02:20.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b1961effddabe9f7edfabb7544524ea93788aa14","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-07 02:39:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"efdafec96faec54bb5e3b379841c4b306ec7a9f7","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-07 02:56:21.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e8ba51385f0bacd0eeea5b030a9be448276eec39","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-07 03:12:32.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"01d5d45187479edfd277d863f20f6d3a7f44f701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 03:58:53.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9a35e075c184427b9a2050092ac744a7 : SUCCESS in 58m 46s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/bb8b5cb62e5f4bdf9ba0b0a96a0019ba : SUCCESS in 1h 18m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04f33036d6f94c83884ce4a7f48679a2 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e139ec2d5974493f911d5848b359a335 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f5215212ca46d487ebc1a0bf5d2c22 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e13e3c2345934304a95acf0ff0f90989 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb59f691dcf84036ab901bcf01115404 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5842a811f5f45488616a4148e4920fc : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfac50c3d46741c994df803863ac637e : SUCCESS in 9m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c8d9ebbc0de4575bf6c3fc78f793278 : SUCCESS in 5m 09s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c9be093a278446bd999c3e05b68095f1 : SUCCESS in 7m 04s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/b7f14b4f247e484a9367eb9b0ee85377 : SUCCESS in 13m 51s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/98c9bc22e30141b490c427a494884a4a : SUCCESS in 30m 54s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/2387509f3deb4634afdf0cf407067429 : SUCCESS in 10m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebddafff51724a3eb24066857b460754 : SUCCESS in 1h 11m 10s","accounts_in_message":[],"_revision_number":11},{"id":"30869d180b52a36d47358a00f1c1950f159df96d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-14 04:12:06.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a448a9ef0567f1ebb299de8e236b9b4b158bedad","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-14 04:17:31.000000000","message":"Patch Set 12:\n\n(2 comments)\n\nThe only change in this PS is to fix the counting of unique consumers in the next patch functional tests.","accounts_in_message":[],"_revision_number":12},{"id":"77f8f4a47a06042740e957decdaef4b794c5ed51","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-14 04:20:50.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"7ebe3a2f676e48a2f26378951f880a4d1da27317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 05:07:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/011da8684a0a423d9e229f556a4bfd91 : SUCCESS in 54m 10s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/634e3e710ed046e4be91284c26d011f3 : FAILURE in 14m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d965d1fa73e743269285594d080ab2f2 : SUCCESS in 10m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84896eb10aab495492dfeb209dde79a7 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e1ae58ca424feda2af15547f6d9116 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa4d73779a244c7f8ac9ac270d239feb : SUCCESS in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce26cd33799c4fcab280a245129b63d5 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8985dfa32d184036987efba343776429 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1811554e098a4fe18fce25744528eddc : SUCCESS in 11m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/753f8dc7a22f4b6497c019e9062dd594 : SUCCESS in 7m 24s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e24dd7c1a468418992a3ad26568fa36b : SUCCESS in 10m 04s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/bb56ca52f5d943e988c64226ab0594d1 : SUCCESS in 18m 29s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/0e046d1cb32348969fa27f10271bc7ce : SUCCESS in 24m 00s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/9e853219d4b84fd2b374586208dd19fe : SUCCESS in 13m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60ba2ce3f56041d0a323610a5ff89068 : FAILURE in 18m 51s","accounts_in_message":[],"_revision_number":12},{"id":"9b087d4df02e121bcdb224bae207a0c038d81f80","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-14 15:43:37.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"38a65aaaf1a4ee151eab08ca8204936e6ebbc264","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-02-14 15:59:34.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"21789511888886116efd1a449871dde34e3c1651","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-14 16:04:35.000000000","message":"Patch Set 12: Code-Review-1\n\nI\u0027m going to rework the sqla queries based on feedback from zzzeek in #openstack-placement","accounts_in_message":[],"_revision_number":12},{"id":"1c2ddf45198261099eee8b7de17cf8757bd866fc","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-25 22:36:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"7c3ec156a932b9c447e8e31dc573a7dfc2bb2a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 00:01:32.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4d767631614e44819e836e5274309011 : SUCCESS in 1h 02m 01s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/8d7c61de475d4420926e32f18e27962a : SUCCESS in 1h 23m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9aa06d49f97d45f4889dca5b053a6fae : SUCCESS in 7m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06d31aabd5374737bea6356707fbbaf6 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c28da095f84f58b8747bee002bc12e : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/617fbf52004549eb91e0511b920801d3 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff5d2cbe159e40e988cf7afb185343dd : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9cfc9ac7cc448d1a61fc3f25f548730 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3776cca442a443b7913f3fd839f17f72 : SUCCESS in 8m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9177f829bf2e4178aeb6cbcc718bcab8 : SUCCESS in 6m 32s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/43c5c7cf3b444515b5267ea53d89a3c2 : SUCCESS in 9m 09s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/475a2fe7e6df45cdbde3c10ef406b64c : FAILURE in 17m 57s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/2c63f3bbd3b84f4cafd3483ab2598aa1 : SUCCESS in 22m 51s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/3a9e6c0b63ff48f385f09e9a3f4ed955 : SUCCESS in 13m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9ab3a637f444b7199fd9cdc335ce9d2 : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":13},{"id":"8e477ae68798023c25ab47ccce523aceb9d69359","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-02-26 08:52:18.000000000","message":"Patch Set 10:\n\n(1 comment)\n\nThe bug has been smashed. Thank you for updating.\nWaiting for another update.","accounts_in_message":[],"_revision_number":10},{"id":"1ecfdf938cb43148a92d4c51999070ddb7d97ea3","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-27 02:53:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7d1a7850d3f510215842297e8ab48249707afb04","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-05 05:51:25.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"66d833fdb555712051a46fe3e36725f5e80ab24b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 07:09:41.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2039e3d2601d48b5866142333953f03b : SUCCESS in 56m 28s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/1af11b50c59a4265ac03cbc5434e2745 : SUCCESS in 1h 16m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0c1c69145cd4fb3b4b2b7ace950ebe8 : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6113ee6c03ad4ad9b7fe8c20d9e6e424 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e24a9e09c3ac458687484ad85a6eeb44 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f2cf103c357498f8968bb72e1b0153d : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e58e0e7baa346359765197ac8b75d43 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7ca48f940c7415c9957738ba2d7abdb : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6d7d74aa834e6bad286826538e61fd : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4ec89b4aec847d9a14f751959d2cefd : SUCCESS in 5m 06s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/2b6260470e6f44b79bcb6c151d387696 : SUCCESS in 8m 23s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c96063f08ad7471f9e573021b5481a75 : SUCCESS in 13m 51s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/e4f9ed7a26f642a1b94161e58257960a : SUCCESS in 21m 04s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/b585b86a277a430abac20c0f447ac9f3 : SUCCESS in 13m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d7d5974f361498d8d5cf41a10f61dfe : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":14},{"id":"d9f7e42f9cf22ab91f7d0fdc15d9767eb17b68de","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-03-10 00:24:02.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c63d7b0e0c606b9054ff1696c1f316b9a587105b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-10 01:17:15.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"15c2c12fb26bd96c0ed20044038baa111c0a12b0","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-03-10 02:38:31.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"aed757107c3042bba2aa6b9592906f0ce6b04a65","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-10 02:55:24.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"9344154d69ab036490197c2dd4f0eed9d82b0db1","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-11 00:54:37.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"45cf49e3735f2246de8056ef0873050e452c8d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 02:14:04.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/19c3c6333c214cb0a195f21b0cccc252 : SUCCESS in 1h 04m 23s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/b4c503bcce14429592a3d66842d50931 : SUCCESS in 1h 14m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c3808df20a040bab0a8bcda7e96b8d9 : SUCCESS in 6m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/953b7417ac1f4901bc52e18f4535f07e : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd6a413b56414e79a6adbcc89cf19de3 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/268b6649da2c483b8a0f263830339a8c : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/074fd6d0b5be4e039d0e8ceae86bd4aa : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9eefba0b2dd84a549f7c5d9d8737819f : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53ffaeb282cd4f978319f45ad0ea1d1c : SUCCESS in 9m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/19028af29ca744ebb4e1ef7c837d1543 : SUCCESS in 6m 28s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/1a9641c02bce42f88e18cd3a1710596a : SUCCESS in 7m 40s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/51a96cea20404b1b8200d6ec360f6fcb : SUCCESS in 16m 46s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/e902e82fcb404671b4f69f2d1cf2c926 : SUCCESS in 25m 00s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/56c3701a76394837822acab1e7492268 : SUCCESS in 12m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aeafbebc85ee475ea139b91a88ce690e : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":15},{"id":"b8860dd3acdfd4df4333cd27af56eee387ce4738","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2020-03-12 12:20:21.000000000","message":"Patch Set 15:\n\n(7 comments)\n\nI checked it out what if we drop the UNKNOWN row in the database...Any thoughts?","accounts_in_message":[],"_revision_number":15},{"id":"b02bda0166634b6f76b1fb75b16e442c4081cd62","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-18 18:14:14.000000000","message":"Patch Set 15:\n\n(3 comments)\n\nPlease see my comments inline about why I kept the default consumer type record UNKNOWN and let me know what you think. I am not sure how else to break the cycle of being able to create new records with a NULL consumer type ID. Or do you think it should be OK to have NULL forever?","accounts_in_message":[],"_revision_number":15},{"id":"f259ea03ecf69ff880fcc915a57d7a627dd8a8d8","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-03-18 18:38:15.000000000","message":"Patch Set 15:\n\n(6 comments)\n\nmostly extra info about the choices here but the one thing you should definitely change is the inspector.from_engine() thing since that\u0027s deprecated.","accounts_in_message":[],"_revision_number":15},{"id":"ef02a3b01ba9ca96c90dd0108894315b3122ed1d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-18 19:14:22.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"4e8f72c30edd36a8b40e51dc9c490549bf0dc5dd","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2020-03-18 19:46:25.000000000","message":"Patch Set 15:\n\n\u003e (1 comment)\n\nlets talk about the two tables, \"consumers\" and \"consumer_types\".  the decision to make here is about the size of the data.\n\nif existing \"consumers\" tables are huge, like 100ks of rows or more, then it becomes more risky to do a migration where there\u0027s a NOT NULL, or the other version where you start with NULL and then backfill then set NOT NULL, because yes, MySQL is not good at this kind of thing and then with Galera involved, that also causes things to work more poorly with very large transactions.  I\u0027m not actually sure what the \"message too long\" error in https://bugs.launchpad.net/nova/+bug/1862205 is, I\u0027ve not seen that before, it\u0027s not exactly what Galera does.   \n\nbut anyway, if consumers is huge, it\u0027s likely not worth the complexity to add a new NOT NULL column with a server default.\n\n\nunfortunately, \"consumers\" being huge is also what would cause a LEFT OUTER JOIN between \"consumers\" and \"consumer_types\" to be slow.     So you might need to make sure you are mindful of what kinds of queries you do with these tables.  SQLAlchemy now has a very efficient form of eager loading called \"selectin\" loading that first and foremost has MySQL\u0027s awful SQL optimizer issues in mind, so if you want to load lots of consumers and their consumer_types eagerly loaded, the \"selectin\" load will emit a second quick SELECT statement for those types with the \"consumers\" SELECT unaffected.","accounts_in_message":[],"_revision_number":15},{"id":"e87aeaf1678b0e6ef0a846f565277ee3ec80565b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-19 03:31:04.000000000","message":"Patch Set 15:\n\nThanks Mike.\n\nWith all of this in mind, I think I\u0027m back to thinking we should (1) not use a server_default so as to avoid a server backfill of default values on existing records and (2) do a batch placement-manage db online_data_migrations to allow population of the default value in chunks and then no longer create new records without a default value for older microversions \u003c\u003d 1.36.\n\nThat way we\u0027ll be able to guarantee that all records have consumer_type_id filled in by the V release, and then we can remove the migration code and change to a left outer join to an inner join.","accounts_in_message":[],"_revision_number":15},{"id":"f49cdfa175da78a77afd1a9aada608e91bec2e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-04-21 01:22:06.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f4588f831d757b24b1726d9a852611192719e95f","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-04-21 01:48:05.000000000","message":"Patch Set 16:\n\nThis update removes the DEFAULT_CONSUMER_TYPE_ID entirely and allows NULL in the Consumers.consumer_type_id column. Records with NULL consumer_type_id will be presented as \u0027unknown\u0027 users but kept as NULL internally.\n\nThis is done to avoid unnecessary data migrations.","accounts_in_message":[],"_revision_number":16},{"id":"261a723dffe524f34e7c988fba1c04a9d5ee4c85","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-04-21 01:48:39.000000000","message":"Patch Set 16:\n\n\u003e Patch Set 16:\n\u003e \n\u003e This update removes the DEFAULT_CONSUMER_TYPE_ID entirely and allows NULL in the Consumers.consumer_type_id column. Records with NULL consumer_type_id will be presented as \u0027unknown\u0027 users but kept as NULL internally.\n\u003e \n\u003e This is done to avoid unnecessary data migrations.\n\ns/\u0027unknown\u0027 users/\u0027unknown\u0027 to users/","accounts_in_message":[],"_revision_number":16},{"id":"849c2809227eca063aeb82239bca4b89882d2c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 02:22:53.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/120865db143f494cb614b188ec0e1a48 : SUCCESS in 57m 38s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/0b6438d58f24445bb39d5700dd67b15f : SUCCESS in 59m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfdf3e31e8524050a72dc6fd62c1f934 : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d31e12b65d9d45ef9d0d07f6e4815b47 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b67afc9e8de4af2b701ed29f032fcd5 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/929448a75b974e359501a7489dc145ad : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a9820b3f1584f0b9a283240ed35a8fd : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/195b680c047743c29b43be8fdcf7ca04 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf58752513894668b48a290ce271a939 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/745b3b3ba3b54e959789388ff3d9f131 : SUCCESS in 6m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ab21b905de4e422eaf25550965cab20c : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8c220f1d7aac47a0bfef0cae6616d637 : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b7348ffc5854c73bcc34a79aa057c52 : SUCCESS in 5m 57s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c6350b9de434cef98821cd095adc104 : SUCCESS in 14m 51s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/bf6a09e8469e4957b1f991c03c62828b : SUCCESS in 16m 05s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/e0f06ae184d0436bad7e62784a1ede1c : SUCCESS in 10m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0931eaf06bb6406e9ec986ab4c203f41 : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":16},{"id":"335c20da22743955fa2bd7eed8c1241b5b2a5205","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-27 14:47:12.000000000","message":"Patch Set 16: Code-Review-1\n\n(3 comments)\n\nI have two questions inline.","accounts_in_message":[],"_revision_number":16},{"id":"de2fb9aa3cf8c8f2792fa257737a8ab791813860","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-27 15:51:36.000000000","message":"Patch Set 16: Code-Review+1\n\n(2 comments)\n\nOK, now as I read the next patch I understood this a bit better. I only have a small nit about a code comment and that can be fixed in a follow up so I upgrade my vote to +1. I still have question in the next patch hence I\u0027m holding my +2 for now.\nCheers,\ngibi","accounts_in_message":[],"_revision_number":16},{"id":"e0344f97c68726c1c362c30adb3418f5115819e5","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-04-27 22:01:32.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f53838fbf670e0150397e60891716700aa9f3aba","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-05-17 15:00:50.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"51a955eaef3c8acf4a204cfd933b12d17d250fc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-22 01:34:29.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"14cc5455e2b052377c6209aafd8400ad7330c819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 03:14:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd441fa73e774540b721174b8f802674 : SUCCESS in 58m 58s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/6596a4690728499f9f93dbe31326fd2c : SUCCESS in 1h 28m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/885c10b5e41b40b7bac4711d82ba2b33 : SUCCESS in 7m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5adde58dc8c4bf7a5720f88f4e35f19 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5a365edb174f5fa417f7f046ac5acd : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/614e46be239e471ab76b11f70abc8e39 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71de771677664bb68b416a78674d9e30 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89df4c436cab45408bdf726b0561f26a : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c168173c224d63a84e5edc155577e1 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bad1dc67e606423e884774ab25c080be : SUCCESS in 6m 50s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/777754f4f3c345aeb4c761bc3e2609ee : SUCCESS in 7m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a6e0fe1ded404c77895d8f7a6911ad4b : SUCCESS in 6m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a49e6663797d4fd18793f1881c44338a : SUCCESS in 6m 03s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d8d53166b7b846128421c89bc97789d1 : SUCCESS in 14m 22s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/105fdd7f1cea4e78ad6a98e04e528eb4 : SUCCESS in 26m 20s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/29713be9961c4230a9f755a14491cb3e : SUCCESS in 8m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4bd88e3e88e4dfd893aaec73476f5de : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":17},{"id":"0d01a73d1ac1eb21ec8deeba6f7bf2b766e285bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-23 20:35:35.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"2394a9fbe1b7157f977e3b8fa71b36be58a2db88","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-23 21:36:11.000000000","message":"Patch Set 18: Workflow-1\n\nTrying some integration testing.","accounts_in_message":[],"_revision_number":18},{"id":"a928ffc2261c757ee67c620bbbafb779d2dd03d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 21:41:24.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d690febed9954d74858eb2ff2a47883e : SUCCESS in 1h 00m 29s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/814258c4e5444719b863b8d2079336c8 : SUCCESS in 1h 03m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c69966452d6e4e95990e66edd5d0078f : SUCCESS in 6m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81112e325fde4aa8829674dbd53c37ea : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/639da3f9964e4df2a8f05e1915e57963 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcd93c69068a4bedafaad9699b6bd88a : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0813e05779b4e47bfad1dedea2106ac : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99810bb823c54629b53530bb597764c8 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1024d2a1e6f0469bb3ba81aa1c0576e8 : SUCCESS in 8m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bff068e425fe458f93bc1a0d1b871428 : SUCCESS in 5m 40s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7169dd37635c41d3aa77b88966c56850 : SUCCESS in 8m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f35ada7ad6b749debac2a0347765939b : SUCCESS in 6m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c7be064cf5641bbba5c5fedd60a8e1f : SUCCESS in 6m 16s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8c89d94498934e5a87fc789c658a7d5a : SUCCESS in 15m 00s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/9efeaaca3d394aeebc1fd89c675a451b : SUCCESS in 31m 07s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/5aef57cb75f24c68ac5a0aebb753603c : SUCCESS in 13m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54b452116f68417888513e96e6333d31 : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":18},{"id":"85c9b458f01773b7ead18fb1fdc6a6dcc22c51e8","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-27 16:57:34.000000000","message":"Patch Set 18: -Workflow\n\nHere is the DNM nova patch that exercises consumer types:\n\nhttps://review.opendev.org/c/openstack/nova/+/802133\n\nIt all looks to work well EXCEPT for hitting a generation conflict during resource deletion in tempest. This situation is an older one that was reported here:\n\nhttps://bugs.launchpad.net/nova/+bug/1836754\n\nSo that would need to be resolved before this series can reliably pass the gate while consumer types are actively being used. Note that in the DNM patch the final run is green because I added Depends-On for the WIP patch addressing the above reported issue ^.","accounts_in_message":[],"_revision_number":18},{"id":"ed7231531601b496a44a45bc50a2707d857814a8","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-07-27 18:01:19.000000000","message":"Patch Set 18: Code-Review-1\n\n(3 comments)\n\nFine adding a new table but getting the \"migrate\" dependency out would be very important here if possible","accounts_in_message":[],"_revision_number":18},{"id":"fd5d5505d1fea8d8254497d6c3157ecfab29b636","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-27 18:22:40.000000000","message":"Patch Set 18:\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"5c9568f6b9a8db65af0bca6d9f19f36ec1786fdd","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2021-07-27 18:52:17.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"2045b467d427b0530c95342fbca2610063c819f6","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-27 19:04:06.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"7d84dcea48b691b91a423f27379f1b7cb6a909eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-27 21:58:29.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"8b9d230d3e81fee514bfb863d08a7f960ad579f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 23:09:45.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8ab85908a7434decaf6ce9c507bc3d12 : SUCCESS in 1h 00m 08s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/253629b2ab50461cb07f7f713ac4a954 : SUCCESS in 1h 06m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ebf5153a39a49168e0cdd218cdb64a7 : SUCCESS in 10m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4c1180c9a9f4aec81dcced280f68883 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/391501e016fb4cdd8968a469d0d961b5 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac52e375815644bcba2e0095f907d87f : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f81060a9d8947179e3ebf58159bb758 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc86437d5e314a6994220a08bf659ee9 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e053969a0b4ed69bca58c775aad9ac : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/853d475eec074f4d906bd9a0bd58abf5 : SUCCESS in 5m 39s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/9457b2ac31f74c158bb965cf2976ee2d : SUCCESS in 6m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/34c9ca23a24a43c283396e3162e1e88b : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9635bd16a4084874af7c04600faa5c1b : SUCCESS in 6m 17s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c36b8f2f52444901a669b88fe71e7cdb : SUCCESS in 13m 58s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/183fdde164f14e468ec45770aae29838 : SUCCESS in 20m 30s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/04329808a47e48ccb843074dc7cf8dea : SUCCESS in 9m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a2240bca9a9429c9106157384846f36 : SUCCESS in 48m 46s","accounts_in_message":[],"_revision_number":19},{"id":"c0cf2d896dc49223e8d007618eaf74cdace33f09","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-07-30 14:59:38.000000000","message":"Patch Set 19: Code-Review+1\n\n(1 comment)\n\nFirst glance looks legit, but I\u0027ll review more once Gerrit is back after the planned outage.","accounts_in_message":[],"_revision_number":19},{"id":"b7396a752e02f21bbb8772a630baa53132d3b7b8","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-07-30 22:20:33.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"5bd866389425845e8aba71bc7351badcfe4f077c","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-08-03 14:25:00.000000000","message":"Patch Set 19: Code-Review-1\n\n(15 comments)\n\nI think 99% of the change is all good, but I see a reason why I need to hold this change. Sorry, but I think we need to uppercause the default name of a ConsumerType object.","accounts_in_message":[],"_revision_number":19},{"id":"beb970c2f9fbb6ac00ecf238e029d010bb76a447","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-08-05 15:50:07.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)\n\nRemoving my -1 so putting +2","accounts_in_message":[],"_revision_number":19},{"id":"d664f91e674c24f8d302662ec9b6b0f73f462ffa","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-08-05 18:06:03.000000000","message":"Patch Set 19: Workflow-1\n\n(5 comments)\n\nGoing to update to rename the global alias variable for \"unknown\". Should probably also add one for \"all\". And then I\u0027ll update the wrong code comment in the test and remove the unnecessary use of label().","accounts_in_message":[],"_revision_number":19},{"id":"3bbb72718c2fde48630e59beb13efed9f7289fcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-08-05 23:23:57.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"019fe255d7fff9ecdbe9ebceabb61c1cc13fd502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 00:50:57.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1ce280377b2e4b5ba59b264fca0e606a : SUCCESS in 1h 24m 10s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/f6bd060e7fed4ec7ba9cb504003751c8 : SUCCESS in 1h 18m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6c1abd760ff408f9c74a5f554fa2a6a : SUCCESS in 7m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f4d6a60498e438293c417478e7d6c8d : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/231dc9a0850f4ce09566ecf4ba21e095 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/730405d279e04024a560d1f9c6661c24 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5769fe163c014282976f64b9356f47b0 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac285505891d4bbda803bfda5dc0f594 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4cc90c55052445591fc93ccefa6a6eb : SUCCESS in 8m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0142bc9400df4b6a9e339b86aec5c93e : SUCCESS in 5m 49s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a53cd6a9bde245e19be097081a4a788f : SUCCESS in 6m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/50f5566fceec4fe29feff03ed4a7555f : SUCCESS in 6m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b868c8cb94654f9cba5498b9b5ff40f5 : SUCCESS in 6m 37s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/73e28eecd6424e528111d7cba1f65815 : SUCCESS in 13m 23s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/37d11904a0e54dde8579fb791c5170f3 : SUCCESS in 32m 28s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/1ed3707ae9de47eb9b978af35e4d0789 : SUCCESS in 11m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d63aecb6844422294537a270a956d10 : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":20},{"id":"09c95f6d4d62b0cd227b8cbd8f91ba83e7202c99","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-08-06 08:07:53.000000000","message":"Patch Set 20: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":20},{"id":"3bb9dc2bfbcaa6f21d4415e8d420422db517bd01","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-08-19 17:13:12.000000000","message":"Patch Set 20: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":20},{"id":"35e152f6d88c1669f203080c5db42b1a40fac11d","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-08-24 10:11:33.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1\n\nLGTM, well tested and sets everything up for the microversion change in the next patch.","accounts_in_message":[],"_revision_number":20},{"id":"ffc4dd315bcae4df4a418bb2a7bc99767510ac45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 10:12:09.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"57744799c50690a2df5301330b5a11b746740787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 11:54:23.000000000","message":"Patch Set 20: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/6790902891574007abff31cf42c6de7d : SUCCESS in 1h 00m 10s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/4c748325963a4ce1b68b5d772c31e5fc : SUCCESS in 1h 32m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/704616360caa44178ef360e5d3cfc110 : TIMED_OUT in 41m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe31b8ded99948169ec2a731effedac6 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9700958ecf0b4a0093026b8c2aba5600 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90432552c1aa4515b99e4f14f6227402 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b05a36111e41fa9a92890f127513c3 : SUCCESS in 8m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1df5b546dd0d40638bc94b5169c62094 : FAILURE in 6m 26s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/ddc3b550dd384a5fa5108cb364ed308f : FAILURE in 7m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/daf7f9a6e4db4107a2cb1ccef63aa0ba : FAILURE in 7m 10s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aa8d3fea1de14cc5a09d5acfd85c50e7 : SUCCESS in 14m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4571619899640bf80f8e17cb553bbce : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":20},{"id":"5d0f45f97762205d48e15bf6a0222aa4f2884bec","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-08-25 11:16:22.000000000","message":"Patch Set 20:\n\nNeeds I89b330d0647b0883eecaf40b988a1a4779bcb1dd","accounts_in_message":[],"_revision_number":20},{"id":"e149e36fd32ca5f24306adc679c093ecde7def43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 12:26:11.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e6910c36c54243e38a64c5bcc4362e89 : SUCCESS in 1h 01m 51s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/bf6a0ac049b94aebb431d0c020af6fd9 : SUCCESS in 1h 07m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1acfaf628514944960c4ca36776c0bf : FAILURE in 8m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8ddf0a95dd849a1b2e7882dc484650f : TIMED_OUT in 41m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51fa7be41fb64e7488450a22b593494c : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f809e5209e74809ba8c486bc5185255 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d42836a0ec8b415abeb912de3ef709cd : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a9c3d15a7f44210a9dc763c28358b32 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3145c708f0534b61950e238d87e198e8 : SUCCESS in 9m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/401ebbc67df1444faf66b024bd15272e : FAILURE in 6m 15s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/a5057986823b4832aa3ae571c49f0358 : FAILURE in 7m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3b8d8b183bb746cfabb42d20308ac1a2 : FAILURE in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5debcf6363be4e16927f1a865238bbb3 : FAILURE in 7m 22s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/499756a861044e3f9f9c1025f5b4794a : SUCCESS in 12m 37s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/e84324231776484ebe80a0b42678e453 : SUCCESS in 27m 06s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/3e664eaea6664e698b4826f9bd94d726 : SUCCESS in 8m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bf8b34a3fc0469697fcdf5b02e877fc : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":20},{"id":"9b7d15cb4fcfb8a27c9517da72caefef44542c59","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-08-25 15:31:16.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"f5848f2044780dd569b322d50b4b3e649e9e3781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 16:49:28.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8816983dcca94f58b51ed8f6623bc117 : SUCCESS in 55m 59s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/0c192edc77f1427bb8adfc9dcaf12a35 : SUCCESS in 1h 11m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd1337d4099b41099a3dfd89097a7c85 : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ad79bdd5f31434aac9a4b7b80c09392 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4923e1433ce34982aabb2fabcf1dc8df : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4e9e26c8755467388c7f7796095cbe8 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7df2ca95b5094824b1747dc8d120d9f6 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/420191c49084493b8c709942d92bf624 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8038d9632a1347d9b6ad8cb802e92117 : SUCCESS in 8m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb145f01ba804f8f8694e92d706d7f10 : SUCCESS in 6m 01s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/4dec60ad9a6f4edb93fabce50b326129 : SUCCESS in 9m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d9240ebbe9ac4e32919f0ea7b44d1e03 : SUCCESS in 6m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7e04a7e0f1bc4fc4b3a4263ed8970686 : SUCCESS in 8m 21s (non-voting)\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/93c22047b8844046a9c309b780d93762 : SUCCESS in 18m 02s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/48ce46806a91490e93306a431d0980c1 : SUCCESS in 19m 18s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/dad11f01709c49d6803a9d057fcee9b5 : SUCCESS in 9m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96661e8fa8a74a57a321f14bcbba23df : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":20},{"id":"397751afd8755f4eeac8acce06792e4502c2be80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 16:50:49.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"a76473eb8ccd1cfb7e119a70acaa28057eab0f49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 22:18:14.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f30eb514a06f49b3a40a007f77b2aad1 : SUCCESS in 57m 43s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/6aaa26cac1564b8aabc4bde28a565f7d : SUCCESS in 1h 00m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db20c7c0314444d887dca2236af3df19 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c641388eb1394e00b5d62eb9de54dd90 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3db6c92881d46eba8e7653bc8ab693f : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e233ce84e6c24436b991425082de2d71 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8aac660d03c483abdff45b43cbda27e : SUCCESS in 10m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/18c295af88144b10b63b89a86b213ec5 : SUCCESS in 7m 15s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8cea892758c34a328ebcdbf81b6d1fbf : SUCCESS in 8m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/914d0a13ca164233ba009ad1b36dd26b : SUCCESS in 8m 08s\n- placement-nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8cfa7a5993694685a0c07d309c4192cc : SUCCESS in 15m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dabe83b069dc42879a0fbfa363ff06f2 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":20},{"id":"521cdd4de2bdad3705dc44ea7504e39982513a82","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 22:18:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"7f670bf639d7ab456f7d8567a537f34851855dea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 22:19:45.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f9b369407644ffb87c8b4a6bcb4c2f1 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"b1f3dd39c3fd140ea77fba474c8a3a118a41de8c","revisions":{"d3f158e057427103c0140d2de369318eba040a18":{"kind":"REWORK","_number":1,"created":"2019-07-04 12:45:30.000000000","uploader":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"ref":"refs/changes/70/669170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/1"}}},"commit":{"parents":[{"commit":"62d65617958414ce7f324af2e9adaf723435fccc","subject":"Merge \"Fix up some inaccuracies in perfload comments and logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/62d65617958414ce7f324af2e9adaf723435fccc"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"subject":"Add consumer_types table and consumer_type_id column to consumers","message":"Add consumer_types table and consumer_type_id column to consumers\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d3f158e057427103c0140d2de369318eba040a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d3f158e057427103c0140d2de369318eba040a18"}]},"branch":"refs/heads/master"},"80809e49672f5dc35630b7ab0d4e3984ade8c2ca":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-24 09:15:06.000000000","uploader":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"ref":"refs/changes/70/669170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/2"}}},"commit":{"parents":[{"commit":"885012cb646d2ff884ea903b7c8dda7cbb465c61","subject":"Merge \"Update api-ref location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/885012cb646d2ff884ea903b7c8dda7cbb465c61"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-23 13:31:36.000000000","tz":120},"subject":"Add consumer_types table and consumer_type_id column to consumers","message":"Add consumer_types table and consumer_type_id column to consumers\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/80809e49672f5dc35630b7ab0d4e3984ade8c2ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/80809e49672f5dc35630b7ab0d4e3984ade8c2ca"}]},"branch":"refs/heads/master"},"6c97cb3a8f4b4552a6052ea978f5c8cfcb7d8cb8":{"kind":"REWORK","_number":3,"created":"2019-08-29 17:00:11.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/70/669170/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/3"}}},"commit":{"parents":[{"commit":"7e885108916f7243ddab0ac9a27d3a540c070fb1","subject":"Merge \"Get usages in _build_provider_summaries()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7e885108916f7243ddab0ac9a27d3a540c070fb1"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-29 13:30:41.000000000","tz":60},"subject":"Add consumer_types table and consumer_type_id column to consumers","message":"Add consumer_types table and consumer_type_id column to consumers\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6c97cb3a8f4b4552a6052ea978f5c8cfcb7d8cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/6c97cb3a8f4b4552a6052ea978f5c8cfcb7d8cb8"}]},"branch":"refs/heads/master"},"30e3ad1c47849cb557f867aab7c4ed9b0c4a4e59":{"kind":"REWORK","_number":4,"created":"2019-08-30 13:09:44.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/70/669170/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/4"}}},"commit":{"parents":[{"commit":"8272dfee98d800566b4e2d169d0fa02a635293c1","subject":"Fix links to migration scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8272dfee98d800566b4e2d169d0fa02a635293c1"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-30 12:11:12.000000000","tz":60},"subject":"Add consumer_types table and consumer_type_id column to consumers","message":"Add consumer_types table and consumer_type_id column to consumers\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThis change predicts the existence of a consumer_types.ensure_sync\nin one of the migration tests. That\u0027s coming in the patch following\nthis one. That ensure_sync will add a row of 1:UNKNOWN to the\nconsumer_types table.\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/30e3ad1c47849cb557f867aab7c4ed9b0c4a4e59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/30e3ad1c47849cb557f867aab7c4ed9b0c4a4e59"}]},"branch":"refs/heads/master"},"66dc6792b44433e4ec235f6002523283f70301f1":{"kind":"REWORK","_number":5,"created":"2019-09-02 09:35:44.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/70/669170/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/5"}}},"commit":{"parents":[{"commit":"8272dfee98d800566b4e2d169d0fa02a635293c1","subject":"Fix links to migration scripts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8272dfee98d800566b4e2d169d0fa02a635293c1"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-02 09:20:18.000000000","tz":60},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and object\ncode to support saving consumer types when creating consumers and\nreporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nA consumer type object is added, similar to trait and resource class.\nLike those there is a set of default data which needs to be ensured\nat process startup time. In this case it is one row in the table:\nID 1 with value UNKNOWN. This is the default value for an\nunspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nCo-Authored-By: Surya Seetharaman \u003csuryaseetharaman.9@gmail.com\u003e\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/66dc6792b44433e4ec235f6002523283f70301f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/66dc6792b44433e4ec235f6002523283f70301f1"}]},"branch":"refs/heads/master"},"273165c8fb8f177f96f196d373eb104ab5e0b1ca":{"kind":"REWORK","_number":6,"created":"2019-09-02 11:12:47.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/70/669170/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/6"}}},"commit":{"parents":[{"commit":"156b08335733f6b13bc994c831bdf5b7add7f417","subject":"Fix typo in microversion sequence test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/156b08335733f6b13bc994c831bdf5b7add7f417"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-02 11:11:36.000000000","tz":60},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and object\ncode to support saving consumer types when creating consumers and\nreporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nA consumer type object is added, similar to trait and resource class.\nLike those there is a set of default data which needs to be ensured\nat process startup time. In this case it is one row in the table:\nID 1 with value UNKNOWN. This is the default value for an\nunspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/273165c8fb8f177f96f196d373eb104ab5e0b1ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/273165c8fb8f177f96f196d373eb104ab5e0b1ca"}]},"branch":"refs/heads/master"},"f783c98e4479e335d5103a1282ba06b2393fdca3":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-09-02 14:16:40.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/70/669170/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/7"}}},"commit":{"parents":[{"commit":"156b08335733f6b13bc994c831bdf5b7add7f417","subject":"Fix typo in microversion sequence test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/156b08335733f6b13bc994c831bdf5b7add7f417"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-02 14:15:38.000000000","tz":60},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and object\ncode to support saving consumer types when creating consumers and\nreporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nA consumer type object is added, similar to trait and resource class.\nLike those there is a set of default data which needs to be ensured\nat process startup time. In this case it is one row in the table:\nID 1 with value UNKNOWN. This is the default value for an\nunspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nThe depends on is to a grenade change that upgrades placement\nindependent of nova. Neeed now because we need to do our own\ndb sync as a result of this new table.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nDepends-On: https://review.opendev.org/679655\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f783c98e4479e335d5103a1282ba06b2393fdca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/f783c98e4479e335d5103a1282ba06b2393fdca3"}]},"branch":"refs/heads/master"},"7e12ce1d1dacd1e2858a4c3f1801119c6964efc1":{"kind":"REWORK","_number":8,"created":"2019-09-03 09:43:59.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/70/669170/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/8"}}},"commit":{"parents":[{"commit":"156b08335733f6b13bc994c831bdf5b7add7f417","subject":"Fix typo in microversion sequence test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/156b08335733f6b13bc994c831bdf5b7add7f417"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-09-03 09:36:47.000000000","tz":60},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and object\ncode to support saving consumer types when creating consumers and\nreporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThere is a set of default data which needs to be ensured at process\nstartup time. In production this is done via the database migration.\nBecause of the way the database-driven functional tests work the row\nadded in the migration is rolled back with each test, so it is manually\nadded back. The default data is one row in the table: ID 1 with value\nUNKNOWN. This is the default value for an unspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nThe depends on is to a grenade change that upgrades placement\nindependent of nova. Neeed now because we need to do our own\ndb sync as a result of this new table.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nDepends-On: https://review.opendev.org/679655\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7e12ce1d1dacd1e2858a4c3f1801119c6964efc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7e12ce1d1dacd1e2858a4c3f1801119c6964efc1"}]},"branch":"refs/heads/master"},"59d8270405d1c4dd8229967124520e948ce98978":{"kind":"REWORK","_number":9,"created":"2019-10-16 20:42:27.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/9"}}},"commit":{"parents":[{"commit":"156b08335733f6b13bc994c831bdf5b7add7f417","subject":"Fix typo in microversion sequence test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/156b08335733f6b13bc994c831bdf5b7add7f417"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2019-10-16 20:41:53.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and object\ncode to support saving consumer types when creating consumers and\nreporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThere is a set of default data which needs to be ensured at process\nstartup time. In production this is done via the database migration.\nBecause of the way the database-driven functional tests work the row\nadded in the migration is rolled back with each test, so it is manually\nadded back. The default data is one row in the table: ID 1 with value\nUNKNOWN. This is the default value for an unspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nThe depends on is to a grenade change that upgrades placement\nindependent of nova. Neeed now because we need to do our own\ndb sync as a result of this new table.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nDepends-On: https://review.opendev.org/679655\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/59d8270405d1c4dd8229967124520e948ce98978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/59d8270405d1c4dd8229967124520e948ce98978"}]},"branch":"refs/heads/master"},"70aeec852b8b1bb9973cd77c20fb28ea6eaacf10":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-11-04 11:51:31.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/70/669170/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/10"}}},"commit":{"parents":[{"commit":"2a665b5d995452af1858373304a05c67c01cc6bb","subject":"Merge \"Remove unused import statement.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2a665b5d995452af1858373304a05c67c01cc6bb"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-11-04 11:44:12.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and object\ncode to support saving consumer types when creating consumers and\nreporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThere is a set of default data which needs to be ensured at process\nstartup time. In production this is done via the database migration.\nBecause of the way the database-driven functional tests work the row\nadded in the migration is rolled back with each test, so it is manually\nadded back. The default data is one row in the table: ID 1 with value\nUNKNOWN. This is the default value for an unspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nThe depends on is to a grenade change that upgrades placement\nindependent of nova. Neeed now because we need to do our own\ndb sync as a result of this new table.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nDepends-On: https://review.opendev.org/679655\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/70aeec852b8b1bb9973cd77c20fb28ea6eaacf10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/70aeec852b8b1bb9973cd77c20fb28ea6eaacf10"}]},"branch":"refs/heads/master"},"62159ce3c1bc8d56dd5465c3f1634ff882ef46d7":{"kind":"REWORK","_number":11,"created":"2020-02-07 02:39:57.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/11"}}},"commit":{"parents":[{"commit":"971c7aa76e1a3455acf8cf2210b3b7af29846696","subject":"Provide more accurate links in doc/source/user/provider-tree.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/971c7aa76e1a3455acf8cf2210b3b7af29846696"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-02-07 01:31:58.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and object\ncode to support saving consumer types when creating consumers and\nreporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThere is a set of default data which needs to be ensured at process\nstartup time. In production this is done via the database migration.\nBecause of the way the database-driven functional tests work the row\nadded in the migration is rolled back with each test, so it is manually\nadded back. The default data is one row in the table: ID 1 with value\nUNKNOWN. This is the default value for an unspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nThe depends on is to a grenade change that upgrades placement\nindependent of nova. Neeed now because we need to do our own\ndb sync as a result of this new table.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nDepends-On: https://review.opendev.org/679655\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/62159ce3c1bc8d56dd5465c3f1634ff882ef46d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/62159ce3c1bc8d56dd5465c3f1634ff882ef46d7"}]},"branch":"refs/heads/master"},"32d13b04825f0dd63784d168b1b106e42980df24":{"kind":"REWORK","_number":12,"created":"2020-02-14 04:12:06.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/12"}}},"commit":{"parents":[{"commit":"971c7aa76e1a3455acf8cf2210b3b7af29846696","subject":"Provide more accurate links in doc/source/user/provider-tree.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/971c7aa76e1a3455acf8cf2210b3b7af29846696"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-02-14 04:10:55.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and object\ncode to support saving consumer types when creating consumers and\nreporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThere is a set of default data which needs to be ensured at process\nstartup time. In production this is done via the database migration.\nBecause of the way the database-driven functional tests work the row\nadded in the migration is rolled back with each test, so it is manually\nadded back. The default data is one row in the table: ID 1 with value\nUNKNOWN. This is the default value for an unspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nThe depends on is to a grenade change that upgrades placement\nindependent of nova. Neeed now because we need to do our own\ndb sync as a result of this new table.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nDepends-On: https://review.opendev.org/679655\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/32d13b04825f0dd63784d168b1b106e42980df24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/32d13b04825f0dd63784d168b1b106e42980df24"}]},"branch":"refs/heads/master"},"e01ef1f99db0da004f281538d795204a729e09aa":{"kind":"REWORK","_number":13,"created":"2020-02-25 22:36:30.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/13"}}},"commit":{"parents":[{"commit":"971c7aa76e1a3455acf8cf2210b3b7af29846696","subject":"Provide more accurate links in doc/source/user/provider-tree.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/971c7aa76e1a3455acf8cf2210b3b7af29846696"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-02-25 22:36:13.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and object\ncode to support saving consumer types when creating consumers and\nreporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThere is a set of default data which needs to be ensured at process\nstartup time. In production this is done via the database migration.\nBecause of the way the database-driven functional tests work the row\nadded in the migration is rolled back with each test, so it is manually\nadded back. The default data is one row in the table: ID 1 with value\nUNKNOWN. This is the default value for an unspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nThe depends on is to a grenade change that upgrades placement\nindependent of nova. Neeed now because we need to do our own\ndb sync as a result of this new table.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nDepends-On: https://review.opendev.org/679655\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e01ef1f99db0da004f281538d795204a729e09aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/e01ef1f99db0da004f281538d795204a729e09aa"}]},"branch":"refs/heads/master"},"5bec6b98267869a5e36a8816148dd2217407ca34":{"kind":"REWORK","_number":14,"created":"2020-03-05 05:51:25.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/14"}}},"commit":{"parents":[{"commit":"971c7aa76e1a3455acf8cf2210b3b7af29846696","subject":"Provide more accurate links in doc/source/user/provider-tree.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/971c7aa76e1a3455acf8cf2210b3b7af29846696"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-05 05:43:32.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and\nobject code to support saving consumer types when creating consumers\nand reporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThere is a set of default data which needs to be ensured at process\nstartup time. In production this is done via the database migration.\nThe default data is one row in the table: ID 1 with value UNKNOWN.\nThis is the default value for an unspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nThe depends on is to a grenade change that upgrades placement\nindependent of nova. Needed now because we need to do our own\ndb sync as a result of this new table.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nDepends-On: https://review.opendev.org/679655\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\nStory: 2005473\nTask: 35683\nTask: 36420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5bec6b98267869a5e36a8816148dd2217407ca34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5bec6b98267869a5e36a8816148dd2217407ca34"}]},"branch":"refs/heads/master"},"ed1180f32959e63aa4b2564102f90011e47f6ed4":{"kind":"REWORK","_number":15,"created":"2020-03-11 00:54:37.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/15"}}},"commit":{"parents":[{"commit":"971c7aa76e1a3455acf8cf2210b3b7af29846696","subject":"Provide more accurate links in doc/source/user/provider-tree.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/971c7aa76e1a3455acf8cf2210b3b7af29846696"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2020-03-11 00:04:08.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and\nobject code to support saving consumer types when creating consumers\nand reporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThere is a set of default data which needs to be ensured at process\nstartup time. In production this is done via the database migration.\nThe default data is one row in the table: ID 1 with value UNKNOWN.\nThis is the default value for an unspecified consumer type.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nThe depends on is to a grenade change that upgrades placement\nindependent of nova. Needed now because we need to do our own\ndb sync as a result of this new table.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nCo-Authored-By: melanie witt \u003cmelwittt@gmail.com\u003e\n\nStory: 2005473\nTask: 35683\nTask: 36420\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ed1180f32959e63aa4b2564102f90011e47f6ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ed1180f32959e63aa4b2564102f90011e47f6ed4"}]},"branch":"refs/heads/master"},"9243b780041a6bda29a70572752ce8e046e18538":{"kind":"REWORK","_number":16,"created":"2021-04-21 01:22:06.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/16"}}},"commit":{"parents":[{"commit":"91b0ff88498a774b6040ff4d6bde6ae245be9943","subject":"Merge \"Correctly handle integrity errors on MySQL 8.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/91b0ff88498a774b6040ff4d6bde6ae245be9943"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-04-21 00:46:45.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and\nobject code to support saving consumer types when creating consumers\nand reporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThe consumer_type_id is nullable and records with a NULL\nconsumer_type_id are considered as \u0027unknown\u0027.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nCo-Authored-By: melanie witt \u003cmelwittt@gmail.com\u003e\n\nStory: 2005473\nTask: 35683\nTask: 36420\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9243b780041a6bda29a70572752ce8e046e18538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9243b780041a6bda29a70572752ce8e046e18538"}]},"branch":"refs/heads/master"},"bad71a78d76a0f8b3d5d34bf9371aa7fb96c11c8":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-07-22 01:34:29.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/17"}}},"commit":{"parents":[{"commit":"8d4c1c20f3a10f13bd1e1bbf818982456c42928e","subject":"Merge \"Bump os-resource-classes requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/8d4c1c20f3a10f13bd1e1bbf818982456c42928e"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-22 01:33:05.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.37, add the necessary database and\nobject code to support saving consumer types when creating consumers\nand reporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThe consumer_type_id is nullable and records with a NULL\nconsumer_type_id are considered as \u0027unknown\u0027.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.37 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nCo-Authored-By: melanie witt \u003cmelwittt@gmail.com\u003e\n\nStory: 2005473\nTask: 35683\nTask: 36420\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bad71a78d76a0f8b3d5d34bf9371aa7fb96c11c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/bad71a78d76a0f8b3d5d34bf9371aa7fb96c11c8"}]},"branch":"refs/heads/master"},"9b8effb9bc5a6bafe47ed75f6d4964e09c53abe1":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":18,"created":"2021-07-23 20:35:35.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/18"}}},"commit":{"parents":[{"commit":"c4d785cc354a80a53f20c678557b89f0b56ef0cf","subject":"Merge \"Add support for RP re-parenting and orphaning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c4d785cc354a80a53f20c678557b89f0b56ef0cf"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-23 20:35:17.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.38, add the necessary database and\nobject code to support saving consumer types when creating consumers\nand reporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThe consumer_type_id is nullable and records with a NULL\nconsumer_type_id are considered as \u0027unknown\u0027.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.38 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nCo-Authored-By: melanie witt \u003cmelwittt@gmail.com\u003e\n\nStory: 2005473\nTask: 35683\nTask: 36420\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9b8effb9bc5a6bafe47ed75f6d4964e09c53abe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9b8effb9bc5a6bafe47ed75f6d4964e09c53abe1"}]},"branch":"refs/heads/master"},"c3554d504f732f4dfbe37f7f2ead9e4a7d81ba34":{"kind":"REWORK","_number":19,"created":"2021-07-27 21:58:29.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/19"}}},"commit":{"parents":[{"commit":"c4d785cc354a80a53f20c678557b89f0b56ef0cf","subject":"Merge \"Add support for RP re-parenting and orphaning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c4d785cc354a80a53f20c678557b89f0b56ef0cf"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-07-27 21:57:58.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.38, add the necessary database and\nobject code to support saving consumer types when creating consumers\nand reporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThe consumer_type_id is nullable and records with a NULL\nconsumer_type_id are considered as \u0027unknown\u0027.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.38 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nCo-Authored-By: melanie witt \u003cmelwittt@gmail.com\u003e\n\nStory: 2005473\nTask: 35683\nTask: 36420\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c3554d504f732f4dfbe37f7f2ead9e4a7d81ba34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c3554d504f732f4dfbe37f7f2ead9e4a7d81ba34"}]},"branch":"refs/heads/master"},"b1f3dd39c3fd140ea77fba474c8a3a118a41de8c":{"kind":"REWORK","_number":20,"created":"2021-08-05 23:23:57.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/70/669170/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/70/669170/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/70/669170/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/70/669170/20"}}},"commit":{"parents":[{"commit":"c4d785cc354a80a53f20c678557b89f0b56ef0cf","subject":"Merge \"Add support for RP re-parenting and orphaning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/c4d785cc354a80a53f20c678557b89f0b56ef0cf"}]}],"author":{"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","date":"2019-07-04 12:41:02.000000000","tz":120},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2021-08-05 23:23:46.000000000","tz":0},"subject":"Add consumer_types migration, database and object changes","message":"Add consumer_types migration, database and object changes\n\nIn advance of a microversion 1.38, add the necessary database and\nobject code to support saving consumer types when creating consumers\nand reporting those types when requesting usage information.\n\nThis patch adds a new table called ``consumer_types`` and also\nadd a new column called ``consumer_type_id`` that references the id\nin the consumer_types table.\n\nThe consumer_type_id is nullable and records with a NULL\nconsumer_type_id are considered as \u0027unknown\u0027.\n\nIn future changes we may wish to AttributeCache consumer types in\nthe same way that traits and resource classes are cached.\n\nUsage object can now be retrieved by consumer type and include\nconsumer type information when retrieved. Until 1.38 that info\nwill not be used in results.\n\nAllocations are written and retrieved with a consumer type but\nthe data is not yet presented in HTTP requests.\n\nCo-Authored-By: Chris Dent \u003ccdent@anticdent.org\u003e\nCo-Authored-By: melanie witt \u003cmelwittt@gmail.com\u003e\n\nStory: 2005473\nTask: 35683\nTask: 36420\n\nChange-Id: Ic74c3d1b8aa272e9c7d1070fe7658b928e693889\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b1f3dd39c3fd140ea77fba474c8a3a118a41de8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b1f3dd39c3fd140ea77fba474c8a3a118a41de8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
