)]}'
{"id":"openstack%2Fplacement~669810","triplet_id":"openstack%2Fplacement~master~I6ffc19eea84b45dd452bc61d2a2007b6572719ad","project":"openstack/placement","branch":"master","topic":"trivial","hashtags":[],"change_id":"I6ffc19eea84b45dd452bc61d2a2007b6572719ad","subject":"Trivial: Fix docs for functions","status":"MERGED","created":"2019-07-09 07:52:08.000000000","updated":"2019-07-09 14:29:09.000000000","submitted":"2019-07-09 14:29:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669810-1562682549462-5870c8eb","meta_rev_id":"437c2104e4c8e20852c2c18fa035f60c1157e2d4","_number":669810,"virtual_id_number":669810,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-07-09 14:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-07-09 10:32:59.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-07-09 07:53:53.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":1,"date":"2019-07-09 10:32:59.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 07:53:53.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-07-09 10:32:59.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-09 14:29:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02728f4fc3edd417063f47f6c2723641cc449630","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-07-09 07:52:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e360e4199e123bf19d19889e2fb6868c7086d15","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-09 07:53:53.000000000","message":"Patch Set 1: Code-Review+2\n\ngood improvements","accounts_in_message":[],"_revision_number":1},{"id":"8632d5a626f5a29fc15678cf3997955f96bc5e82","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-09 10:32:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"880b86ea841bb3fb0b3123ff8cab6306ca4ff673","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 11:16:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/10/669810/1/check/grenade-py3/e08b17c/ : SUCCESS in 58m 13s\n- tempest-full-py3 http://logs.openstack.org/10/669810/1/check/tempest-full-py3/d518cef/ : SUCCESS in 1h 38m 32s\n- openstack-tox-docs http://logs.openstack.org/10/669810/1/check/openstack-tox-docs/a65e525/html/ : SUCCESS in 4m 21s\n- openstack-tox-cover http://logs.openstack.org/10/669810/1/check/openstack-tox-cover/c9035e4/cover/ : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/669810/1/check/openstack-tox-lower-constraints/b150a4a/ : SUCCESS in 5m 28s\n- openstack-tox-pep8 http://logs.openstack.org/10/669810/1/check/openstack-tox-pep8/0370c38/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/10/669810/1/check/openstack-tox-py27/d34be56/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/10/669810/1/check/openstack-tox-py36/1d2bafc/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/10/669810/1/check/openstack-tox-py37/fb1cc63/ : SUCCESS in 5m 24s\n- openstack-tox-functional http://logs.openstack.org/10/669810/1/check/openstack-tox-functional/d375989/ : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/10/669810/1/check/openstack-tox-functional-py36/c49e279/ : SUCCESS in 5m 02s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/10/669810/1/check/placement-nova-tox-functional-py36/a41f962/ : SUCCESS in 11m 25s\n- placement-nested-perfload http://logs.openstack.org/10/669810/1/check/placement-nested-perfload/4f52799/ : SUCCESS in 16m 17s (non-voting)\n- placement-perfload http://logs.openstack.org/10/669810/1/check/placement-perfload/ba48b92/ : SUCCESS in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"090541d24eacd6549c48e7a281998548eda5fd25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 11:16:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"afc8b84af935747f2e765d82a3c65464a4a58ebe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:29:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44d15786380af4b168cff3a9d5a8f6e8743c28ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 14:29:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/10/669810/1/gate/grenade-py3/0db6f2d/ : SUCCESS in 52m 27s\n- tempest-full-py3 http://logs.openstack.org/10/669810/1/gate/tempest-full-py3/d3c735d/ : SUCCESS in 1h 36m 25s\n- openstack-tox-docs http://logs.openstack.org/10/669810/1/gate/openstack-tox-docs/772be72/html/ : SUCCESS in 4m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/669810/1/gate/openstack-tox-lower-constraints/bb4743f/ : SUCCESS in 6m 55s\n- openstack-tox-pep8 http://logs.openstack.org/10/669810/1/gate/openstack-tox-pep8/8c9de81/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/10/669810/1/gate/openstack-tox-py27/10c3afd/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/10/669810/1/gate/openstack-tox-py36/07bb64d/ : SUCCESS in 4m 30s\n- openstack-tox-py37 http://logs.openstack.org/10/669810/1/gate/openstack-tox-py37/7afbf16/ : SUCCESS in 5m 10s\n- openstack-tox-functional http://logs.openstack.org/10/669810/1/gate/openstack-tox-functional/ccb6da3/ : SUCCESS in 5m 04s\n- openstack-tox-functional-py36 http://logs.openstack.org/10/669810/1/gate/openstack-tox-functional-py36/1fdf866/ : SUCCESS in 4m 41s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/10/669810/1/gate/placement-nova-tox-functional-py36/1e9127b/ : SUCCESS in 12m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"23300f2813e4ae02c566122896b05e883c3ec1c3","revisions":{"23300f2813e4ae02c566122896b05e883c3ec1c3":{"kind":"REWORK","_number":1,"created":"2019-07-09 07:52:08.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/10/669810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/10/669810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/10/669810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/10/669810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/10/669810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/10/669810/1"}}},"commit":{"parents":[{"commit":"7f72d1e30e7e7bc04cded154731ef391d928a59b","subject":"Merge \"tox: Stop building api-ref docs with the main docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7f72d1e30e7e7bc04cded154731ef391d928a59b"}]}],"author":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-07-09 07:42:56.000000000","tz":0},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-07-09 07:44:39.000000000","tz":0},"subject":"Trivial: Fix docs for functions","message":"Trivial: Fix docs for functions\n\nThis patch fixes some comments in code:\n - Now suffixes can be string in _merge_candidates()\n - Remove old doc for `allow_forbidden` parameter in\n   RequestGroup.dict_from_request\n - Delete unnecessary space in ResourceClassCache\n\nChange-Id: I6ffc19eea84b45dd452bc61d2a2007b6572719ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/23300f2813e4ae02c566122896b05e883c3ec1c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/23300f2813e4ae02c566122896b05e883c3ec1c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
