)]}'
{"id":"openstack%2Fplacement~670553","triplet_id":"openstack%2Fplacement~master~Id8f96801116818dda6c68c301619aee185e8fbff","project":"openstack/placement","branch":"master","topic":"cd/remove-redundant-name-finding","hashtags":[],"change_id":"Id8f96801116818dda6c68c301619aee185e8fbff","subject":"Simplify placement.microversion:_fully_qualified_name","status":"MERGED","created":"2019-07-12 13:33:33.000000000","updated":"2019-07-23 16:22:32.000000000","submitted":"2019-07-23 16:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670553-1563898952776-83992aa8","meta_rev_id":"61fdad95ba4f458b90517334dec70d629d3aaf60","_number":670553,"virtual_id_number":670553,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-07-23 16:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2019-07-23 10:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-23 12:47:55.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-07-23 14:17:13.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 10:44:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-23 14:17:13.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-23 16:22:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d043bd2a2cdf76a94f7d91d612fa4027a8d0ed82","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-12 13:33:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95bb467c79bbea16a93f9c10776603064a3234c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 15:09:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/53/670553/1/check/grenade-py3/b1ec2dc/ : SUCCESS in 1h 03m 03s\n- tempest-full-py3 http://logs.openstack.org/53/670553/1/check/tempest-full-py3/4102f6d/ : SUCCESS in 1h 34m 11s\n- openstack-tox-docs http://logs.openstack.org/53/670553/1/check/openstack-tox-docs/76892d9/html/ : SUCCESS in 4m 45s\n- openstack-tox-cover http://logs.openstack.org/53/670553/1/check/openstack-tox-cover/d76a0b4/cover/ : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/670553/1/check/openstack-tox-lower-constraints/86d4752/ : SUCCESS in 7m 00s\n- openstack-tox-pep8 http://logs.openstack.org/53/670553/1/check/openstack-tox-pep8/ac1660b/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/53/670553/1/check/openstack-tox-py27/209b14f/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/53/670553/1/check/openstack-tox-py36/32c04f7/ : SUCCESS in 4m 08s\n- openstack-tox-py37 http://logs.openstack.org/53/670553/1/check/openstack-tox-py37/3af8e86/ : SUCCESS in 5m 07s\n- openstack-tox-functional http://logs.openstack.org/53/670553/1/check/openstack-tox-functional/37c6472/ : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 http://logs.openstack.org/53/670553/1/check/openstack-tox-functional-py36/7e77ff1/ : SUCCESS in 5m 15s\n- placement-nova-tox-functional-py36 http://logs.openstack.org/53/670553/1/check/placement-nova-tox-functional-py36/3b60a6e/ : SUCCESS in 11m 02s\n- placement-nested-perfload http://logs.openstack.org/53/670553/1/check/placement-nested-perfload/d70e547/ : SUCCESS in 19m 52s (non-voting)\n- placement-perfload http://logs.openstack.org/53/670553/1/check/placement-perfload/f469c2e/ : SUCCESS in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"211c180839e478d784c85db3b34c87589173d3f1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-23 10:44:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12e3d25413970d608e1f22519a79f99113175053","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-23 12:47:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)\n\nI confirmed there\u0027s only module-level functions calling in here. That being the case, further simplification is possible, noted inline. And another, unrelated simplification.","accounts_in_message":[],"_revision_number":1},{"id":"53991fcc423f566da83aeb8d619753ef374d62da","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-23 14:01:58.000000000","message":"Patch Set 1:\n\nAddressed my comments here [1]. We can squash or keep separate, whatever.\n\n[1] https://review.opendev.org/#/c/672293/","accounts_in_message":[],"_revision_number":1},{"id":"7e6ebdb99b3f3676e4a3b00bcb99be4857559a3e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-23 14:17:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca82db0f147d2e414e58f8c57b920001e7c3e799","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:17:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e05cf6b0e60218ded2445d4f6537e9291e0aee6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 16:22:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/53/670553/1/gate/grenade-py3/98b8852/ : SUCCESS in 1h 05m 10s\n- tempest-full-py3 http://logs.openstack.org/53/670553/1/gate/tempest-full-py3/c96aab6/ : SUCCESS in 1h 29m 55s\n- openstack-tox-docs http://logs.openstack.org/53/670553/1/gate/openstack-tox-docs/4823053/html/ : SUCCESS in 5m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/53/670553/1/gate/openstack-tox-lower-constraints/8ac5a6a/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/53/670553/1/gate/openstack-tox-pep8/ca052f5/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/53/670553/1/gate/openstack-tox-py27/d3c2fe9/ : SUCCESS in 5m 01s\n- openstack-tox-py36 http://logs.openstack.org/53/670553/1/gate/openstack-tox-py36/1f30afa/ : SUCCESS in 4m 23s\n- openstack-tox-py37 http://logs.openstack.org/53/670553/1/gate/openstack-tox-py37/15f5ab7/ : SUCCESS in 6m 47s\n- openstack-tox-functional http://logs.openstack.org/53/670553/1/gate/openstack-tox-functional/195aa7f/ : SUCCESS in 7m 18s\n- openstack-tox-functional-py36 http://logs.openstack.org/53/670553/1/gate/openstack-tox-functional-py36/0d9538d/ : SUCCESS in 5m 36s\n- openstack-tox-functional-py37 http://logs.openstack.org/53/670553/1/gate/openstack-tox-functional-py37/01f4342/ : SUCCESS in 6m 32s\n- placement-nova-tox-functional-py37 http://logs.openstack.org/53/670553/1/gate/placement-nova-tox-functional-py37/225d0df/ : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a0b6095877014d4e969b8f8459474a743e0cb8aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 16:22:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2b452d7f5e03970a5341d84066f5ce0d1997f8e","revisions":{"d2b452d7f5e03970a5341d84066f5ce0d1997f8e":{"kind":"REWORK","_number":1,"created":"2019-07-12 13:33:33.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/53/670553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/53/670553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/53/670553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/53/670553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/53/670553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/53/670553/1"}}},"commit":{"parents":[{"commit":"b9cbda42a1833d4839e93c69c04e422092ec54a6","subject":"Merge \"Doc `same_subtree` queryparam\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b9cbda42a1833d4839e93c69c04e422092ec54a6"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-07-12 13:27:41.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-07-12 13:27:41.000000000","tz":60},"subject":"Simplify placement.microversion:_fully_qualified_name","message":"Simplify placement.microversion:_fully_qualified_name\n\n_fully_qualified_name was copied from Twisted. The original\nversion does more than Placement needs. We only need to\nhandle module-level classes and functions (and really only\nfunctions). No nested of those, nor object methods in any\nform.\n\nThis is because we only use the method for finding the\nright microversioned handler of a method with the same\nname and placement HTTP handlers are universally functions.\n\nThe previous version of the method had code that was never\ncalled. Here, that dead code is removed and the remainder\nis simplified.\n\nIf it is called incorrectly there will be an (intentionally)\nuntrapped TypeError during running of the tests.\n\nN.B.: This was discovered while reviewing coverage data.\n\nChange-Id: Id8f96801116818dda6c68c301619aee185e8fbff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d2b452d7f5e03970a5341d84066f5ce0d1997f8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/d2b452d7f5e03970a5341d84066f5ce0d1997f8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
