)]}'
{"id":"openstack%2Fplacement~672110","triplet_id":"openstack%2Fplacement~master~Iccddd4356b5d7c28977ad954654d0875972dae97","project":"openstack/placement","branch":"master","topic":"retire-api-site","hashtags":[],"change_id":"Iccddd4356b5d7c28977ad954654d0875972dae97","subject":"Update api-ref location","status":"MERGED","created":"2019-07-22 16:56:44.000000000","updated":"2019-07-23 12:23:33.000000000","submitted":"2019-07-23 12:23:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672110-1563884613638-586d1149","meta_rev_id":"b217cb706c4c68217545df37b47ea63dc21f9923","_number":672110,"virtual_id_number":672110,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-07-23 12:23:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-07-23 10:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-23 10:31:49.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"date":"2019-07-22 18:34:06.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-07-23 10:59:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 18:34:06.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-23 10:31:49.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-07-23 10:59:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-23 12:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af8140d0444147340157709f94ea3deb52455481","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 16:56:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76ca125484923b72c659a0a46abb515c3ae4824b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 17:17:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"962af1f9196bd838e19cc2d15bfefe3d766b9929","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:37:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/10/672110/2/check/build-openstack-api-ref/f42a72a/html/ : SUCCESS in 5m 06s\n- test-release-openstack http://logs.openstack.org/10/672110/2/check/test-release-openstack/c987e61/ : SUCCESS in 4m 02s\n- openstack-tox-docs http://logs.openstack.org/10/672110/2/check/openstack-tox-docs/3fa5055/html/ : SUCCESS in 5m 28s\n- openstack-tox-cover http://logs.openstack.org/10/672110/2/check/openstack-tox-cover/67f5feb/cover/ : SUCCESS in 6m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/672110/2/check/openstack-tox-lower-constraints/79bf142/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/10/672110/2/check/openstack-tox-pep8/f88b60a/ : SUCCESS in 6m 29s\n- openstack-tox-py27 http://logs.openstack.org/10/672110/2/check/openstack-tox-py27/bf3e2f4/ : SUCCESS in 6m 04s\n- openstack-tox-py36 http://logs.openstack.org/10/672110/2/check/openstack-tox-py36/154c7e4/ : SUCCESS in 6m 12s\n- openstack-tox-py37 http://logs.openstack.org/10/672110/2/check/openstack-tox-py37/b8d3b2a/ : SUCCESS in 7m 34s\n- build-openstack-releasenotes http://logs.openstack.org/10/672110/2/check/build-openstack-releasenotes/8d37283/html/ : SUCCESS in 5m 10s\n- openstack-tox-functional http://logs.openstack.org/10/672110/2/check/openstack-tox-functional/d6fb80c/ : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 http://logs.openstack.org/10/672110/2/check/openstack-tox-functional-py36/1edb18c/ : SUCCESS in 6m 05s\n- openstack-tox-functional-py37 http://logs.openstack.org/10/672110/2/check/openstack-tox-functional-py37/fb468c2/ : SUCCESS in 6m 42s\n- placement-nova-tox-functional-py37 http://logs.openstack.org/10/672110/2/check/placement-nova-tox-functional-py37/6aafc41/ : SUCCESS in 15m 37s","accounts_in_message":[],"_revision_number":2},{"id":"c2be4da65c3a803b3f57cdf6a568604b2019545f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-23 10:31:49.000000000","message":"Patch Set 2: Code-Review+2\n\nI feel weird about this for reasons I can\u0027t quite pin down, but it appears to be the desired outcome, so... \n\nI noticed a wart about microversions which I\u0027ll fix in a fup.","accounts_in_message":[],"_revision_number":2},{"id":"f3d7cf851816e31d005052f9914f25201997a2da","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-23 10:59:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11974d2938dfb55d9c8f5a2ab0571757121e53f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 10:59:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30b33ab150688f1d9c1b87adcc45c872179a1cf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:23:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2e2f3ea33fc69521809b6cc4510a4957864cc58a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:23:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/10/672110/2/gate/build-openstack-api-ref/20a7c41/html/ : SUCCESS in 4m 48s\n- test-release-openstack http://logs.openstack.org/10/672110/2/gate/test-release-openstack/3743f3f/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/10/672110/2/gate/openstack-tox-docs/5195d33/html/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/10/672110/2/gate/openstack-tox-lower-constraints/419265c/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/10/672110/2/gate/openstack-tox-pep8/5bfdb9a/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/10/672110/2/gate/openstack-tox-py27/d46ac07/ : SUCCESS in 7m 42s\n- openstack-tox-py36 http://logs.openstack.org/10/672110/2/gate/openstack-tox-py36/49872ad/ : SUCCESS in 6m 04s\n- openstack-tox-py37 http://logs.openstack.org/10/672110/2/gate/openstack-tox-py37/8bf4884/ : SUCCESS in 5m 19s\n- build-openstack-releasenotes http://logs.openstack.org/10/672110/2/gate/build-openstack-releasenotes/6e1d50d/html/ : SUCCESS in 4m 43s\n- openstack-tox-functional http://logs.openstack.org/10/672110/2/gate/openstack-tox-functional/e347e61/ : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 http://logs.openstack.org/10/672110/2/gate/openstack-tox-functional-py36/2a77830/ : SUCCESS in 5m 44s\n- openstack-tox-functional-py37 http://logs.openstack.org/10/672110/2/gate/openstack-tox-functional-py37/7c8f19e/ : SUCCESS in 8m 19s\n- placement-nova-tox-functional-py37 http://logs.openstack.org/10/672110/2/gate/placement-nova-tox-functional-py37/1034f69/ : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b3c179fe4a67c99e366c6420e1b387a14d8e871","revisions":{"346ea245e5098b886f70809708c4da01b1b36b44":{"kind":"REWORK","_number":1,"created":"2019-07-22 16:56:44.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/10/672110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/10/672110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/10/672110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/10/672110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/10/672110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/10/672110/1"}}},"commit":{"parents":[{"commit":"9e8040bf0cb4278f45bc23670e8a639b6b242997","subject":"Merge \"Move rc_cache onto RequestContext\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9e8040bf0cb4278f45bc23670e8a639b6b242997"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:56:36.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:56:36.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: Iccddd4356b5d7c28977ad954654d0875972dae97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/346ea245e5098b886f70809708c4da01b1b36b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/346ea245e5098b886f70809708c4da01b1b36b44"}]},"branch":"refs/heads/master"},"7b3c179fe4a67c99e366c6420e1b387a14d8e871":{"kind":"REWORK","_number":2,"created":"2019-07-22 17:17:09.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/10/672110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/10/672110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/10/672110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/10/672110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/10/672110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/10/672110/2"}}},"commit":{"parents":[{"commit":"9e8040bf0cb4278f45bc23670e8a639b6b242997","subject":"Merge \"Move rc_cache onto RequestContext\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9e8040bf0cb4278f45bc23670e8a639b6b242997"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:56:36.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 17:17:01.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: Iccddd4356b5d7c28977ad954654d0875972dae97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7b3c179fe4a67c99e366c6420e1b387a14d8e871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7b3c179fe4a67c99e366c6420e1b387a14d8e871"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
