)]}'
{"id":"openstack%2Fplacement~673505","triplet_id":"openstack%2Fplacement~master~Ie4809abc31212711b96f69e5f291104ae761059e","project":"openstack/placement","branch":"master","topic":"story/2005443","hashtags":[],"change_id":"Ie4809abc31212711b96f69e5f291104ae761059e","subject":"Run nested-perfload parallel correctly","status":"MERGED","created":"2019-07-30 12:49:26.000000000","updated":"2019-08-07 16:11:45.000000000","submitted":"2019-08-07 16:11:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673505-1565194305405-ca213fc5","meta_rev_id":"d95d0898155b6edca78f8afacd7232185ffd929b","_number":673505,"virtual_id_number":673505,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":2,"date":"2019-08-07 16:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-07 12:13:23.000000000","_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":2,"date":"2019-08-06 15:12:29.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-08-07 00:44:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-08-07 09:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":1,"date":"2019-08-07 09:09:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 15:12:29.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-07 00:44:45.000000000","updated_by":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"reviewer":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"state":"REVIEWER"},{"updated":"2019-08-07 09:09:45.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-08-07 16:11:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e93998a51a1b4d22b0d0608fb0e2e119648568e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-30 12:49:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f80b6a9e871b0a456ab39d70fc796d52183943c9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-30 15:50:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac30bbb3ca257bf4229b37ee978ffdcee821acdf","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-30 15:54:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b374d415ed8c09887694548df6c5c9b6ad8bb5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 16:40:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/05/673505/1/check/grenade-py3/5cf5a16/ : SUCCESS in 1h 02m 48s\n- tempest-integrated-placement http://logs.openstack.org/05/673505/1/check/tempest-integrated-placement/0c7f623/ : SUCCESS in 1h 19m 46s\n- openstack-tox-docs http://logs.openstack.org/05/673505/1/check/openstack-tox-docs/1dce848/html/ : SUCCESS in 6m 32s\n- openstack-tox-cover http://logs.openstack.org/05/673505/1/check/openstack-tox-cover/688cc25/cover/ : SUCCESS in 8m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/673505/1/check/openstack-tox-lower-constraints/f074895/ : SUCCESS in 6m 27s\n- openstack-tox-pep8 http://logs.openstack.org/05/673505/1/check/openstack-tox-pep8/42ee4d8/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/05/673505/1/check/openstack-tox-py27/2e051e8/ : SUCCESS in 6m 13s\n- openstack-tox-py36 http://logs.openstack.org/05/673505/1/check/openstack-tox-py36/dc72b46/ : SUCCESS in 5m 49s\n- openstack-tox-py37 http://logs.openstack.org/05/673505/1/check/openstack-tox-py37/cf7f7fe/ : SUCCESS in 6m 27s\n- openstack-tox-functional http://logs.openstack.org/05/673505/1/check/openstack-tox-functional/db20964/ : SUCCESS in 4m 28s\n- openstack-tox-functional-py36 http://logs.openstack.org/05/673505/1/check/openstack-tox-functional-py36/24cd37c/ : SUCCESS in 4m 14s\n- openstack-tox-functional-py37 http://logs.openstack.org/05/673505/1/check/openstack-tox-functional-py37/6528573/ : SUCCESS in 6m 03s\n- placement-nova-tox-functional-py37 http://logs.openstack.org/05/673505/1/check/placement-nova-tox-functional-py37/d86b6a9/ : SUCCESS in 12m 49s\n- placement-nested-perfload http://logs.openstack.org/05/673505/1/check/placement-nested-perfload/5b008b3/ : SUCCESS in 8m 04s (non-voting)\n- placement-perfload http://logs.openstack.org/05/673505/1/check/placement-perfload/4bb912c/ : SUCCESS in 8m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc93a9d106967430e5642699d067313061d907c8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-30 16:50:11.000000000","message":"Patch Set 1:\n\nIs it possible to tell the difference from the job results?","accounts_in_message":[],"_revision_number":1},{"id":"043839909ab13928075802f65c285c54c3e314ed","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-30 16:58:54.000000000","message":"Patch Set 1:\n\n\u003e Is it possible to tell the difference from the job results?\n\nnot directly. you can see how long it takes here: https://opendev.org/openstack/placement/src/commit/2f56f379a568d427ac9057086386074a7801b00c/placement/objects/resource_class.py#L72\n\nand I suspect that analysing the placement-api.log would show some different patterns. Compare the first few requests of http://logs.openstack.org/67/670567/3/check/placement-nested-perfload/fc76783/logs/placement-api.log with http://logs.openstack.org/67/670567/3/check/placement-nested-perfload/fc76783/logs/placement-api.log\n\nIn the latter (this patch) there are several POST to create resource providers at the same time.","accounts_in_message":[],"_revision_number":1},{"id":"8607be0de69f83d12dc4dc4de6c68ca3f77e17a0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-30 17:07:44.000000000","message":"Patch Set 1: Code-Review+2\n\nassume that second link was supposed to be\n\nhttp://logs.openstack.org/05/673505/1/check/placement-nested-perfload/5b008b3/logs/placement-api.log\n\nAnd \"at the same time\" is \"within the order of 10ms vs 100ms of each other\" which, while not probative, is at least somewhat convincing.\n\nI suppose if we logged thread IDs as well as process IDs it might be clearer?\n\nAnyway, this isn\u0027t breaking anything, so I see no harm in merging it.","accounts_in_message":[],"_revision_number":1},{"id":"dfddf0f6993c2b0608a66774be366ae8024208c8","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-31 10:19:30.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3280de77dafc2884673c233099b79ac9ea2fa4a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 13:00:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/05/673505/2/check/grenade-py3/fe08e0f/ : SUCCESS in 1h 03m 16s\n- tempest-integrated-placement http://logs.openstack.org/05/673505/2/check/tempest-integrated-placement/fc5d677/ : SUCCESS in 1h 19m 16s\n- openstack-tox-docs http://logs.openstack.org/05/673505/2/check/openstack-tox-docs/2be9ef3/html/ : SUCCESS in 4m 09s\n- openstack-tox-cover http://logs.openstack.org/05/673505/2/check/openstack-tox-cover/fce6662/cover/ : SUCCESS in 6m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/673505/2/check/openstack-tox-lower-constraints/b61bd04/ : SUCCESS in 4m 36s\n- openstack-tox-pep8 http://logs.openstack.org/05/673505/2/check/openstack-tox-pep8/144224a/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/05/673505/2/check/openstack-tox-py27/74dbcdc/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/05/673505/2/check/openstack-tox-py36/7f143f9/ : SUCCESS in 4m 38s\n- openstack-tox-py37 http://logs.openstack.org/05/673505/2/check/openstack-tox-py37/57e9c75/ : SUCCESS in 4m 38s\n- openstack-tox-functional http://logs.openstack.org/05/673505/2/check/openstack-tox-functional/3574c31/ : SUCCESS in 5m 24s\n- openstack-tox-functional-py36 http://logs.openstack.org/05/673505/2/check/openstack-tox-functional-py36/332a10a/ : SUCCESS in 4m 55s\n- openstack-tox-functional-py37 http://logs.openstack.org/05/673505/2/check/openstack-tox-functional-py37/0f4f88b/ : SUCCESS in 7m 12s\n- placement-nova-tox-functional-py37 http://logs.openstack.org/05/673505/2/check/placement-nova-tox-functional-py37/049909a/ : SUCCESS in 12m 57s\n- placement-nested-perfload http://logs.openstack.org/05/673505/2/check/placement-nested-perfload/ae89643/ : SUCCESS in 6m 44s (non-voting)\n- placement-perfload http://logs.openstack.org/05/673505/2/check/placement-perfload/936e33f/ : SUCCESS in 9m 25s (non-voting)\n- tempest-ipv6-only http://logs.openstack.org/05/673505/2/check/tempest-ipv6-only/4855829/ : SUCCESS in 1h 12m 20s","accounts_in_message":[],"_revision_number":2},{"id":"07057acaa789586f2ea7e49b3c4f38ad785cb71f","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-08-06 05:13:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\n-1 because we need to update the doc accordingly.","accounts_in_message":[],"_revision_number":2},{"id":"a50b6ed0e89899a35b7e9e11f5928bcfb24b484e","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-06 08:02:20.000000000","message":"Patch Set 2: Workflow-1\n\nSee, not only can Tetsuro write better English than the natives, he can also read it...\n\nI\u0027ll make some adjustments. \n\nThank you!","accounts_in_message":[],"_revision_number":2},{"id":"f52c5c5b04d093020426a9af5b605b3ece2423cb","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-06 08:19:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"027cfc3304ecf98428a43871698bd4ffe8566b5f","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-06 08:23:00.000000000","message":"Patch Set 3:\n\nI\u0027ve decided to try to share gabbi-run\u0027s cpu usage with the web server and database a bit more fairly, and gone for 50%. This makes sense based on the manual testing I did last week, but we can compare the numbers...\n\nThe previous run [1] was about 2 minutes, it gets a lot slower with next patch (the more complex topology)\n\n[1] http://logs.openstack.org/05/673505/2/check/placement-nested-perfload/ae89643/job-output.txt.gz#_2019-07-31_11_48_34_126349","accounts_in_message":[],"_revision_number":3},{"id":"f1483994884c931a64b49798ddfba3ca0f185d26","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 11:40:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://logs.opendev.org/05/673505/3/check/grenade-py3/983a951/ : SUCCESS in 51m 43s\n- tempest-integrated-placement https://logs.opendev.org/05/673505/3/check/tempest-integrated-placement/7225ed9/ : SUCCESS in 1h 22m 16s\n- openstack-tox-docs https://logs.opendev.org/05/673505/3/check/openstack-tox-docs/5fca3e0/html/ : SUCCESS in 5m 22s\n- openstack-tox-cover https://logs.opendev.org/05/673505/3/check/openstack-tox-cover/ebdd1e1/cover/ : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints https://logs.opendev.org/05/673505/3/check/openstack-tox-lower-constraints/33a9fbd/ : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://logs.opendev.org/05/673505/3/check/openstack-tox-pep8/04b4f0d/ : SUCCESS in 5m 51s\n- openstack-tox-py27 https://logs.opendev.org/05/673505/3/check/openstack-tox-py27/ec514ee/ : SUCCESS in 5m 57s\n- openstack-tox-py36 https://logs.opendev.org/05/673505/3/check/openstack-tox-py36/40bcd0e/ : SUCCESS in 5m 10s\n- openstack-tox-py37 https://logs.opendev.org/05/673505/3/check/openstack-tox-py37/d3de845/ : SUCCESS in 7m 12s\n- openstack-tox-functional https://logs.opendev.org/05/673505/3/check/openstack-tox-functional/1d604a3/ : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://logs.opendev.org/05/673505/3/check/openstack-tox-functional-py36/9b2485b/ : SUCCESS in 7m 02s\n- openstack-tox-functional-py37 https://logs.opendev.org/05/673505/3/check/openstack-tox-functional-py37/c763f50/ : SUCCESS in 7m 06s\n- placement-nova-tox-functional-py37 https://logs.opendev.org/05/673505/3/check/placement-nova-tox-functional-py37/848fb69/ : SUCCESS in 12m 50s\n- placement-nested-perfload https://logs.opendev.org/05/673505/3/check/placement-nested-perfload/815c7d8/ : SUCCESS in 8m 29s (non-voting)\n- placement-perfload https://logs.opendev.org/05/673505/3/check/placement-perfload/b4ce354/ : SUCCESS in 8m 49s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/05/673505/3/check/tempest-ipv6-only/999cda5/ : SUCCESS in 1h 11m 46s","accounts_in_message":[],"_revision_number":3},{"id":"0479263ba1124bd70dcdff8f18bc49a9ce58aaf9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-06 15:12:29.000000000","message":"Patch Set 3: Code-Review+2\n\nNice","accounts_in_message":[],"_revision_number":3},{"id":"7c2db98c761d71e2113d98d524f37addef680787","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-08-07 00:44:45.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69664c49783c35460139fabf46b8bf1d27e11f81","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-07 08:33:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"656c2e6a1bcdd3c4b280f6fad9f1fcbf077e4dc8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-07 09:09:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"0dd650f72edb88f78ce63fffaa96e00c78121317","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 09:16:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3461d66e116b327ee99daea0bf1ece1ea2684670","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 12:00:49.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://logs.opendev.org/05/673505/3/gate/grenade-py3/776d183/ : SUCCESS in 1h 15m 51s\n- tempest-integrated-placement https://logs.opendev.org/05/673505/3/gate/tempest-integrated-placement/c7eea40/ : TIMED_OUT in 2h 06m 38s\n- openstack-tox-docs https://logs.opendev.org/05/673505/3/gate/openstack-tox-docs/299d199/html/ : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints https://logs.opendev.org/05/673505/3/gate/openstack-tox-lower-constraints/f8c9543/ : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://logs.opendev.org/05/673505/3/gate/openstack-tox-pep8/3ae268c/ : SUCCESS in 5m 10s\n- openstack-tox-py27 https://logs.opendev.org/05/673505/3/gate/openstack-tox-py27/d02e1e0/ : SUCCESS in 5m 26s\n- openstack-tox-py36 https://logs.opendev.org/05/673505/3/gate/openstack-tox-py36/c5ec282/ : SUCCESS in 6m 38s\n- openstack-tox-py37 https://logs.opendev.org/05/673505/3/gate/openstack-tox-py37/459f553/ : SUCCESS in 5m 32s\n- openstack-tox-functional https://logs.opendev.org/05/673505/3/gate/openstack-tox-functional/e6a6cf3/ : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://logs.opendev.org/05/673505/3/gate/openstack-tox-functional-py36/486ebf2/ : SUCCESS in 5m 26s\n- openstack-tox-functional-py37 https://logs.opendev.org/05/673505/3/gate/openstack-tox-functional-py37/da96890/ : SUCCESS in 7m 09s\n- placement-nova-tox-functional-py37 https://logs.opendev.org/05/673505/3/gate/placement-nova-tox-functional-py37/8cdda01/ : SUCCESS in 11m 55s\n- tempest-ipv6-only https://logs.opendev.org/05/673505/3/gate/tempest-ipv6-only/74998b5/ : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":3},{"id":"7c6ae5a112eebeb95bdf153e5b2ceb0a73582c09","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-07 12:13:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e485a50f6db4e817a11193a5884e7d57004f6126","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:52:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://logs.opendev.org/05/673505/3/check/grenade-py3/1da8145/ : SUCCESS in 58m 49s\n- tempest-integrated-placement https://logs.opendev.org/05/673505/3/check/tempest-integrated-placement/8e419b7/ : SUCCESS in 1h 30m 02s\n- openstack-tox-docs https://logs.opendev.org/05/673505/3/check/openstack-tox-docs/99c4d42/html/ : SUCCESS in 4m 43s\n- openstack-tox-cover https://logs.opendev.org/05/673505/3/check/openstack-tox-cover/8c814ba/cover/ : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://logs.opendev.org/05/673505/3/check/openstack-tox-lower-constraints/477bbf3/ : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://logs.opendev.org/05/673505/3/check/openstack-tox-pep8/1c47ba8/ : SUCCESS in 5m 07s\n- openstack-tox-py27 https://logs.opendev.org/05/673505/3/check/openstack-tox-py27/55d7117/ : SUCCESS in 6m 24s\n- openstack-tox-py36 https://logs.opendev.org/05/673505/3/check/openstack-tox-py36/482f834/ : SUCCESS in 5m 30s\n- openstack-tox-py37 https://logs.opendev.org/05/673505/3/check/openstack-tox-py37/076e8c5/ : SUCCESS in 7m 16s\n- openstack-tox-functional https://logs.opendev.org/05/673505/3/check/openstack-tox-functional/058395e/ : SUCCESS in 4m 46s\n- openstack-tox-functional-py36 https://logs.opendev.org/05/673505/3/check/openstack-tox-functional-py36/f43ab14/ : SUCCESS in 4m 56s\n- openstack-tox-functional-py37 https://logs.opendev.org/05/673505/3/check/openstack-tox-functional-py37/4d823e2/ : SUCCESS in 5m 57s\n- placement-nova-tox-functional-py37 https://logs.opendev.org/05/673505/3/check/placement-nova-tox-functional-py37/d7be653/ : SUCCESS in 14m 51s\n- placement-nested-perfload https://logs.opendev.org/05/673505/3/check/placement-nested-perfload/0661f5b/ : SUCCESS in 8m 25s (non-voting)\n- placement-perfload https://logs.opendev.org/05/673505/3/check/placement-perfload/89691d0/ : SUCCESS in 8m 37s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/05/673505/3/check/tempest-ipv6-only/0f7869d/ : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":3},{"id":"93d1efc4b2dd29c83f5811d28ac127f3e846b231","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 13:52:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aea5ecb3bc381059282a6b3e9f60311efe5c19e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 16:11:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://logs.opendev.org/05/673505/3/gate/grenade-py3/c9985de/ : SUCCESS in 1h 00m 12s\n- tempest-integrated-placement https://logs.opendev.org/05/673505/3/gate/tempest-integrated-placement/3eedfb9/ : SUCCESS in 1h 21m 07s\n- openstack-tox-docs https://logs.opendev.org/05/673505/3/gate/openstack-tox-docs/bcd7de4/html/ : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://logs.opendev.org/05/673505/3/gate/openstack-tox-lower-constraints/e6042e9/ : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://logs.opendev.org/05/673505/3/gate/openstack-tox-pep8/6d5aefd/ : SUCCESS in 6m 51s\n- openstack-tox-py27 https://logs.opendev.org/05/673505/3/gate/openstack-tox-py27/9b5e521/ : SUCCESS in 5m 03s\n- openstack-tox-py36 https://logs.opendev.org/05/673505/3/gate/openstack-tox-py36/ce00fb7/ : SUCCESS in 7m 47s\n- openstack-tox-py37 https://logs.opendev.org/05/673505/3/gate/openstack-tox-py37/28f6c0f/ : SUCCESS in 6m 29s\n- openstack-tox-functional https://logs.opendev.org/05/673505/3/gate/openstack-tox-functional/6beb940/ : SUCCESS in 5m 59s\n- openstack-tox-functional-py36 https://logs.opendev.org/05/673505/3/gate/openstack-tox-functional-py36/e6c4b09/ : SUCCESS in 6m 36s\n- openstack-tox-functional-py37 https://logs.opendev.org/05/673505/3/gate/openstack-tox-functional-py37/9b9413d/ : SUCCESS in 7m 59s\n- placement-nova-tox-functional-py37 https://logs.opendev.org/05/673505/3/gate/placement-nova-tox-functional-py37/3bef777/ : SUCCESS in 11m 50s\n- tempest-ipv6-only https://logs.opendev.org/05/673505/3/gate/tempest-ipv6-only/a74c765/ : SUCCESS in 1h 17m 11s","accounts_in_message":[],"_revision_number":3},{"id":"c09c4c01116b8d1d1426bc52990575fb1152f039","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 16:11:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7464ff6e24c34e412ec3def799edbc9dedd3a2ef","revisions":{"ec3b3ae1303ba7824181b30bc7d9468e1f4f0a86":{"kind":"REWORK","_number":1,"created":"2019-07-30 12:49:26.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/05/673505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/673505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/673505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/673505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/673505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/673505/1"}}},"commit":{"parents":[{"commit":"2f56f379a568d427ac9057086386074a7801b00c","subject":"Merge \"Use integrated-gate-placement zuul template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/2f56f379a568d427ac9057086386074a7801b00c"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-07-30 12:46:13.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-07-30 12:46:13.000000000","tz":60},"subject":"Run nested-perfload parallel correctly","message":"Run nested-perfload parallel correctly\n\nWhile experimenting with expanding the nested perfload tests,\nit became clear that the call to parallel was not working as\ndocumented. What\u0027s supposed to be happening is that 3*nprocs\nprocesses are run, but in local testing that\u0027s not what happens:\ninstead just one, in sequence.\n\nThis changes it to use available processors.\n\nSubsequent patches will add a more complicated nested structure.\n\nChange-Id: Ie4809abc31212711b96f69e5f291104ae761059e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ec3b3ae1303ba7824181b30bc7d9468e1f4f0a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/ec3b3ae1303ba7824181b30bc7d9468e1f4f0a86"}]},"branch":"refs/heads/master"},"735e8d52a51a5acb31463cd4c4a4e5a81440e8a8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-31 10:19:30.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/05/673505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/673505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/673505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/673505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/673505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/673505/2"}}},"commit":{"parents":[{"commit":"541052ad5edcdd6328f59d71d94b0ae39879d74a","subject":"Merge \"Extra gabbi tests for same_subtree\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/541052ad5edcdd6328f59d71d94b0ae39879d74a"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-07-30 12:46:13.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-07-31 10:04:56.000000000","tz":60},"subject":"Run nested-perfload parallel correctly","message":"Run nested-perfload parallel correctly\n\nWhile experimenting with expanding the nested perfload tests,\nit became clear that the call to parallel was not working as\ndocumented. What\u0027s supposed to be happening is that 3*nprocs\nprocesses are run, but in local testing that\u0027s not what happens:\ninstead just one, in sequence.\n\nThis changes it to use available processors.\n\nSubsequent patches will add a more complicated nested structure.\n\nChange-Id: Ie4809abc31212711b96f69e5f291104ae761059e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/735e8d52a51a5acb31463cd4c4a4e5a81440e8a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/735e8d52a51a5acb31463cd4c4a4e5a81440e8a8"}]},"branch":"refs/heads/master"},"7464ff6e24c34e412ec3def799edbc9dedd3a2ef":{"kind":"REWORK","_number":3,"created":"2019-08-06 08:19:01.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/05/673505/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/05/673505/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/673505/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/673505/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/05/673505/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/05/673505/3"}}},"commit":{"parents":[{"commit":"b12cba973d3c655a2a3ef304692c5358847859b8","subject":"Merge \"Use expanding bindparam in provider_ids_from_rp_ids in_\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/b12cba973d3c655a2a3ef304692c5358847859b8"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-07-30 12:46:13.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-06 08:14:58.000000000","tz":60},"subject":"Run nested-perfload parallel correctly","message":"Run nested-perfload parallel correctly\n\nWhile experimenting with expanding the nested perfload tests,\nit became clear that the call to parallel was not working as\nexpected because the documents were misread. With help from\nTetsuro the correct incantation was determined so that we\nuse 50% of available CPUs.\n\nThis should leave some space for the database and the web\nserver.\n\nSubsequent patches will add a more complicated nested structure.\n\nCo-Authored-By: Tetsuro Nakamura \u003ctetsuro.nakamura.bc@hco.ntt.co.jp\u003e\nChange-Id: Ie4809abc31212711b96f69e5f291104ae761059e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7464ff6e24c34e412ec3def799edbc9dedd3a2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/7464ff6e24c34e412ec3def799edbc9dedd3a2ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
