)]}'
{"id":"openstack%2Fplacement~674661","triplet_id":"openstack%2Fplacement~master~I4e4db57555e8d570c446c6432151c276996a42ac","project":"openstack/placement","branch":"master","topic":"optimize-_build_provider_summaries","hashtags":[],"change_id":"I4e4db57555e8d570c446c6432151c276996a42ac","subject":"WIP: Use orjson in python3 for allocation candidate dump","status":"ABANDONED","created":"2019-08-05 17:49:25.000000000","updated":"2019-09-10 09:55:43.000000000","total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9a3bc8e50e19a3d15d3c85b3239fde93a72caae1","_number":674661,"virtual_id_number":674661,"owner":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-08-19 22:15:41.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":1,"date":"2019-08-16 15:08:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":-1,"date":"2019-08-29 09:46:01.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 11:37:10.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-08-16 15:08:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 22:15:41.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"}],"messages":[{"id":"2e2f8aa37cc903e1044c1623d85fa037df4ed8ad","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-05 17:49:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f24a60cb66a54cc204869b3fcccef01328c9df4c","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-05 18:10:07.000000000","message":"Patch Set 1:\n\nGetting much better concurrency numbers at this end of the stack.\n\nab -c 80 -n 800 (80 concurrency, 800 total requests) leads to 5.38 requests per seconds and mean request of 14.8 seconds.\n\nwhen this work was started it was less than a request per second, so we\u0027re making headway.","accounts_in_message":[],"_revision_number":1},{"id":"40ce985f81299dc854d9b34cceacf39bf2616199","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-05 18:34:30.000000000","message":"Patch Set 1:\n\noops, yet again, wrong request. right numbers:\n\n80x800\n\n3.19 requests per seconds\n25 seconds per request\n\nnot as good as the wrong request numbers just now, but still much better than before","accounts_in_message":[],"_revision_number":1},{"id":"d2836d3de1aa10da9eeafdfb5ceaef8999ee116c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 19:10:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://logs.opendev.org/61/674661/1/check/grenade-py3/45fdfab/ : SUCCESS in 1h 02m 07s\n- tempest-integrated-placement https://logs.opendev.org/61/674661/1/check/tempest-integrated-placement/478be61/ : SUCCESS in 1h 18m 33s\n- openstack-tox-docs https://logs.opendev.org/61/674661/1/check/openstack-tox-docs/2ee3a04/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover https://logs.opendev.org/61/674661/1/check/openstack-tox-cover/302a60d/cover/ : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://logs.opendev.org/61/674661/1/check/openstack-tox-lower-constraints/7e89589/ : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://logs.opendev.org/61/674661/1/check/openstack-tox-pep8/dca501f/ : SUCCESS in 4m 54s\n- openstack-tox-py27 https://logs.opendev.org/61/674661/1/check/openstack-tox-py27/5855515/ : SUCCESS in 6m 48s\n- openstack-tox-py36 https://logs.opendev.org/61/674661/1/check/openstack-tox-py36/9d2b806/ : SUCCESS in 6m 19s\n- openstack-tox-py37 https://logs.opendev.org/61/674661/1/check/openstack-tox-py37/ebef04c/ : SUCCESS in 7m 14s\n- openstack-tox-functional https://logs.opendev.org/61/674661/1/check/openstack-tox-functional/63f3924/ : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://logs.opendev.org/61/674661/1/check/openstack-tox-functional-py36/33497e0/ : SUCCESS in 7m 07s\n- openstack-tox-functional-py37 https://logs.opendev.org/61/674661/1/check/openstack-tox-functional-py37/453406d/ : SUCCESS in 6m 02s\n- placement-nova-tox-functional-py37 https://logs.opendev.org/61/674661/1/check/placement-nova-tox-functional-py37/a2ac529/ : SUCCESS in 13m 43s\n- placement-nested-perfload https://logs.opendev.org/61/674661/1/check/placement-nested-perfload/f0921f5/ : SUCCESS in 17m 12s (non-voting)\n- placement-perfload https://logs.opendev.org/61/674661/1/check/placement-perfload/fa0f633/ : SUCCESS in 8m 23s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/61/674661/1/check/tempest-ipv6-only/22fbd3a/ : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":1},{"id":"6aaa4203c13fb6e85500d50455a67e7d7f9adf9d","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-06 08:33:13.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"864a4af1a7af96eb7929777454c31aaad050fb2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:03:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://logs.opendev.org/61/674661/2/check/grenade-py3/0d0b1a7/ : SUCCESS in 1h 00m 35s\n- tempest-integrated-placement https://logs.opendev.org/61/674661/2/check/tempest-integrated-placement/f018930/ : SUCCESS in 1h 24m 55s\n- openstack-tox-docs https://logs.opendev.org/61/674661/2/check/openstack-tox-docs/6307243/html/ : SUCCESS in 4m 12s\n- openstack-tox-cover https://logs.opendev.org/61/674661/2/check/openstack-tox-cover/69b5830/cover/ : SUCCESS in 5m 34s\n- openstack-tox-lower-constraints https://logs.opendev.org/61/674661/2/check/openstack-tox-lower-constraints/9507f5b/ : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://logs.opendev.org/61/674661/2/check/openstack-tox-pep8/6fc4e96/ : SUCCESS in 4m 17s\n- openstack-tox-py27 https://logs.opendev.org/61/674661/2/check/openstack-tox-py27/2d8889d/ : SUCCESS in 6m 26s\n- openstack-tox-py36 https://logs.opendev.org/61/674661/2/check/openstack-tox-py36/e93ab86/ : SUCCESS in 7m 02s\n- openstack-tox-py37 https://logs.opendev.org/61/674661/2/check/openstack-tox-py37/c6e920c/ : SUCCESS in 5m 39s\n- openstack-tox-functional https://logs.opendev.org/61/674661/2/check/openstack-tox-functional/eb75dbf/ : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://logs.opendev.org/61/674661/2/check/openstack-tox-functional-py36/8eda667/ : SUCCESS in 7m 32s\n- openstack-tox-functional-py37 https://logs.opendev.org/61/674661/2/check/openstack-tox-functional-py37/499cd86/ : SUCCESS in 5m 30s\n- placement-nova-tox-functional-py37 https://logs.opendev.org/61/674661/2/check/placement-nova-tox-functional-py37/9b2091a/ : SUCCESS in 12m 57s\n- placement-nested-perfload https://logs.opendev.org/61/674661/2/check/placement-nested-perfload/220d492/ : SUCCESS in 16m 57s (non-voting)\n- placement-perfload https://logs.opendev.org/61/674661/2/check/placement-perfload/f1e3ce8/ : SUCCESS in 9m 05s (non-voting)\n- tempest-ipv6-only https://logs.opendev.org/61/674661/2/check/tempest-ipv6-only/9781e63/ : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":2},{"id":"270bad163620edb9e27a88d8346c2eb993d4ddc2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-08-08 11:37:10.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a0cd9ce9c6e50e2eceb2c4bc2e3236f421b5636a","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-08 11:42:41.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nthanks for the feedback gibi","accounts_in_message":[],"_revision_number":2},{"id":"cd2504a9fd44823cd087d40bd06ac63e8a84d057","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-15 15:26:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1f63243a348519d416a6c125a1d459a4e6719e24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 17:02:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/98f252bf2e704cfc83801c89b1b4f87b : SUCCESS in 59m 12s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/74eee761e7d04642a69eff5b185fc692 : SUCCESS in 1h 28m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4fe15975fd494a8de2f593cf194653 : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3334544da8940e096bb7ce333b9bde9 : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e72e6b118f8b4557b8a7bec722be05e4 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e121835d28f44e2fa70c6203c06d8d09 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a90fa253c6454b8498a941e4baed07bf : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce5f69e37a804e6a8cbc76d5a79b7e8a : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09c1968ce4b64c4f88a24b23c03eba2a : SUCCESS in 4m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9cec79176ac44fc3a25a1a82b44ec8c4 : SUCCESS in 5m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4a12dc015b542f0a71821b00aa63655 : SUCCESS in 5m 12s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/8389ed3040a14441836b3cf99abe8a25 : SUCCESS in 5m 46s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c8981b21e39d4497b9f1a65c55df13d9 : SUCCESS in 12m 29s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/069306b6cc4143f99576282e5f823996 : SUCCESS in 33m 37s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/31d9d1a8ecf9406db4914417c8ed3247 : SUCCESS in 11m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/691809fc1e8b4adc8e88d5366946ba08 : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":3},{"id":"00ffe7edf226575f53b0bab6cc1f1a63b2567b3b","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-16 09:52:29.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"2eec2b690da57640efe3a9785f8fe5c13372b1ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 11:23:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/11e3afb0697d4df3bcc135ba2102f04c : SUCCESS in 1h 13m 21s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/c127c58976134027b9e0be25f7825463 : SUCCESS in 1h 24m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92792127919d41ec8e770b9541764849 : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb5a8ba40ebf4ad8b1c5455346e01cea : SUCCESS in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb3e95e480ce4dabb7ab43fe683634cf : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d48c4344117410db752bcd9686e518f : SUCCESS in 7m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13eabe889d51490bbbedc09b3c94ae5c : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9456baf41c2428b83ea149235be7d82 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f076076f96e24e4eb93170b7c299a69c : SUCCESS in 7m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cfed31f6edb5416e970888134c510c92 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/469b44b0700a4f5eacf999803167e8c3 : SUCCESS in 5m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/edd262fd149a462c9e9ce451f09fc3b6 : SUCCESS in 7m 41s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/35a6a6b1f31448b88c34a578c06c7e9f : SUCCESS in 11m 21s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/280513e828ee499289ec5151ac2f5ac0 : SUCCESS in 27m 27s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/e7ca0230081f45f690e4d63da043efc8 : SUCCESS in 11m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c46571e2d62d4e2f8cc32269c62d13cb : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":4},{"id":"9a62473fe303e66f03201883e8fef3e7e9ecb7b2","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-16 13:39:22.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6514808d60e9c134d43c76b611f16125540a4715","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:08:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/65776493008c4ac39c652c2cd8f359bd : SUCCESS in 1h 16m 48s\n- tempest-integrated-placement https://zuul.opendev.org/t/openstack/build/44c2f52c854b41da80dfe3a476c45e27 : SUCCESS in 1h 25m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78bd322de12247d0a150600a30306122 : SUCCESS in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/febb6be27c6940bdb17c5b64992646e6 : SUCCESS in 5m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cdaa5544de5405fba54957865c1c5cf : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1193622b354fd9aac6834bc99c25de : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c15d595f9a2d4c90abd951dfe0b8ec75 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5cf4fb1c78a40bd84fe440a228a5d7b : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d84d736dc12b4c92b6ffd012190e04ea : SUCCESS in 8m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/88a0d320998a42e1ad4f97fe6800ae78 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc869199a87f411cac68595370f1826a : SUCCESS in 7m 15s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/480b87b8c3534f20a334babedb2786f9 : SUCCESS in 6m 42s\n- placement-nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d202eee3dd34473cb93c626511556bac : SUCCESS in 13m 20s\n- placement-nested-perfload https://zuul.opendev.org/t/openstack/build/ce608a23d57a4c40aa38c93dc823f9e0 : SUCCESS in 10m 46s (non-voting)\n- placement-perfload https://zuul.opendev.org/t/openstack/build/9f0fc076de4249248aeba781153a32bd : SUCCESS in 12m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/754cbaf9509046199a9c2c3d071049c3 : SUCCESS in 1h 07m 19s","accounts_in_message":[],"_revision_number":5},{"id":"db6b9e33dd715bb2f58bc186fa0a69798c8f9387","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 22:15:06.000000000","message":"Patch Set 5:\n\nIf this is giving us a big improvement, I\u0027m in favor.\n\nIs there a reason orjson isn\u0027t already in g-r? (Because py3.5 and up perhaps?)","accounts_in_message":[],"_revision_number":5},{"id":"7a9375f50e3aa7552a5b70ba7ea5af501d3a00be","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-19 22:15:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d9b24edfa3c8c45a0024b410ff5c7f3bd25c6384","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-08-29 09:46:01.000000000","message":"Patch Set 5: Workflow-1\n\ndiscussion [1] with requirements and other folk have pointed out some potential issues with orjson, so it might be a non-starter\n\nfalling back to other options, already in requirements, is probably worth investigating\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-August/thread.html#8849","accounts_in_message":[],"_revision_number":5},{"id":"a454d840bbed7448f508f93b4a27f59a6b5b6ecf","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-09-10 09:55:43.000000000","message":"Abandoned\n\nnot gonna go this way because of requirements limitations\n\nalso, not going to explore other options in the immediate future because there\u0027s work in progress in the requirements group to narrow which json libraries are used/available","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a494ba5e0593c76b3530ca74181d6fa2284db305","revisions":{"9a98807ff551e838c7f9c6c7abfca840c2cd7fc2":{"kind":"REWORK","_number":1,"created":"2019-08-05 17:49:25.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/61/674661/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/61/674661/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/61/674661/1"}}},"commit":{"parents":[{"commit":"88ab5f69f2b398ffc3d1fe4cd9f5a2a025cc747e","subject":"Use expanding bindparam in extend_usages_by_provider_tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/88ab5f69f2b398ffc3d1fe4cd9f5a2a025cc747e"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-05 17:44:16.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-05 17:44:16.000000000","tz":60},"subject":"WIP: Use orjson in python3 for allocation candidate dump","message":"WIP: Use orjson in python3 for allocation candidate dump\n\norjson [1] is about 2x faster than the stdlib json library when\ndump large objects however it comes with a couple of caveats:\n\n* It only works for python\u003e\u003d3.5 (easy to work around)\n* It only writes types that are native to JSON, so sets and default\n  dicts need to be worked around (as seen in this change).\n\nThe results are useful though. A request which was taking 1.57 seconds\ntakes 1.35s with these changes. That doesn\u0027t sound like much, but it\na pretty reasonable improvement that in aggregate would add up.\n\nA WIP because:\n\n* we should talk about it\n* we\u0027d need global requirements\n* we\u0027d probably want to do it anywhere we dump large (for loads it would\n  not be much of a win)\n\n[1] https://pypi.org/project/orjson/\n\nChange-Id: I4e4db57555e8d570c446c6432151c276996a42ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9a98807ff551e838c7f9c6c7abfca840c2cd7fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9a98807ff551e838c7f9c6c7abfca840c2cd7fc2"}]},"branch":"refs/heads/master"},"92cc250c604d8b12e920f135fb8fe8b0ee693997":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-06 08:33:13.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/61/674661/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/61/674661/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/61/674661/2"}}},"commit":{"parents":[{"commit":"5883c794f969b8fda14078137704b4dd34919ab0","subject":"Use expanding bindparam in extend_usages_by_provider_tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5883c794f969b8fda14078137704b4dd34919ab0"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-05 17:44:16.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-06 08:33:05.000000000","tz":60},"subject":"WIP: Use orjson in python3 for allocation candidate dump","message":"WIP: Use orjson in python3 for allocation candidate dump\n\norjson [1] is about 2x faster than the stdlib json library when\ndump large objects however it comes with a couple of caveats:\n\n* It only works for python\u003e\u003d3.5 (easy to work around)\n* It only writes types that are native to JSON, so sets and default\n  dicts need to be worked around (as seen in this change).\n\nThe results are useful though. A request which was taking 1.57 seconds\ntakes 1.35s with these changes. That doesn\u0027t sound like much, but it\na pretty reasonable improvement that in aggregate would add up.\n\nA WIP because:\n\n* we should talk about it\n* we\u0027d need global requirements\n* we\u0027d probably want to do it anywhere we dump large (for loads it would\n  not be much of a win)\n\n[1] https://pypi.org/project/orjson/\n\nChange-Id: I4e4db57555e8d570c446c6432151c276996a42ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/92cc250c604d8b12e920f135fb8fe8b0ee693997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/92cc250c604d8b12e920f135fb8fe8b0ee693997"}]},"branch":"refs/heads/master"},"9732bcadc0dce86a080f9478375b33fd62c841f2":{"kind":"REWORK","_number":3,"created":"2019-08-15 15:26:00.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/61/674661/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/61/674661/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/61/674661/3"}}},"commit":{"parents":[{"commit":"5e718c3a6fefed807919a8bf271e5ff36f0a5f17","subject":"Avoid duplicate ProviderSummary in _merge_candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/5e718c3a6fefed807919a8bf271e5ff36f0a5f17"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-05 17:44:16.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-15 15:26:00.000000000","tz":0},"subject":"WIP: Use orjson in python3 for allocation candidate dump","message":"WIP: Use orjson in python3 for allocation candidate dump\n\norjson [1] is about 2x faster than the stdlib json library when\ndump large objects however it comes with a couple of caveats:\n\n* It only works for python\u003e\u003d3.5 (easy to work around)\n* It only writes types that are native to JSON, so sets and default\n  dicts need to be worked around (as seen in this change).\n\nThe results are useful though. A request which was taking 1.57 seconds\ntakes 1.35s with these changes. That doesn\u0027t sound like much, but it\na pretty reasonable improvement that in aggregate would add up.\n\nA WIP because:\n\n* we should talk about it\n* we\u0027d need global requirements\n* we\u0027d probably want to do it anywhere we dump large (for loads it would\n  not be much of a win)\n\n[1] https://pypi.org/project/orjson/\n\nChange-Id: I4e4db57555e8d570c446c6432151c276996a42ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9732bcadc0dce86a080f9478375b33fd62c841f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/9732bcadc0dce86a080f9478375b33fd62c841f2"}]},"branch":"refs/heads/master"},"1cc97ec1326d8980a0bb8f0ed774936d0620a2e9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-16 09:52:29.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/61/674661/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/61/674661/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/61/674661/4"}}},"commit":{"parents":[{"commit":"aecd0f3497c5e4f1e73bf6c76a9add3db6270e27","subject":"Avoid duplicate ProviderSummary in _merge_candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/aecd0f3497c5e4f1e73bf6c76a9add3db6270e27"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-05 17:44:16.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-16 09:44:37.000000000","tz":60},"subject":"WIP: Use orjson in python3 for allocation candidate dump","message":"WIP: Use orjson in python3 for allocation candidate dump\n\norjson [1] is about 2x faster than the stdlib json library when\ndump large objects however it comes with a couple of caveats:\n\n* It only works for python\u003e\u003d3.5 (easy to work around)\n* It only writes types that are native to JSON, so sets and default\n  dicts need to be worked around (as seen in this change).\n\nThe results are useful though. A request which was taking 1.57 seconds\ntakes 1.35s with these changes. That doesn\u0027t sound like much, but it\na pretty reasonable improvement that in aggregate would add up.\n\nA WIP because:\n\n* we should talk about it\n* we\u0027d need global requirements\n* we\u0027d probably want to do it anywhere we dump large (for loads it would\n  not be much of a win)\n\n[1] https://pypi.org/project/orjson/\n\nChange-Id: I4e4db57555e8d570c446c6432151c276996a42ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1cc97ec1326d8980a0bb8f0ed774936d0620a2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/1cc97ec1326d8980a0bb8f0ed774936d0620a2e9"}]},"branch":"refs/heads/master"},"a494ba5e0593c76b3530ca74181d6fa2284db305":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-08-16 13:39:22.000000000","uploader":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"ref":"refs/changes/61/674661/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/placement","ref":"refs/changes/61/674661/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/placement refs/changes/61/674661/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/placement refs/changes/61/674661/5"}}},"commit":{"parents":[{"commit":"aecd0f3497c5e4f1e73bf6c76a9add3db6270e27","subject":"Avoid duplicate ProviderSummary in _merge_candidates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/aecd0f3497c5e4f1e73bf6c76a9add3db6270e27"}]}],"author":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-05 17:44:16.000000000","tz":60},"committer":{"name":"Chris Dent","email":"cdent@anticdent.org","date":"2019-08-16 13:37:05.000000000","tz":60},"subject":"WIP: Use orjson in python3 for allocation candidate dump","message":"WIP: Use orjson in python3 for allocation candidate dump\n\norjson [1] is about 2x faster than the stdlib json library when\ndumping large objects however it comes with a couple of caveats:\n\n* It only works for python\u003e\u003d3.5 (easy to work around)\n* It only writes types that are native to JSON, so sets and default\n  dicts need to be worked around (as seen in this change).\n\nThe results are useful though. A request which was taking 0.953 seconds\ntakes 0.731s with these changes. That is, json dumping is 24.18% of\ntime when using the default code, but only 1.16% using the new.\n\nA WIP because:\n\n* we should talk about it\n* we\u0027d need global requirements\n* we\u0027d probably want to do it anywhere we dump large (for loads it would\n  not be much of a win)\n\n[1] https://pypi.org/project/orjson/\n\nChange-Id: I4e4db57555e8d570c446c6432151c276996a42ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a494ba5e0593c76b3530ca74181d6fa2284db305"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/placement/commit/a494ba5e0593c76b3530ca74181d6fa2284db305"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
